)]}'
{"id":"openstack%2Fsushy~757907","triplet_id":"openstack%2Fsushy~master~I5f52d07c1bc9e67c543d32365568b284b57e6500","project":"openstack/sushy","branch":"master","topic":"overview","hashtags":[],"change_id":"I5f52d07c1bc9e67c543d32365568b284b57e6500","subject":"Adds sushy library overview","status":"ABANDONED","created":"2020-10-13 16:10:01.000000000","updated":"2020-10-27 09:16:36.000000000","total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"16eaa415be906d8b32aec55d46160f08fa7d0b2e","_number":757907,"virtual_id_number":757907,"owner":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"date":"2020-10-27 07:33:25.000000000","_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"tag":"autogenerated:zuul:check","value":1,"date":"2020-10-27 08:13:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"date":"2020-10-27 06:42:07.000000000","_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"disliked":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},{"value":-1,"date":"2020-10-27 08:33:47.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":0},"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-10-18 13:08:53.000000000","updated_by":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"reviewer":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"state":"REVIEWER"},{"updated":"2020-10-26 17:38:19.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2020-10-27 07:33:25.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"REVIEWER"},{"updated":"2020-10-27 08:13:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-10-27 08:33:47.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"a12a260489ec2fb1ec3fb8a90d915abde3db89f5","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-13 16:10:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9ca5c81772789bcb92f364d6cca473543b77d961","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-13 16:34:26.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"100358163e39de1acecec7a2d6e3ae2674afc0a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-13 17:49:35.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0f2997ebf1774fcea2680fa426012ab9 : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/544913a591464988a84964aaf60485f5 : SUCCESS in 6m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6a6b4171b52d4d83af216d5b72a83c12 : SUCCESS in 50m 50s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/eb76a72e69ef4f3d85cb841de1ad9293 : SUCCESS in 55m 47s","accounts_in_message":[],"_revision_number":2},{"id":"feed9d736861f47bc8b19bbb93253ada6191714c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-13 20:50:28.000000000","message":"Patch Set 2:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-10-13:11:34/757907/2/ : SUCCESS in 53m 15s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-10-13:11:34/757907/2/ : FAILURE in 1h 37m 08s","accounts_in_message":[],"_revision_number":2},{"id":"438637516a940744ef0c7ef347b13a9891e4230a","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-10-15 14:43:35.000000000","message":"Patch Set 2: Code-Review-1\n\nHi Kafilat, \n\nThanks for the patch,  according to the description of the Task https://storyboard.openstack.org/#!/story/2006610 , the Overview should provide some use-cases","accounts_in_message":[],"_revision_number":2},{"id":"181e4cd48c68549f5039e2769ab0397754e1c1d6","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-18 08:08:57.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"a30e56a4560e3d1a731995e897028cb6f8eb9340","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-18 08:19:23.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3f9f77654a061e77ab55a5b4c87cd78b8a6b6ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-18 09:17:44.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6c16e45a9f4c418486a2671cc46f9d8c : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e6f00c5719844279957e26fd97083a4 : SUCCESS in 9m 01s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3ee186cb8ebc49b9bd01fdff73741c0c : SUCCESS in 52m 23s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/976866be723b4edd813d35a60b75ae65 : SUCCESS in 55m 01s","accounts_in_message":[],"_revision_number":4},{"id":"750d1fe4c0d6fa033d9798e5ad79479363387c39","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-18 10:05:29.000000000","message":"Patch Set 4:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-10-18:03:19/757907/4/ : FAILURE in 50m 40s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-10-18:03:19/757907/4/ : FAILURE in 1h 45m 54s","accounts_in_message":[],"_revision_number":4},{"id":"8a4f2c6baf4a0fa0e70e96b6a6c7b7f06dc289f5","author":{"_account_id":28182,"name":"Jiaping LI","email":"lijiaping@yovole.com","username":"lijiaping"},"date":"2020-10-18 13:08:53.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"e1230fd9a115437938daa607bf29122f58b59c9c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-10-19 11:06:42.000000000","message":"Patch Set 4: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"e36268b3af57a0c604a58c1f7afe586e71a8251b","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-23 06:13:26.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1db79285f2991e5e554cd9c537ecfad6bcfd822a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-23 07:15:39.000000000","message":"Patch Set 5:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-10-23:01:13/757907/5/ : SUCCESS in 1h 01m 42s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-10-23:01:13/757907/5/ : FAILURE in 1h 02m 01s","accounts_in_message":[],"_revision_number":5},{"id":"4a5eb13c3644687904820fe9a8589f60eb6cbe43","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-23 07:25:43.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a40669c6d3d47889ca606a0d541265d : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1a084c7b3aa4d459581871c9f818f81 : SUCCESS in 8m 43s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/39c1ac61a4a541e78b6a93f596db8b9a : SUCCESS in 53m 08s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c4371e385fc140f08f06633117a42f62 : SUCCESS in 53m 49s","accounts_in_message":[],"_revision_number":5},{"id":"71e57324bbe6e61db6293119c9adbfe29b6ddba2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2020-10-26 17:38:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c4d78e3508513b948613a283f4667453ad54c5a1","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-26 17:52:06.000000000","message":"Patch Set 6: Published edit on patch set 5.","accounts_in_message":[],"_revision_number":6},{"id":"6e66fb7f5f0a573aa88e69c9bbec672fe40ff3a6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-26 18:59:18.000000000","message":"Patch Set 6:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-10-26:12:52/757907/6/ : SUCCESS in 52m 39s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-10-26:12:52/757907/6/ : FAILURE in 1h 06m 58s","accounts_in_message":[],"_revision_number":6},{"id":"6339d9390eb5c0f189c4dd13602ea20b5019a97c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-26 20:14:38.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4624708032b3410db50089b0f120a238 : SUCCESS in 3m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/33f73f8e1e8c4e6c9ad16d229a9f2f3f : TIMED_OUT in 31m 11s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e2d3ef69582a485684ae8a4ab9055ec6 : SUCCESS in 2h 10m 58s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/08daa9620a7e4754940723615cc176d8 : SUCCESS in 57m 57s","accounts_in_message":[],"_revision_number":6},{"id":"22d6c7eb24058513ddfa12962f04db3825e6fa51","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-27 06:27:00.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.","accounts_in_message":[],"_revision_number":7},{"id":"fec72044d2ffc6ca5c1dbd76286fe637e4bde5c0","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-27 06:42:07.000000000","message":"Patch Set 7:\n\nrecheck","accounts_in_message":[],"_revision_number":7},{"id":"1d1c190ae392c1b950cc8523d6badba208221eec","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-10-27 07:33:25.000000000","message":"Patch Set 7:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-10-27:01:27/757907/7/ : FAILURE in 1h 06m 13s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-10-27:01:27/757907/7/ : FAILURE in 1h 01m 17s","accounts_in_message":[],"_revision_number":7},{"id":"b298a9597518426c9cab2f9c1e4fa7ec12411428","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-10-27 08:13:13.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0a1e390f16144b1d89d326802c5b7c2e : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da815e9793734266bedbd3267ae56326 : SUCCESS in 6m 27s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6fe3afe3b609440b948a07df4e69df62 : SUCCESS in 51m 23s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/87a2b8c3a597481da2f83f1766a83078 : SUCCESS in 58m 52s","accounts_in_message":[],"_revision_number":7},{"id":"094c9f7475d6bf6af35ad353e26eeff730d4ca25","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-10-27 08:33:47.000000000","message":"Patch Set 7: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":7},{"id":"05aa67b2bb0231b74135ba4fbd4cc658f0b7b953","author":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"date":"2020-10-27 09:16:36.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":7}],"current_revision_number":7,"current_revision":"fc32ad71d0380798a860f3523b5da32fbbe2e667","revisions":{"1971302499c2b7c00a34519a894a6caa76ca87a4":{"kind":"REWORK","_number":1,"created":"2020-10-13 16:10:01.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/07/757907/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/07/757907/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/07/757907/1"}}},"commit":{"parents":[{"commit":"f91033cae06c1a078c9a1f375efbe3dc555aec6b","subject":"Lower log severity for unknown type registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f91033cae06c1a078c9a1f375efbe3dc555aec6b"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"subject":"Adds sushy library overview","message":"Adds sushy library overview\n\nChange-Id: I5f52d07c1bc9e67c543d32365568b284b57e6500\nStory: #2006610\nTask: #36858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1971302499c2b7c00a34519a894a6caa76ca87a4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1971302499c2b7c00a34519a894a6caa76ca87a4"}]},"branch":"refs/heads/master"},"52cf83e26b0baf5e1d66aeebe104dd3703c7447a":{"kind":"REWORK","_number":2,"created":"2020-10-13 16:34:26.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/07/757907/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/07/757907/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/07/757907/2"}}},"commit":{"parents":[{"commit":"f91033cae06c1a078c9a1f375efbe3dc555aec6b","subject":"Lower log severity for unknown type registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f91033cae06c1a078c9a1f375efbe3dc555aec6b"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:30:51.000000000","tz":-420},"subject":"Adds sushy library overview","message":"Adds sushy library overview\n\nChange-Id: I5f52d07c1bc9e67c543d32365568b284b57e6500\nStory: #2006610\nTask: #36858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/52cf83e26b0baf5e1d66aeebe104dd3703c7447a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/52cf83e26b0baf5e1d66aeebe104dd3703c7447a"}]},"branch":"refs/heads/master"},"75e352a01b51f6eeab80e9fb28f373f80461eeed":{"kind":"REWORK","_number":3,"created":"2020-10-18 08:08:57.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/07/757907/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/07/757907/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/07/757907/3"}}},"commit":{"parents":[{"commit":"f91033cae06c1a078c9a1f375efbe3dc555aec6b","subject":"Lower log severity for unknown type registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f91033cae06c1a078c9a1f375efbe3dc555aec6b"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-18 08:05:03.000000000","tz":-420},"subject":"Adds sushy library overview","message":"Adds sushy library overview\n\nChange-Id: I5f52d07c1bc9e67c543d32365568b284b57e6500\nStory: #2006610\nTask: #36858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/75e352a01b51f6eeab80e9fb28f373f80461eeed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/75e352a01b51f6eeab80e9fb28f373f80461eeed"}]},"branch":"refs/heads/master"},"116094d620708e6650de23d665073134a1a8c0ed":{"kind":"REWORK","_number":4,"created":"2020-10-18 08:19:23.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/07/757907/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/07/757907/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/07/757907/4"}}},"commit":{"parents":[{"commit":"f91033cae06c1a078c9a1f375efbe3dc555aec6b","subject":"Lower log severity for unknown type registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f91033cae06c1a078c9a1f375efbe3dc555aec6b"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-18 08:18:06.000000000","tz":-420},"subject":"Adds sushy library overview","message":"Adds sushy library overview\n\nChange-Id: I5f52d07c1bc9e67c543d32365568b284b57e6500\nStory: #2006610\nTask: #36858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/116094d620708e6650de23d665073134a1a8c0ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/116094d620708e6650de23d665073134a1a8c0ed"}]},"branch":"refs/heads/master"},"4bd8a2e30cba45024cab079590e64a6459b26971":{"kind":"REWORK","_number":5,"created":"2020-10-23 06:13:26.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/07/757907/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/07/757907/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/07/757907/5"}}},"commit":{"parents":[{"commit":"f91033cae06c1a078c9a1f375efbe3dc555aec6b","subject":"Lower log severity for unknown type registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f91033cae06c1a078c9a1f375efbe3dc555aec6b"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-23 06:11:38.000000000","tz":-420},"subject":"Adds sushy library overview","message":"Adds sushy library overview\n\nChange-Id: I5f52d07c1bc9e67c543d32365568b284b57e6500\nStory: #2006610\nTask: #36858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4bd8a2e30cba45024cab079590e64a6459b26971"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4bd8a2e30cba45024cab079590e64a6459b26971"}]},"branch":"refs/heads/master"},"9e3d41b6221aea6c23be144bc91b1a5d62c2de9a":{"kind":"REWORK","_number":6,"created":"2020-10-26 17:52:06.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/07/757907/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/07/757907/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/07/757907/6"}}},"commit":{"parents":[{"commit":"f91033cae06c1a078c9a1f375efbe3dc555aec6b","subject":"Lower log severity for unknown type registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f91033cae06c1a078c9a1f375efbe3dc555aec6b"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-26 17:51:31.000000000","tz":0},"subject":"Adds sushy library overview","message":"Adds sushy library overview\n\nChange-Id: I5f52d07c1bc9e67c543d32365568b284b57e6500\nStory: #2006610\nTask: #36858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9e3d41b6221aea6c23be144bc91b1a5d62c2de9a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9e3d41b6221aea6c23be144bc91b1a5d62c2de9a"}]},"branch":"refs/heads/master"},"fc32ad71d0380798a860f3523b5da32fbbe2e667":{"kind":"NO_CHANGE","_number":7,"created":"2020-10-27 06:27:00.000000000","uploader":{"_account_id":32493,"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","username":"kafilat-adeleke"},"ref":"refs/changes/07/757907/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/07/757907/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/07/757907/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/07/757907/7"}}},"commit":{"parents":[{"commit":"f91033cae06c1a078c9a1f375efbe3dc555aec6b","subject":"Lower log severity for unknown type registries","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f91033cae06c1a078c9a1f375efbe3dc555aec6b"}]}],"author":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-13 16:05:16.000000000","tz":-420},"committer":{"name":"Kafilat Adeleke","email":"adelekekafilatadenike@gmail.com","date":"2020-10-27 06:20:29.000000000","tz":-420},"subject":"Adds sushy library overview","message":"Adds sushy library overview\n\nChange-Id: I5f52d07c1bc9e67c543d32365568b284b57e6500\nStory: #2006610\nTask: #36858\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fc32ad71d0380798a860f3523b5da32fbbe2e667"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fc32ad71d0380798a860f3523b5da32fbbe2e667"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
