)]}'
{"id":"openstack%2Fsushy~764915","triplet_id":"openstack%2Fsushy~master~Ida1d314f77b9f114e572639c7ad46545be1ce962","project":"openstack/sushy","branch":"master","topic":"firmware","hashtags":[],"change_id":"Ida1d314f77b9f114e572639c7ad46545be1ce962","subject":"Fixes issue of redfish firmware update","status":"MERGED","created":"2020-12-01 09:28:39.000000000","updated":"2021-01-18 13:26:49.000000000","submitted":"2021-01-18 13:24:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":60,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"764915-firmware","meta_rev_id":"101a598b929f0d0099024291605463324c88a147","_number":764915,"virtual_id_number":764915,"owner":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-01-18 13:24:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},{"value":2,"date":"2021-01-16 12:46:48.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2021-01-14 14:54:31.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},{"value":1,"date":"2021-01-18 12:15:39.000000000","_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"value":0,"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},{"value":0,"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"}],"reviewers":{"REVIEWER":[{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},{"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"}],"CC":[{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2020-12-01 10:29:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2020-12-01 10:55:27.000000000","updated_by":{"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},"reviewer":{"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},"state":"REVIEWER"},{"updated":"2020-12-01 11:50:21.000000000","updated_by":{"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},"reviewer":{"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},"state":"REVIEWER"},{"updated":"2020-12-02 05:44:54.000000000","updated_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"reviewer":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"state":"REVIEWER"},{"updated":"2020-12-03 16:46:10.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2020-12-03 16:46:10.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"state":"REVIEWER"},{"updated":"2020-12-03 16:46:10.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"state":"REVIEWER"},{"updated":"2020-12-04 00:51:57.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":10250,"name":"Christopher Dearborn","email":"Christopher.Dearborn@dell.com","username":"cdearbor"},"state":"CC"},{"updated":"2020-12-04 00:51:57.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":31616,"name":"Eric Barrera","email":"Eric_Barrera@Dell.com","username":"erbarr"},"state":"REVIEWER"},{"updated":"2020-12-04 00:51:57.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"state":"REVIEWER"},{"updated":"2020-12-04 00:51:57.000000000","updated_by":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2020-12-14 12:15:07.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2020-12-16 07:54:06.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"97cf3b10fe34dd861bb179dc25a2036a7b3f3f5e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-01 09:28:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74c7536567ee8116338ca6fb10ecd314b7b1393b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-01 10:29:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f57ed1646a74313ad2a4684d4745d08 : SUCCESS in 4m 45s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/602d178eb4aa4ed0876d7af8d6d2db0e : SUCCESS in 4m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23ce0d24c0654553934d101b30aa4f8a : SUCCESS in 3m 21s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b45bc0187dfb4d26a24014029e2f5e83 : SUCCESS in 3m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9bf9dcc98d034536878e2880993f241c : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a07b9da08da84a6a9befc405c1bdee63 : SUCCESS in 3m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/abc7d12c3dd94fcab3bdc9f6ec388419 : SUCCESS in 5m 46s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7c53b270dae94957b9bd08607017604d : SUCCESS in 53m 03s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/00133c29b1274fe8a9f797b84c9623ba : SUCCESS in 54m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5c786cc73e0d9e7110990506ce1231ed57f47619","author":{"_account_id":11297,"name":"Nisha Agarwal","email":"agarwalnisha1980@gmail.com","username":"Nisha"},"date":"2020-12-01 10:55:27.000000000","message":"Patch Set 1: Code-Review+1\n\nThanks for the patch Ankit. Please test this patch with ilo hardware type as well.","accounts_in_message":[],"_revision_number":1},{"id":"4eac3ebf0e4af85ca35b1240af72b36bf9233337","author":{"_account_id":18781,"name":"vinay muddu","email":"vinay50muddu@yahoo.com","username":"vmud213"},"date":"2020-12-01 11:50:21.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nThanks Ankit for the patch. LGTM except for a minor clarification.","accounts_in_message":[],"_revision_number":1},{"id":"002f836af66f2a8077209723ad996d6b8d61c556","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-12-02 05:44:54.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)\n\nThank you Ankit. Can you please add a story for this and releasenotes.","accounts_in_message":[],"_revision_number":1},{"id":"d400eb3e4a192baa55daf1faec8b1988c37d841b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2020-12-03 16:46:10.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)\n\nGreetings, A release note and a story would be very helpful to understand. I do agree the fix is needed, I\u0027m just slightly worried about why the check was there in the first place. I\u0027ve asked around.","accounts_in_message":[],"_revision_number":1},{"id":"09e91b100183fdf65b66f0324634e5599507c0a6","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-12-03 18:48:11.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nI added a couple of comments related to the case of getting a 202 response with no body.","accounts_in_message":[],"_revision_number":1},{"id":"5de98a2669f78bd0f5f8ab0935ec4fbf3a4b6e03","author":{"_account_id":23847,"name":"Richard G. Pioso","email":"richard.pioso@gmail.com","username":"richard.pioso"},"date":"2020-12-04 00:51:57.000000000","message":"Patch Set 1: Code-Review-1\n\n(4 comments)\n\nThank you!","accounts_in_message":[],"_revision_number":1},{"id":"e171372220a784955c38287237434a50a43c9eee","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-12-04 10:17:19.000000000","message":"Patch Set 1: Code-Review-1\n\n(3 comments)\n\nthanks for the patch, comments inline.","accounts_in_message":[],"_revision_number":1},{"id":"69c018b650f41da25ee4cb54a77df2b8e3aa6e9a","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-12-04 11:15:35.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"3082cc47e34e819ac74879d987766c4cb94d4874","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-12-04 11:59:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0682ce2f4af1011780f52e088e4c6af89a362246","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2020-12-04 12:21:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c18073a7b8622094fa3ef42fe2564f30422be21","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-12-04 12:57:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d5ce673eccdd0cb377c2a65010542788fa5d0c8a","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-12-04 17:10:06.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"73cd3b0d98be5041f018e6b52804706a565b15ab","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-12-04 21:33:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3bef891100d5882e41b39f2b021cd2fb8d526e83","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-08 13:17:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e6248d4fa549002b40c0c7ba71802a6b237303f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-14 10:15:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1189fade88ef7cea84738209c0b0e0b85738d038","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-14 11:16:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/04fae0ba0706441c98ae20aba6b35d94 : SUCCESS in 3m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/bdf1dd6f9c3e41e0b9566d3a94104adf : SUCCESS in 3m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b6050805820a46cfa92917cad94f2f5e : SUCCESS in 2m 52s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4883c83a66ec415582091a9150cbf5ad : SUCCESS in 3m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7672339f4f8b4bf983367918d3c7d470 : SUCCESS in 3m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f47b5d766a6448e184301fb65b0a657e : SUCCESS in 3m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e27c1a3fc0914c999dba6255dc914364 : SUCCESS in 5m 31s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/746a891b2ef4484b8493dd57767ced0f : SUCCESS in 47m 53s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/37bf69e16dc24d56bfbf4580e1b8b8d6 : SUCCESS in 52m 49s","accounts_in_message":[],"_revision_number":2},{"id":"cfa6580dd47f9ca7dd86143d69bbd70cc9b7f608","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-14 12:15:07.000000000","message":"Patch Set 2: Code-Review-1\n\n(5 comments)\n\nSome comments inline + please add a release note","accounts_in_message":[],"_revision_number":2},{"id":"c0f07511c23ab5cc2207a1c36269750814a19a28","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-15 07:47:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2a721654ad9db71882dab7641d4adaaed6ac74e6","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-15 07:48:33.000000000","message":"Patch Set 3:\n\n(7 comments)\n\nThanks for comments. All are fixed.","accounts_in_message":[],"_revision_number":3},{"id":"8bc636d72d1305d2d7ac3f50f8febaad533145d5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-15 08:48:19.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0709a274ac724b04b275325a1976aa52 : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/63cf5208bce74e169cb6cc0429d4f83e : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bdcbc161644e43b49a054a5aa2d771a7 : SUCCESS in 3m 10s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cfe7a7bbbf414eac96a56fa47b7c0bc9 : SUCCESS in 4m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1e16ab0c14a94730b7d4d363e8dea1fc : SUCCESS in 3m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ce02b38f1db4434b868e0d1f68ee0234 : SUCCESS in 3m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54c0a861bc144968bcc2485716a4f0ef : SUCCESS in 5m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7d794084ddb4b5ba56d138eba1fb395 : SUCCESS in 3m 04s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/29b91eb6c3ea45bda43ee59b581f20ed : SUCCESS in 48m 28s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/02dfc1c9ba0d433bac0c5bb9355312d2 : SUCCESS in 56m 22s","accounts_in_message":[],"_revision_number":3},{"id":"40285a0a6f7d70fcb227dcafeae2edc76cbf5801","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2020-12-15 16:33:39.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nSorry, missed the release note the last time.","accounts_in_message":[],"_revision_number":3},{"id":"f62521708dba84f45a2621927f580eb4ec2553c3","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2020-12-16 05:34:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)\n\nThanks. A couple of suggested changes for the case where a service does not return a Task resource in a 202 response.","accounts_in_message":[],"_revision_number":3},{"id":"5124474efed924ed5ace3734b3ce65846b920bb0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-16 05:47:53.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"50a17e57f7843e191d3d7abe841532f66e1bdfc1","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-16 05:49:00.000000000","message":"Patch Set 4:\n\n(2 comments)\n\nThanks Dmitry for comments. All are fixed.","accounts_in_message":[],"_revision_number":4},{"id":"1365a6755d8ca6a0d6e5f4f68b03200c4e9bd6f0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-16 06:44:19.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"aa41100f640669383a0f53951ef29d14aa7e2572","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2020-12-16 06:45:44.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks Bill for comments. Its fixed.","accounts_in_message":[],"_revision_number":5},{"id":"48a5ff3611940602bb30880fc2334071483dbaa0","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2020-12-16 07:54:06.000000000","message":"Patch Set 5:\n\nBuild failed. For 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems\n\n- dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R740-tempest-dvsm-ironic-sushy-redfish/2020-12-16:00:44/764915/5/ : FAILURE in 1h 09m 00s\n- dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish https://stash.dellemc-community.org/logs/dellemc-poweredge-R7515-R6515-tempest-dvsm-ironic-sushy-redfish/2020-12-16:00:44/764915/5/ : FAILURE in 53m 25s","accounts_in_message":[],"_revision_number":5},{"id":"caec32d0d734ba3856f6c7f4d70bd719cfcb8fa6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2020-12-16 09:48:44.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e881a535852345f287f7fb0090969984 : SUCCESS in 3m 20s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/6323203d06c44cd980a926f6d94af5ad : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3608beaa5af84c479209a7835b73b2ab : SUCCESS in 2m 47s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9ce43866f2a64506ad37630808825ad9 : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c09224ae332b46d2ab6eb2459389c3c8 : SUCCESS in 3m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b5afc3ceaac402c8022648c599d467e : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fd337e932d1441e93a392478593f262 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/610b8ca8528e4d27915d0c32f05dae04 : SUCCESS in 3m 22s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b7ea9d4e5fb34decaa91b6bc04568031 : SUCCESS in 47m 19s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/827cba913ccd43e28cd084846a3ca442 : SUCCESS in 1h 56m 56s","accounts_in_message":[],"_revision_number":5},{"id":"81861a0e8160e244afd749aaa422a7b7540131ff","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2020-12-29 10:48:32.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nthanks for updates, some comments inline.","accounts_in_message":[],"_revision_number":5},{"id":"846c41bc55c4b6a7009b21dffa862b8b7a7931e2","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-05 16:30:31.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"8456c4044bcefb0735b2164a7bab02bdd27722bc","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-01-05 16:38:58.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"51362bce7026ae56489f3dadaafd732fbcdd04e6","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2021-01-05 17:20:24.000000000","message":"Patch Set 5:\n\n(2 comments)\n\nThanks Aija, Julia for comments. See my comments inline.","accounts_in_message":[],"_revision_number":5},{"id":"502c515c15314f0891a09310a4c80a3a115c4057","author":{"_account_id":28128,"name":"Bill Dodd","email":"billdodd@gmail.com","username":"billdodd"},"date":"2021-01-05 17:24:09.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"9952aa5380ab85e5cc031b1f5919282ba430b2f5","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-01-05 17:51:38.000000000","message":"Patch Set 5:\n\n(2 comments)","accounts_in_message":[],"_revision_number":5},{"id":"db93ccf22a8541a9f5619f514b332bde9115c4fb","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-01-05 18:20:36.000000000","message":"Patch Set 5: Code-Review-1\n\ncan this be put on hold/WF-1 until I test this with iDRAC? I plan to do it this week.","accounts_in_message":[],"_revision_number":5},{"id":"67aa272b9f3f0c99b0e9780fb182db530c5af6b9","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-01-07 10:21:55.000000000","message":"Patch Set 5:\n\n(3 comments)\n\nGood news everyone, iDRAC is still working with these changes.\n\nI still have some comments to make the implementation clearer for now and future developers.\n\nFirst, more unit test cases should be added to fill in the coverage gaps that this code has created and also to document the issue these changes are fixing. For example, there are no unit test updates for updateservice.py changes. I\u0027d see that a new test case can be added that mimic iLO behavior. Current JSON examples used in unit tests also don\u0027t depict scenario where URL joining produces meaningful URL.\n\nSecond, is there any value introducing task_uri in TaskMonitor constructor? In the end in ironic it gets assigned to task_monitor anyway. See [1], [2] and [3]. Why not assign to task_monitor at the very beginning in updateservice.py? Otherwise I\u0027m missing why this \"roundtrip\" is necessary, but unit test cases should clarify need for this if there is any.\n\nI have left some inline comments to add more details.\n\n[1] https://review.opendev.org/c/openstack/ironic/+/764914/5/ironic/drivers/modules/redfish/management.py#749\n[2] https://review.opendev.org/c/openstack/ironic/+/764914/5/ironic/drivers/modules/redfish/management.py#956\n[3] https://review.opendev.org/c/openstack/sushy/+/764915/5/sushy/resources/updateservice/updateservice.py#176","accounts_in_message":[],"_revision_number":5},{"id":"63f86383618d3bb8c40319df1229cc5f5d8d4565","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2021-01-12 07:59:27.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"4761e4d1f2ff2e9f71954a2427801b75a4f022a0","author":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"date":"2021-01-12 08:10:09.000000000","message":"Patch Set 6:\n\n(4 comments)\n\nThanks for comments. All are fixed.","accounts_in_message":[],"_revision_number":6},{"id":"dd36a9f3f99cd8562c84376e34c732f85f2186b3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-12 09:10:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9ff8ee1e20549809a60908f9481eb9c : SUCCESS in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d19ec93568bd4536bcd34e7631b5b6f2 : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/81b8fe355c5943f88d0dc98817b72bf4 : SUCCESS in 3m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f9835ea64144b23857fc2608dd130da : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2665c81369284386ac9d7184a5bae8fa : SUCCESS in 7m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb82c1411183400183e14de46f87eba7 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8aafd83326cc4989939e540484a7cead : SUCCESS in 6m 24s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2ae40253e54848bf87feef6ff4dd5db8 : SUCCESS in 48m 04s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/069f3585d3b447c9aa291acadf23045f : SUCCESS in 1h 08m 40s","accounts_in_message":[],"_revision_number":6},{"id":"77c6dd6112b516d8394b0f7cf7344d68b4c26f28","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-01-13 10:26:05.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"02b9d5412adf3550c9b9c2519a2a266afa9f7dce","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-01-14 14:54:31.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9451250cbf8ffa7386c4edea004ec52634ff44b9","author":{"_account_id":7160,"name":"arkady kanevsky","email":"akanevsk@redhat.com","username":"arkady"},"date":"2021-01-15 20:27:36.000000000","message":"Patch Set 6:\n\nlooks good.\nBut need to check for Dell sushy CI results.","accounts_in_message":[],"_revision_number":6},{"id":"746a6c0f9cee1497a3efc83ea52d638ebed1ec82","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-01-16 12:46:48.000000000","message":"Patch Set 6: Code-Review+2\n\n@arkady: This patch has been manually tested by Aija on Dell hardware and has found it good.","accounts_in_message":[],"_revision_number":6},{"id":"a69ec52681f16b924eac7560e017d9327347577f","author":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"},"date":"2021-01-18 12:15:39.000000000","message":"Patch Set 6: Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"fb50c578832570c9dab0666920224a13068456b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 12:16:00.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"1a2c74b8f3b5f063600b5426c6f40775f38a1d70","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 13:24:43.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37ca599173b9429eb291109bd289eafa : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/374882c2c62044819f584353eb3c8213 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/77c85d853ba843e281f3cbbfdcf66845 : SUCCESS in 3m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0abe025738bd4ed4ac8f6e932b13abb6 : SUCCESS in 5m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cea005d57d724bdeb97062fea603fa2e : SUCCESS in 3m 22s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b55a0ecc71244e3589202324125ca1d2 : SUCCESS in 49m 56s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/18a21eba98474a2a9e000bbad03c4943 : SUCCESS in 58m 54s","accounts_in_message":[],"_revision_number":6},{"id":"ce668da8bd4ddd8a8a8a7b5dacf1ddd08e13e5b8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 13:24:46.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":6},{"id":"101a598b929f0d0099024291605463324c88a147","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-01-18 13:26:49.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b98eafe213448f3977fe8831e66ecd1 : SUCCESS in 1m 17s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4637eab592d94b46b7fa6b8651fbdf89 : SUCCESS in 1m 11s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"3dcffb9a76b894eeeb837007f137452beba8439b","revisions":{"d4be3ccb56ba4a4a9ae3560913408872aa81619d":{"kind":"REWORK","_number":1,"created":"2020-12-01 09:28:39.000000000","uploader":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"ref":"refs/changes/15/764915/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/15/764915/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/15/764915/1"}}},"commit":{"parents":[{"commit":"7016cc0f31050ab656e1e26c80bd44ce3e9fd57a","subject":"Merge \"Add support to expose oem_vendors from resource links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7016cc0f31050ab656e1e26c80bd44ce3e9fd57a"}]}],"author":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-01 09:31:24.000000000","tz":0},"committer":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-01 09:34:29.000000000","tz":0},"subject":"Fixes issue of redfish firmware update","message":"Fixes issue of redfish firmware update\n\nThis patch fixes issue of redfish firmware update\nfor ilo5 based hardware.\n\nChange-Id: Ida1d314f77b9f114e572639c7ad46545be1ce962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d4be3ccb56ba4a4a9ae3560913408872aa81619d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d4be3ccb56ba4a4a9ae3560913408872aa81619d"}]},"branch":"refs/heads/master"},"0cddf766360c2e2f1c4711cc48b8dcb30be39f2a":{"kind":"REWORK","_number":2,"created":"2020-12-14 10:15:01.000000000","uploader":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"ref":"refs/changes/15/764915/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/15/764915/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/15/764915/2"}}},"commit":{"parents":[{"commit":"7016cc0f31050ab656e1e26c80bd44ce3e9fd57a","subject":"Merge \"Add support to expose oem_vendors from resource links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7016cc0f31050ab656e1e26c80bd44ce3e9fd57a"}]}],"author":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-01 09:31:24.000000000","tz":0},"committer":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-14 10:21:09.000000000","tz":0},"subject":"Fixes issue of redfish firmware update","message":"Fixes issue of redfish firmware update\n\nThis patch fixes issue of redfish firmware update\nfor ilo5 based hardware.\n\nChange-Id: Ida1d314f77b9f114e572639c7ad46545be1ce962\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0cddf766360c2e2f1c4711cc48b8dcb30be39f2a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0cddf766360c2e2f1c4711cc48b8dcb30be39f2a"}]},"branch":"refs/heads/master"},"b369de2fdedc1eb20636901feb0e035dba81f97b":{"kind":"REWORK","_number":3,"created":"2020-12-15 07:47:52.000000000","uploader":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"ref":"refs/changes/15/764915/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/15/764915/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/15/764915/3"}}},"commit":{"parents":[{"commit":"7016cc0f31050ab656e1e26c80bd44ce3e9fd57a","subject":"Merge \"Add support to expose oem_vendors from resource links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7016cc0f31050ab656e1e26c80bd44ce3e9fd57a"}]}],"author":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-01 09:31:24.000000000","tz":0},"committer":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-15 07:54:04.000000000","tz":0},"subject":"Fixes issue of redfish firmware update","message":"Fixes issue of redfish firmware update\n\nThis patch fixes issue of redfish firmware update\nfor ilo5 based hardware.\n\nChange-Id: Ida1d314f77b9f114e572639c7ad46545be1ce962\nStory: #2008403\nTask: #41339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b369de2fdedc1eb20636901feb0e035dba81f97b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b369de2fdedc1eb20636901feb0e035dba81f97b"}]},"branch":"refs/heads/master"},"f89ee876fdfd127f5125fbbd55848b9bd3036d56":{"kind":"REWORK","_number":4,"created":"2020-12-16 05:47:53.000000000","uploader":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"ref":"refs/changes/15/764915/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/15/764915/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/15/764915/4"}}},"commit":{"parents":[{"commit":"7016cc0f31050ab656e1e26c80bd44ce3e9fd57a","subject":"Merge \"Add support to expose oem_vendors from resource links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7016cc0f31050ab656e1e26c80bd44ce3e9fd57a"}]}],"author":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-01 09:31:24.000000000","tz":0},"committer":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-16 05:53:10.000000000","tz":0},"subject":"Fixes issue of redfish firmware update","message":"Fixes issue of redfish firmware update\n\nThis patch fixes issue of redfish firmware update\nfor ilo5 based hardware by adding a new attribute\ntask_uri to monitor and validate the firmware\nupdate since task_monitor uri does not provide the\nappropriate response.\n\nChange-Id: Ida1d314f77b9f114e572639c7ad46545be1ce962\nStory: #2008403\nTask: #41339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f89ee876fdfd127f5125fbbd55848b9bd3036d56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f89ee876fdfd127f5125fbbd55848b9bd3036d56"}]},"branch":"refs/heads/master"},"913689c3ebdc36c5bad2cb319f2154f8018d0dbd":{"kind":"REWORK","_number":5,"created":"2020-12-16 06:44:19.000000000","uploader":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"ref":"refs/changes/15/764915/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/15/764915/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/15/764915/5"}}},"commit":{"parents":[{"commit":"7016cc0f31050ab656e1e26c80bd44ce3e9fd57a","subject":"Merge \"Add support to expose oem_vendors from resource links\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7016cc0f31050ab656e1e26c80bd44ce3e9fd57a"}]}],"author":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-01 09:31:24.000000000","tz":0},"committer":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-16 06:48:55.000000000","tz":0},"subject":"Fixes issue of redfish firmware update","message":"Fixes issue of redfish firmware update\n\nThis patch fixes issue of redfish firmware update\nfor ilo5 based hardware by adding a new attribute\ntask_uri to monitor and validate the firmware\nupdate since task_monitor uri does not provide the\nappropriate response.\n\nChange-Id: Ida1d314f77b9f114e572639c7ad46545be1ce962\nStory: #2008403\nTask: #41339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/913689c3ebdc36c5bad2cb319f2154f8018d0dbd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/913689c3ebdc36c5bad2cb319f2154f8018d0dbd"}]},"branch":"refs/heads/master"},"3dcffb9a76b894eeeb837007f137452beba8439b":{"kind":"REWORK","_number":6,"created":"2021-01-12 07:59:27.000000000","uploader":{"_account_id":19604,"name":"ankit","email":"ankit.dhn31@gmail.com","username":"ankit"},"ref":"refs/changes/15/764915/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/15/764915/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/15/764915/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/15/764915/6"}}},"commit":{"parents":[{"commit":"e4a3e18de079968f23aa6473750ed017ff43fa4c","subject":"Add doc/requirements","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e4a3e18de079968f23aa6473750ed017ff43fa4c"}]}],"author":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2020-12-01 09:31:24.000000000","tz":0},"committer":{"name":"ankit","email":"ankit.dhn31@gmail.com","date":"2021-01-12 08:06:44.000000000","tz":0},"subject":"Fixes issue of redfish firmware update","message":"Fixes issue of redfish firmware update\n\nThis patch fixes issue of redfish firmware update\nfor ilo5 based hardware by adding a new attribute\ntask_uri to monitor and validate the firmware\nupdate since task_monitor uri does not provide the\nappropriate response.\n\nChange-Id: Ida1d314f77b9f114e572639c7ad46545be1ce962\nStory: #2008403\nTask: #41339\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3dcffb9a76b894eeeb837007f137452beba8439b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3dcffb9a76b894eeeb837007f137452beba8439b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11076,"name":"Shivanand Tendulker","email":"stendulker@gmail.com","username":"stendulker"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
