)]}'
{"id":"openstack%2Fsushy~784516","triplet_id":"openstack%2Fsushy~master~I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478","project":"openstack/sushy","branch":"master","topic":"bios-attribute-registry","hashtags":["ironic-week-prio"],"change_id":"I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478","subject":"Add support for BIOS Attribute Registry","status":"MERGED","created":"2021-04-02 00:20:43.000000000","updated":"2021-05-05 15:45:59.000000000","submitted":"2021-05-05 15:41:06.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":4,"has_review_started":true,"submission_id":"784516-bios-attribute-registry","meta_rev_id":"755147f70f7cc6d015db07c9334afbc57ed23aa2","_number":784516,"virtual_id_number":784516,"owner":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-05 15:41:05.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2021-05-04 17:39:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-05 14:40:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-05 14:40:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-04-02 00:56:58.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-04-02 01:24:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-04-02 01:29:01.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-04-07 11:27:13.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2021-04-14 11:58:45.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2021-04-20 11:39:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2021-04-22 01:23:42.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2021-05-04 17:39:54.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"40c64f1b2b29711c4077aad26f2dd0ac6412ba2a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-02 00:20:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"51acb24c1964642f9dc0a460f8509d031755aa0c","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-04-02 00:25:19.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"cc16343195cb9ff3fefe8d5f8bc79703b0d376f5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-02 00:56:58.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/2 : SUCCESS in 35m 50s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/5 : SUCCESS in 32m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"3927991c557e3bca996e8e4a5ba07611ea313eb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-02 01:24:18.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ca5064325bd24ba585c84dc6f5059cd1 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18848749537c4307885ff734020a55d3 : SUCCESS in 3m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/14c9b6915fb14b9d89f4b74644d72e1a : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f5491ba21c05431c8cbfd85b19938038 : SUCCESS in 3m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/712cc206cfdb4a43a1ddd87305354b31 : SUCCESS in 3m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf5644ada4a645668502f3b9f31caf41 : SUCCESS in 7m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73141befed1e4127b4761e003dc587e4 : SUCCESS in 3m 27s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2a1fa682760949cdb71e067d2ef9b2be : SUCCESS in 57m 48s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6b08df4853274fdda9ed38a6893c0d1d : SUCCESS in 56m 17s","accounts_in_message":[],"_revision_number":1},{"id":"cdc95917f8a5179e89431b5d0d6d2495796cafef","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-02 01:29:01.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/54ba0aa/54ba0aa.html : FAILURE in 1h 06m 35s","accounts_in_message":[],"_revision_number":1},{"id":"4e575265bc24b656672d8712e5849315f6e39d8c","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-07 11:27:13.000000000","message":"Patch Set 1:\n\n(1 comment)\n\nOne quick comment, haven\u0027t fully reviewed.","accounts_in_message":[],"_revision_number":1},{"id":"6cdc59312837345b81cb13001e5803bdf89e5780","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-04-14 12:46:05.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"21687374426a8059761084612ddd25c8374b8a4a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-15 00:23:51.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"a57b45974a1783488fb22a3f66d4a0af843b49ff","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-15 00:28:47.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b24e48f8506315453f44e2398b1d44b39c6114a3","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-15 01:01:44.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/5 : SUCCESS in 37m 29s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/8 : SUCCESS in 32m 12s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"95a0d88fd9c62f4c2035c6f0a45a8c0ed1a913fd","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-15 01:02:01.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/7b6187c/7b6187c.html : FAILURE in 34m 43s","accounts_in_message":[],"_revision_number":2},{"id":"aa90694819230bc60eddf6690a68a9d89269a813","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 01:27:43.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/28a27ab787d5439eb2db10eb7a2de2e9 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/67329b4263424e2ca200391bb29f7de0 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1a06512d71e84b32bfe878b9c919cb2a : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/415c3be7d4684070abaa2270162a8dde : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39b144035b7f490dbd66c30e7ca83d9a : SUCCESS in 3m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3baeb6b3c4244700a3573b119b74527b : FAILURE in 5m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b7ed999db9c14a7898b0792b075dc4cd : SUCCESS in 3m 40s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/98a1edd9b3134bd6979dc66a62282f5c : SUCCESS in 48m 46s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/98644d4e448d4266a760e996f80887b1 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"12258cc8d223e9bf166c11b387cd16d99628d048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-15 12:06:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2b18118cfd96f177176022b8781d56b296074182","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 13:04:17.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/17dade64aef242eb8b12ca024c2e8565 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/27d3d9b0d85a42798f0f2c9a2ba9def9 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0bf4c0da670e4ff697459acf7db5eaf7 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/95b7b2dc6a3c450db7e1531e0e4cc573 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/14ca791713d046fb99495793892a4012 : SUCCESS in 4m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/440193784af840a2a5306309af0597ef : FAILURE in 4m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2e983fda26ad48d49f8451dbbe16f37f : SUCCESS in 3m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d8e5555cce0047be805725fe5fab8fb8 : SUCCESS in 49m 10s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8c192990ffd645269545ba2c3da929fe : SUCCESS in 52m 24s","accounts_in_message":[],"_revision_number":3},{"id":"36390ab04602910dc848a6748d19684228d654cb","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-15 13:17:17.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/53e1368/53e1368.html : FAILURE in 1h 08m 14s","accounts_in_message":[],"_revision_number":3},{"id":"fd900954c2a566294bf7865b1dc5c9cbbdc62cb9","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-15 13:50:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/6 : SUCCESS in 1h 42m 56s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/9 : SUCCESS in 1h 40m 59s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"868700c7e23e25ac76c05b160b49eacf47bc76bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-15 13:55:15.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"d2a886c6ffc4e4c9149aa674904a133624e5f1c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-15 14:53:19.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e25b487bd8fa49399bb879cdd98c52ea : SUCCESS in 4m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9369aee720fe49b18f428a22825bf8f1 : SUCCESS in 3m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4b3f19e43649488aaf07df302afcbb35 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/113911e85aca4f2a97a591f57b48accc : SUCCESS in 4m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc431d893b6c42c3b12615a094df6f43 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/623644d714c247318f926f269506a453 : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/33e14c16565a4ebd9401c05f4451cfc3 : SUCCESS in 4m 34s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0db4f9e0e2eb40e8a90619fe207731d0 : SUCCESS in 50m 44s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a3834e6d82844183a45d0e5ea9a76458 : SUCCESS in 51m 29s","accounts_in_message":[],"_revision_number":4},{"id":"9a88a19475321d80769971fb54d64b534218bf09","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-15 14:58:27.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/4/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/7cd15be/7cd15be.html : FAILURE in 1h 01m 28s","accounts_in_message":[],"_revision_number":4},{"id":"89e852b58504fc5880de6c44bfe5a11a821c6681","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-15 16:35:51.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/7 : SUCCESS in 2h 40m 11s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/10 : SUCCESS in 2h 17m 19s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"142c85b3f2b2c26584c3020cb7bc8e80c7ac0cad","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-04-16 08:27:14.000000000","message":"Patch Set 4:\n\n(5 comments)\n\nthanks for updates, some ideas inline.","accounts_in_message":[],"_revision_number":4},{"id":"7083a9c345ac20eaa9d0bac4d666132f9ac14c22","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-17 13:03:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"50368c5952f0cf755ba2e406c0b1f01766578b96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-17 13:16:43.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"666c55e6f76ab602a894e954471ffad473f00f1c","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-17 13:17:48.000000000","message":"Patch Set 6:\n\n(5 comments)\n\nThanks for the review Aija! Comments below.","accounts_in_message":[],"_revision_number":6},{"id":"714206a01017738b764b5af062df2c43fcdc102b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-17 13:40:41.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/8 : SUCCESS in 36m 41s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/11 : SUCCESS in 32m 1s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"eec8765f0d668e89f477142961431e0fd0ffc842","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-17 14:06:37.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d19bfd1960d44d90b9ebca4e0554cf3d : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/619e3064cc724f85b26a8668c503a29b : SUCCESS in 3m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b6db4b8e8b5a40d3bf7f5152cbb667a6 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c87f270a91c4c0eaf826e1028e7a14c : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a714ababb52c44948bef27a9843f9bfd : SUCCESS in 3m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b540846898da4ea9ab8b3f1e7dabab4b : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/362b560b58f04bcb9c77cb62a902556e : SUCCESS in 3m 52s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4c06cff63b704f44bb67992af96db790 : SUCCESS in 47m 17s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f49fa0065d554680818f5ac110ac2bbc : SUCCESS in 47m 44s","accounts_in_message":[],"_revision_number":6},{"id":"139231bf3305831594cc31a903dd92493e6c76f5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-17 14:18:43.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/9 : SUCCESS in 37m 39s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/12 : SUCCESS in 31m 57s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"fe89828ca3884fa61a4213f73b7d5be3dcf4ca84","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-17 14:19:13.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/6/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/9a3b0e4/9a3b0e4.html : FAILURE in 1h 00m 43s","accounts_in_message":[],"_revision_number":6},{"id":"725d8d7ef972ad8a4545641d409c4e4f4d8a5d04","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-04-19 10:12:28.000000000","message":"Patch Set 6:\n\n(3 comments)\n\nthanks for updates, LGTM.","accounts_in_message":[],"_revision_number":6},{"id":"6cca6e515c054427104dba490d0ab19a3c472775","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-04-19 10:12:41.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4b9b34f8e6cb3910c06b694019528f625e6e5c80","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-19 19:44:14.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"31c7607ae6cb2914f20028b9dc9f6b4500b1f077","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-19 19:46:17.000000000","message":"Patch Set 7:\n\n(3 comments)\n\nThanks Aija.","accounts_in_message":[],"_revision_number":7},{"id":"28335b9c45bf4f9237c8d6c946f8f96e2ef6e439","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-19 20:52:45.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0873c8eb12434ad2aa04ded6076458d4 : SUCCESS in 4m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c0fefb0ea4dc426bbb5e6d6ac830bf8a : SUCCESS in 4m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/67c3990566a5402aa73a63869070a222 : SUCCESS in 3m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d8b7e79c2ac4826b68a860d343d8e4d : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/422fae2a7b054ed3ab75a7fa12696451 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20525a8c829a476eb362377bd122672b : SUCCESS in 6m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/907b912a125142f9add02d86ebfc2954 : SUCCESS in 4m 07s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ea8da1b834dc448eb387015f4b83a922 : SUCCESS in 1h 03m 16s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/cce61f24ff9a4e77a27dc0ea04a1e5c8 : SUCCESS in 51m 07s","accounts_in_message":[],"_revision_number":7},{"id":"f3fdb22eea8a922aa9b56f390018a62718719362","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-19 20:54:01.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/7/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/707b98d/707b98d.html : FAILURE in 1h 01m 30s","accounts_in_message":[],"_revision_number":7},{"id":"e883e41ce4701dcdd72f2b8fc2674fa3b8b6e422","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-19 20:59:45.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/12 : SUCCESS in 1h 15m 7s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/15 : SUCCESS in 1h 6m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":7},{"id":"10548b62414c97ef5efc1b3b4f7a1a5555adef96","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-04-20 08:06:53.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"1177ce01721168060408e5a7ee5112b652cf313f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-04-20 11:39:39.000000000","message":"Patch Set 7: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":7},{"id":"16451efb4af1901bab97468d130847e8f88d9c1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-20 17:48:48.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"f1703a66a6a5dbe13660abd45d4f399c496351a2","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-20 17:56:04.000000000","message":"Patch Set 8:\n\n(4 comments)","accounts_in_message":[],"_revision_number":8},{"id":"4dbbe9e1ff3f9f18d8c7205aba0468def73d3e2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-20 18:48:37.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f6c09578c0504aaf9c30f36b36e0145c : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68b0c108f5664881ae0920eb1ec79d93 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0b387680cac54f04a4d4896f6c423a78 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26702398db594185b8b053a85158c217 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/998a20dbbba046f8ba524a0f400a68d3 : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fd3e615bb47d448abf8a79ab8bebef4c : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2ac5f2636684d1380848a7f763fed50 : SUCCESS in 4m 51s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d5a4e5c6ce9f47928a723f6d15fbd9c6 : SUCCESS in 56m 41s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2f7093b0278c4c8fbee7b7207c32195a : SUCCESS in 41m 37s","accounts_in_message":[],"_revision_number":8},{"id":"8358e81e412d36110fb4f7d9766e720e1f98986b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-20 18:58:27.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/8/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/150723f/150723f.html : FAILURE in 1h 01m 31s","accounts_in_message":[],"_revision_number":8},{"id":"4a7b836b939c72c919069cd905950491615f9b2c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-20 19:07:30.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/16 : SUCCESS in 1h 18m 21s\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/13 : SUCCESS in 57m 34s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":8},{"id":"7f3b1def4d617dd9238110d43788e6394acd8f7e","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-04-21 10:58:40.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)\n\none comment from previous version inline","accounts_in_message":[],"_revision_number":8},{"id":"6153e78655b9b4676beb203ffe0c782e8bca6914","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-21 12:26:32.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"8df0a1cbef139217e33e85ce389d29df9495e97b","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-21 12:27:20.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"290ddc9bf642d59ba0344b10b63e0e4e85f0e8ba","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-21 13:17:26.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc61c14c94d549aea35437faf6c4ccaa : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97cd300bd70e4a1bab14dbba0a3df675 : SUCCESS in 5m 11s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3a43a4be04b04ca29e9a436240782420 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c54d60b6a6c04c2eb8d54abcc9dd94f1 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/18768ee2978f4bc3a903d95c6ff767c9 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a293b2873bc44924997a9de4e21fa98e : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5dfc4c1e66e94a2fb7df6341d0545d0d : SUCCESS in 5m 10s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c9230b2f1937478b984e83909fd3f6b3 : SUCCESS in 46m 32s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/74797ee801b64528aabfccdbfd78d758 : SUCCESS in 46m 40s","accounts_in_message":[],"_revision_number":9},{"id":"d00cd9de5c2b0a4c617a57936ace07f802c077fb","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-21 14:01:57.000000000","message":"Patch Set 9:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/9/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/be1b5f1/be1b5f1.html : FAILURE in 1h 02m 51s","accounts_in_message":[],"_revision_number":9},{"id":"12ca026d33a19d27a5c365121a991c4b3e5fd39f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-21 14:04:49.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/17 : SUCCESS in 1h 37m 48s\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/14 : SUCCESS in 1h 36m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":9},{"id":"fc8ab47658b7412d1ff8cdba0dcd156d1e54a663","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-04-22 11:40:27.000000000","message":"Patch Set 9: Code-Review+1","accounts_in_message":[],"_revision_number":9},{"id":"73b2e56316abafb8e7961ac3c2df68581c8d2851","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-04-27 14:06:25.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"3b9e09f202478751f990cd2b036a92051e6ed601","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-04-27 15:10:50.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4fc5d75cdb0848deaa5b27ad62739a23 : SUCCESS in 4m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5fcb5340559f46db80f3ca9694453102 : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a33f916214454233a8a144063f7ddaf2 : SUCCESS in 4m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d160b6b9f184c18b216d5cd7e2bd899 : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/73104a7319484057b087a3a353c2dee7 : SUCCESS in 3m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c305938c46c4b81a20b3c92bce156a0 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a084c94c5dec45218986e1e85b4f320a : SUCCESS in 3m 46s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/03c17fc215b64d68a24b9e6f3cd4024b : SUCCESS in 48m 50s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e6d84b04f4d340afafedb29b9d20d745 : SUCCESS in 52m 16s","accounts_in_message":[],"_revision_number":10},{"id":"637d8f919bcd142ba85d56741c90b1c09aac1121","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-04-27 15:15:03.000000000","message":"Patch Set 10:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/16/784516/10/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/3840989/3840989.html : FAILURE in 1h 03m 01s","accounts_in_message":[],"_revision_number":10},{"id":"aa7683dc7399d58dcfa526541a3f3fffcdce2891","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-04-27 17:54:29.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/19 : SUCCESS in 36m 53s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/22 : SUCCESS in 32m 46s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":10},{"id":"b96a3208f3feb5e4919d5f63bd4e125f42a8d583","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-05-04 17:39:54.000000000","message":"Patch Set 10: Code-Review+2\n\nLGTM Thanks Bob!","accounts_in_message":[],"_revision_number":10},{"id":"767bae847b587cb702043a9269023ba6753f1418","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-05 14:40:48.000000000","message":"Patch Set 10: Code-Review+2 Workflow+1\n\nThanks Bob ✔","accounts_in_message":[],"_revision_number":10},{"id":"05a66a8ed03e89245947fd7183d72932e257b404","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 14:41:18.000000000","message":"Patch Set 10: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":10},{"id":"ff822040e1e7bce7600a38d34fa980f72401f07c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:41:05.000000000","message":"Patch Set 10: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2761f8b0046b4293bdae8fa87a1208db : SUCCESS in 5m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e24950b3aae2459faf3ec8accc32ae89 : SUCCESS in 4m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40e957755645476bb8f1ff1692e4090c : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7d4b0c70b3245efab9128600c6b6018 : SUCCESS in 6m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/814b650df935479e95f1840c4a7418bf : SUCCESS in 4m 44s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6f9a15d8e0b04778a4e138e17c3cf81e : SUCCESS in 39m 37s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/030cda09e8c548f6869d69741638827e : SUCCESS in 58m 36s","accounts_in_message":[],"_revision_number":10},{"id":"07ac5c118b431ee0a448804765a975ce40adbf41","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:41:06.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":10},{"id":"755147f70f7cc6d015db07c9334afbc57ed23aa2","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-05 15:45:59.000000000","message":"Patch Set 10:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/969f1d4f35454645ad6dead210aa8d73 : SUCCESS in 1m 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8dcabbbcafee414ea097752d1e84f6d5 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":10}],"current_revision_number":10,"current_revision":"2d88bc9d862b729195841f6909dbe49fc03ad3d1","revisions":{"ec6882a626e6c98e6b8765124df0cd22b799da29":{"kind":"REWORK","_number":1,"created":"2021-04-02 00:20:43.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/1"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:20:34.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ec6882a626e6c98e6b8765124df0cd22b799da29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ec6882a626e6c98e6b8765124df0cd22b799da29"}]},"branch":"refs/heads/master"},"2263d9de27beff2bdcf07e539235c4374b7d5ea8":{"kind":"REWORK","_number":2,"created":"2021-04-15 00:23:51.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/2"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-15 00:23:26.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2263d9de27beff2bdcf07e539235c4374b7d5ea8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2263d9de27beff2bdcf07e539235c4374b7d5ea8"}]},"branch":"refs/heads/master"},"e98b68e09c68d989b5bce648a9f719f0b8233f97":{"kind":"REWORK","_number":3,"created":"2021-04-15 12:06:46.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/3"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-15 12:06:38.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e98b68e09c68d989b5bce648a9f719f0b8233f97"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e98b68e09c68d989b5bce648a9f719f0b8233f97"}]},"branch":"refs/heads/master"},"eb058bae1e6d439354b674581f5588677c3681f7":{"kind":"REWORK","_number":4,"created":"2021-04-15 13:55:15.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/4"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-15 13:55:07.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eb058bae1e6d439354b674581f5588677c3681f7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eb058bae1e6d439354b674581f5588677c3681f7"}]},"branch":"refs/heads/master"},"4911b317514e17c2dc41a3c6890c191eeb9fbdbf":{"kind":"REWORK","_number":5,"created":"2021-04-17 13:03:37.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/5"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-17 13:03:11.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4911b317514e17c2dc41a3c6890c191eeb9fbdbf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4911b317514e17c2dc41a3c6890c191eeb9fbdbf"}]},"branch":"refs/heads/master"},"26f36ca5797504cee7398d89c23ebc4ddbbb8a1a":{"kind":"REWORK","_number":6,"created":"2021-04-17 13:16:43.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/6"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-17 13:16:33.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/26f36ca5797504cee7398d89c23ebc4ddbbb8a1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/26f36ca5797504cee7398d89c23ebc4ddbbb8a1a"}]},"branch":"refs/heads/master"},"fd4ceae45fdc4f5227c812557c6681c077c826c4":{"kind":"REWORK","_number":7,"created":"2021-04-19 19:44:14.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/7"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-19 19:44:04.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fd4ceae45fdc4f5227c812557c6681c077c826c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fd4ceae45fdc4f5227c812557c6681c077c826c4"}]},"branch":"refs/heads/master"},"66de4fd8ce63791a93f75e0dd9048fbb362a5aab":{"kind":"REWORK","_number":8,"created":"2021-04-20 17:48:48.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/8"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-20 17:48:40.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/66de4fd8ce63791a93f75e0dd9048fbb362a5aab"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/66de4fd8ce63791a93f75e0dd9048fbb362a5aab"}]},"branch":"refs/heads/master"},"e898a0259f5fef32aed708095960458876e8771d":{"kind":"REWORK","_number":9,"created":"2021-04-21 12:26:32.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/9"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-21 12:26:21.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e898a0259f5fef32aed708095960458876e8771d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e898a0259f5fef32aed708095960458876e8771d"}]},"branch":"refs/heads/master"},"2d88bc9d862b729195841f6909dbe49fc03ad3d1":{"kind":"REWORK","_number":10,"created":"2021-04-27 14:06:25.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/16/784516/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/16/784516/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/16/784516/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/16/784516/10"}}},"commit":{"parents":[{"commit":"c0da841700ca685a42161082fe4ef43604a560e3","subject":"Add Python3 xena unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c0da841700ca685a42161082fe4ef43604a560e3"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-02 00:18:26.000000000","tz":-240},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-04-27 14:06:11.000000000","tz":-240},"subject":"Add support for BIOS Attribute Registry","message":"Add support for BIOS Attribute Registry\n\nAdd support for caching Attribute Registries. In particular,\ncache the BIOS Attribute Registry and provide a function to return it\nif it matches the AttributeRegistry field in System BIOS.\n\nChange-Id: I1b83cfd9d9a44ac00e423c589bcc2c53c14ad478\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2d88bc9d862b729195841f6909dbe49fc03ad3d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2d88bc9d862b729195841f6909dbe49fc03ad3d1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
