)]}'
{"id":"openstack%2Fsushy~792053","triplet_id":"openstack%2Fsushy~master~I8991bed87cea60e2834b05b1af879a146b12e04c","project":"openstack/sushy","branch":"master","topic":"eventservice","hashtags":["ironic-week-prio"],"change_id":"I8991bed87cea60e2834b05b1af879a146b12e04c","subject":"EventService support","status":"MERGED","created":"2021-05-19 00:26:57.000000000","updated":"2021-05-28 10:20:51.000000000","submitted":"2021-05-28 10:19:13.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":71,"unresolved_comment_count":11,"has_review_started":true,"submission_id":"792053-eventservice","meta_rev_id":"947b1bcef2e35b1d861da40aa53cbedfeb278664","_number":792053,"virtual_id_number":792053,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-05-28 10:19:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":1,"date":"2021-05-27 08:40:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2021-05-27 14:08:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-05-28 09:19:48.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-05-28 09:19:48.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-05-19 00:41:31.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-05-19 01:31:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-05-20 18:06:44.000000000","updated_by":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"reviewer":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"state":"REVIEWER"},{"updated":"2021-05-20 20:03:55.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-05-21 13:47:52.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"CC"},{"updated":"2021-05-24 08:17:58.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2021-05-27 14:08:01.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-05-28 09:19:48.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"4d9abf8828cba369c9f132211777fa7774411f7d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-19 00:26:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5382d37e86e6699206da87006bdb1bf20d371c35","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-19 00:41:31.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/53/792053/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/b3498fc/b3498fc.html : FAILURE in 12m 05s","accounts_in_message":[],"_revision_number":1},{"id":"bad421d26dddbd15c61ca075ea9ce5f437b04234","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 01:31:21.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9be9834b1eb0459eb155058630c3596c : SUCCESS in 4m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e008f1d783844432841cfea5e0d87839 : SUCCESS in 3m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ff5556d901d14a34aa747f135cf9a974 : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e603f37f617b4a77a3d20f023b41dac0 : SUCCESS in 3m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2bfd9a94dfa48f1835dc0e91e3e3d39 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e89b0f708dd4da0a7717f53d08f89e1 : FAILURE in 4m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19f58ad4b7304551b4491a45519da4fb : SUCCESS in 4m 08s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b1a93f8ee7e547d3a5ef5e9c1de22f20 : SUCCESS in 1h 01m 40s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/33c29448a281468dbb1ac8173e568855 : SUCCESS in 57m 31s","accounts_in_message":[],"_revision_number":1},{"id":"26ef1cf53a2646c2b4570264fb7b8a0d82997d65","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-19 07:25:36.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"24a5cb4a62b8b6c155f7d29f3c910f3c91b1f308","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-19 07:25:59.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"5125561e0810aa7cb21252e792b644683d4ff025","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-19 07:50:28.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/53/792053/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/74464cf/74464cf.html : FAILURE in 19m 38s","accounts_in_message":[],"_revision_number":2},{"id":"d83027e7b77597d50fd46a2f552e2301f1d56f11","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-19 08:18:20.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10b0a531111d4cf5b32514d97d545873 : SUCCESS in 4m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b3ebaff73c9045f589895eb61c2e22b6 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a433a2ec51ef46e5965b1a6e617ca2a7 : SUCCESS in 3m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/75024736894a4c6a8389d879840ebb89 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2cd1c3a16cb64e3ab91c025d6ae42111 : SUCCESS in 3m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ca26bc19b1e4c3a830121c7eef848b8 : SUCCESS in 5m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/693520b55bab4dbf95d1a1951c3adb33 : SUCCESS in 3m 43s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6ba28bc9ef8442efb62b55eba5179f5e : SUCCESS in 46m 42s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c92bc35d3aff454288a97ca5be57ed64 : SUCCESS in 48m 38s","accounts_in_message":[],"_revision_number":2},{"id":"9619678eb425c0c7d1c493ff585dbbc1c1a07cd8","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-20 18:06:44.000000000","message":"Patch Set 2: Code-Review-1\n\n(6 comments)\n\nLooks good, just a couple things below.","accounts_in_message":[],"_revision_number":2},{"id":"b03232a7c8a471f18a98f8452388d0310803fd4b","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-20 18:15:09.000000000","message":"Patch Set 2:\n\n(5 comments)","accounts_in_message":[],"_revision_number":2},{"id":"aafb13a77b4dd620eaf60dc3f0bde60c93b0d80f","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-20 18:17:20.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"526cc66f1b19b8a0ac5097762dcdfcf19cee36d7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-20 18:23:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"0b4eb34382d00f41f95af5c833d46b00a93417b7","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-20 18:31:59.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4dedd1dee4563bc55d71487588ac6d2c29b17da4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-20 19:15:55.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5ba16a357e9c45f58870b9a9eb3065ad : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da955b16486c4434bc7c46336ec24de5 : SUCCESS in 3m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/714065e616a0448690babb52135a02af : SUCCESS in 4m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f779244230d04d16b4b5871c0f717627 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2c0f17f5ec0c49bc94e2095f5442205b : SUCCESS in 6m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f20eb65e173e4b08815fc4d4c91e18e8 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/38fec84ac5874471abb4f175fb3a1ee6 : SUCCESS in 3m 36s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fea6f0ed98f849c48d8471008b201c63 : SUCCESS in 47m 07s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/282e529a1846468baccda0a34d0e4a29 : SUCCESS in 51m 47s","accounts_in_message":[],"_revision_number":3},{"id":"3d1d04c3c0c42a026f268734b6df8891094142ac","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-20 19:18:21.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/53/792053/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/7ddedb1/7ddedb1.html : FAILURE in 37m 06s","accounts_in_message":[],"_revision_number":3},{"id":"5a138cbd7b57994dc9889f44604d175d7bedbee6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-20 20:03:55.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/28 : SUCCESS in 1h 40m 20s\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/25 : SUCCESS in 1h 34m 15s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"97cd503c4c38f884e15586ccd730a591220b567d","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-05-20 20:13:24.000000000","message":"Patch Set 3: Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"d9a857868245ceccf442bb3a329186271793151b","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-21 13:47:52.000000000","message":"Patch Set 3:\n\n(10 comments)\n\nthanks for the patch, haven\u0027t gone through till the end yet, but leaving some comments now.","accounts_in_message":[],"_revision_number":3},{"id":"d82eedf4c4a038fb2caa9dd78b3bbf6743037dbd","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-21 14:10:13.000000000","message":"Patch Set 3:\n\n(6 comments)","accounts_in_message":[],"_revision_number":3},{"id":"defa5c2a944054c58decd250d5de539197a212b2","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-21 14:40:17.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4c5cffc84d34fefe7e3974dee7483f063b094f81","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-21 15:03:29.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"608d588067cb41edfb0c81df0954d8b4b55576ae","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-23 16:29:10.000000000","message":"Patch Set 3:\n\n(4 comments)","accounts_in_message":[],"_revision_number":3},{"id":"f15a76a50b11f6dd2aaa10547f94158ee741bf65","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-23 22:42:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"88b918c0de5a3aea8f84703123f2f785d833c080","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-23 23:05:48.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 finger://HPE-CI/013a367e70614a61991daa1bfb59e552 : RETRY_LIMIT in 1m 04s","accounts_in_message":[],"_revision_number":4},{"id":"337612c152e319ac493cd391a13238380b096c7a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-23 23:26:49.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/26 : SUCCESS in 44m 4s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/29 : SUCCESS in 38m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"20b51b19cf2743d757a391012665d197b634e2fb","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-23 23:27:18.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dd637ca23aae4681b933a09c351b97ef : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d84e6709a09d415c88297df888c7a27d : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7ada8e121df145b68993512fc5e731db : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/194895900fac4d3b824b155f1d477d7e : SUCCESS in 7m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a859c59c241146d18b456422c1052c50 : SUCCESS in 3m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38fdc68c575c4c25ab9079c52b2916a5 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29ced7d9adf64f74bba8cb87d661d016 : SUCCESS in 3m 23s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c415857df4744aa78149f01fc23d8237 : SUCCESS in 41m 22s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a8dd803487864de98e98cdd0e090fb8b : SUCCESS in 44m 06s","accounts_in_message":[],"_revision_number":4},{"id":"19da2b4a71312eebf634c3ad658f35a76a08fbce","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-24 08:17:58.000000000","message":"Patch Set 4: Code-Review-1\n\n(7 comments)\n\nthanks for updates","accounts_in_message":[],"_revision_number":4},{"id":"1afb7c830f98ffc40c10d21cb912d46e7dacda47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-24 09:43:17.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"0b5f9e8027b68c02eb34c481c55206ba318e4c82","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-24 10:27:34.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/27 : SUCCESS in 43m 53s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/30 : SUCCESS in 38m 23s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"62e8bc2508545f8ee5deb9df9fa71c595df77b2f","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-24 10:32:01.000000000","message":"Patch Set 5:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/53/792053/5/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/1c39edf/1c39edf.html : FAILURE in 45m 55s","accounts_in_message":[],"_revision_number":5},{"id":"17677c4604e0d6fdbf1132b5b36ce18c10f2a127","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-24 10:35:04.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6cee9879df84c4aa79a1e27a732bd9f : SUCCESS in 4m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fe54e98194b047dc8de98cec10486a1f : SUCCESS in 3m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7426c241060f4b3cb7470de0f1bcf2e7 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9a31d70debaf42b6a20254ddf9e0f9d3 : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6f685ec75304413596b169cae27e7eb3 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c57d21d50154488b3cc3c04e0c769a0 : SUCCESS in 6m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0d2aa3f7e9aa44baa16770945f84a105 : SUCCESS in 4m 08s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3e210db663254ed0bf3f74e4cb9be6f0 : SUCCESS in 45m 46s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/aa50bfd6ade3452faf08e59344a68791 : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":5},{"id":"7c754a60c65aadb3e7a63dcb12b810d8a14ed3f7","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-24 14:54:17.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"b1e5c0fdedd8169a5a8702ae28e33f19515fb741","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-25 19:25:53.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"75eba1cdfd22ff0ade7198baf2de5f536c5e0fc2","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-25 19:38:18.000000000","message":"Patch Set 6:\n\nAdded HttpHeaders","accounts_in_message":[],"_revision_number":6},{"id":"17248e5696501a548a3e2c250205959743201fda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-25 20:09:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2a4a78108b5c4806ae390045181cadaf : SUCCESS in 3m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0013c06f819e42d2bfa42ec0151ebd85 : SUCCESS in 4m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/648abfcc80ec4e3e9f4f2797c79720aa : SUCCESS in 4m 21s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47aaeef12e6c419eb2041f48165a38b7 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/52c448bdd76c444d89a751e9809defae : SUCCESS in 3m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bec382e5d5da45b4bac5cce43ca38c93 : SUCCESS in 6m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0acefeff66db406ab449bcdfa5e849be : SUCCESS in 4m 20s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c24c0b8eaffc4196ac489ff6c53dbf52 : SUCCESS in 35m 30s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6aa2f08c1ce04757a0d2002e5b475ceb : SUCCESS in 42m 32s","accounts_in_message":[],"_revision_number":6},{"id":"85cdd6a3be670171b08e106235ff2ebcdfba2478","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-05-25 20:19:59.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/28 : SUCCESS in 44m 18s\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/31 : SUCCESS in 39m 33s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"b0bff790d107d52369752abc5a0803b87fca255f","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-25 20:24:21.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 finger://HPE-CI/2bcb516f8b5441dd8346ae9d1d283e01 : RETRY_LIMIT in 1m 00s","accounts_in_message":[],"_revision_number":6},{"id":"a2bba5e3922d68a641295c3d076133a3f0a01e79","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-26 08:12:07.000000000","message":"Patch Set 6: Code-Review-1\n\n(7 comments)\n\nthanks for updates, looks good, some comments inline","accounts_in_message":[],"_revision_number":6},{"id":"8d6cb9506aec15ee74f2566d89ef9d8665a6fc68","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-26 09:06:05.000000000","message":"Patch Set 6:\n\n(7 comments)","accounts_in_message":[],"_revision_number":6},{"id":"73bbf050d1970af792ed77feb21b5dc9a33ef1af","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-26 09:28:01.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9860f749780bf0624314c4efe4c9b9a64418e231","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-26 10:37:24.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"98786e623b3146e380a9b086910744ae73745e6b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-05-26 10:42:47.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"128c18af169397d83325b7b22f887e80e2827a1a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-26 11:22:06.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b6a1557d0d85458596f5867c7765f377 : SUCCESS in 3m 48s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e07684737ca54951bbca75436c4c3820 : SUCCESS in 3m 27s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b0aa872e9a484b74a761d331df8ab1aa : SUCCESS in 4m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/175dddcc52dd4918a36e65b12c3f6830 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aac723644f64445a9f5aad25cfaa7cca : SUCCESS in 3m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62d07ef188364059a6819840b3a0e562 : SUCCESS in 7m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1fbe94d5a484b509f85e4f3879bc9e5 : SUCCESS in 3m 47s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ae91b39f2f6e431283d054c411f99d33 : SUCCESS in 38m 43s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/84fb865bdf294f6ca5f595adc50268a1 : SUCCESS in 43m 13s","accounts_in_message":[],"_revision_number":7},{"id":"95d7a6216d8450b4383ad961c32d4617e7222d69","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-26 12:56:17.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"68530cc95406383f390b890f8cc878e9241ba031","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-27 07:49:48.000000000","message":"Patch Set 7: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"bd15d41ab06db53bc5d8843b89aef4fef6ef26fe","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-27 08:18:06.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"b753335b9eef8b50c170a22128955511beb3b932","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-27 08:18:25.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"77fe6618116e620ac296491f4adf1c90fb00b803","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-27 08:31:11.000000000","message":"Patch Set 8: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"e4fab0cc07581de4443c2cff2293a639180172de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-27 08:37:34.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"62d8b30f577588702a6d7892050ff29cbc8239e4","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2021-05-27 08:40:27.000000000","message":"Patch Set 9: Code-Review+1\n\nthank you","accounts_in_message":[],"_revision_number":9},{"id":"726bd81f27b76ca00254189fd9434b335a6c86c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-27 09:38:54.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f8fde0da5e7f4d0daaaa959a268284b2 : SUCCESS in 4m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9c95d354e4cc48838236dc2eb631ef5d : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a30a1871e295423f92857233420fc05b : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3eff585229f94ab299eddb4622ae2ee3 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/dc0f000cce534d93afa081c16a5011b0 : SUCCESS in 4m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/32fd164e072341249498f48580ae6dd4 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dacb795c1984422b2f50aaba5c21e38 : SUCCESS in 4m 04s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bc2e9713d70644b48dc610ceabd8e691 : SUCCESS in 49m 15s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ba06c7a0ad6148949ce489b2e0dcf97a : SUCCESS in 51m 21s","accounts_in_message":[],"_revision_number":9},{"id":"e42801dedc5d99b79dd90bbc8946fa1bad99bc7b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-05-27 14:08:01.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"169361d85cb24224b3dd1933b2f2a62f6c8e96fd","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-05-27 15:26:21.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"e07c80b021b5d29c3ad7bf2c9a26f2bd88786c70","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-05-28 09:19:48.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\nLGTM, thanks Iury!","accounts_in_message":[],"_revision_number":9},{"id":"4e6d82e08207ecb01436aab3d16784816c28397c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 09:20:13.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"f056ef96d48d71d20e0963a90a05026ed46a838e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 10:19:12.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d7a81ada29841f1b0580927f826ea8c : SUCCESS in 3m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/350796440dc14aae9cce1c6dc8beb95f : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0ef22e21ad6840b28a1a778a2f75aadf : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18c06ec8d7994f128b2ca4b60dfae4f0 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/97a6484d23ae4a76808ee02eb07fb6cb : SUCCESS in 4m 17s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/931a84ded16442f8ae149a96c34b4c88 : SUCCESS in 51m 16s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fad3427f4bc6436794a1be53a2b3bd0a : SUCCESS in 53m 26s","accounts_in_message":[],"_revision_number":9},{"id":"aab3b411a1344f7973d59835f0a23467ec11369f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 10:19:13.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":9},{"id":"947b1bcef2e35b1d861da40aa53cbedfeb278664","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-05-28 10:20:51.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e64f4dfa44754eb7aa32290be36e3d33 : SUCCESS in 1m 22s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c39cde90ffd14aff9e7a96f507e2b21d : SUCCESS in 1m 09s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"b11e99a8619ea934c8d3399d7de5dfa60cbf5d03","revisions":{"15f9d42c1db3a799bfb90d8331800ce558171c58":{"kind":"REWORK","_number":1,"created":"2021-05-19 00:26:57.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/1"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-19 00:23:11.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/15f9d42c1db3a799bfb90d8331800ce558171c58"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/15f9d42c1db3a799bfb90d8331800ce558171c58"}]},"branch":"refs/heads/master"},"17551d2399c8a8ab7cfca3a9e1b0889c5875229c":{"kind":"REWORK","_number":2,"created":"2021-05-19 07:25:59.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/2"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-19 07:25:08.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/17551d2399c8a8ab7cfca3a9e1b0889c5875229c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/17551d2399c8a8ab7cfca3a9e1b0889c5875229c"}]},"branch":"refs/heads/master"},"dc8bf357124a8117c3664a196e35c6e128c3ebdf":{"kind":"REWORK","_number":3,"created":"2021-05-20 18:23:05.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/3"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-20 18:22:58.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc8bf357124a8117c3664a196e35c6e128c3ebdf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc8bf357124a8117c3664a196e35c6e128c3ebdf"}]},"branch":"refs/heads/master"},"832ddac014cccdc0e138eba5c0dc7fc2a67873d2":{"kind":"REWORK","_number":4,"created":"2021-05-23 22:42:18.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/4"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-23 22:42:05.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/832ddac014cccdc0e138eba5c0dc7fc2a67873d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/832ddac014cccdc0e138eba5c0dc7fc2a67873d2"}]},"branch":"refs/heads/master"},"0f0f7d53bec7be42fce320b8d160efc2ef864371":{"kind":"REWORK","_number":5,"created":"2021-05-24 09:43:17.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/5"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-24 09:43:06.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0f0f7d53bec7be42fce320b8d160efc2ef864371"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0f0f7d53bec7be42fce320b8d160efc2ef864371"}]},"branch":"refs/heads/master"},"f63b94f3fabe3df0ed5bacbfd0c0b8ecf573b8d1":{"kind":"REWORK","_number":6,"created":"2021-05-25 19:25:53.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/6"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-25 19:25:28.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f63b94f3fabe3df0ed5bacbfd0c0b8ecf573b8d1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f63b94f3fabe3df0ed5bacbfd0c0b8ecf573b8d1"}]},"branch":"refs/heads/master"},"356b60a2c330d72e33fcec9a80a70bc4c58eed4f":{"kind":"REWORK","_number":7,"created":"2021-05-26 10:37:24.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/7"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-26 10:37:17.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/356b60a2c330d72e33fcec9a80a70bc4c58eed4f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/356b60a2c330d72e33fcec9a80a70bc4c58eed4f"}]},"branch":"refs/heads/master"},"0157d523a236cf5494b1335e0a2ed8ccf7807dc1":{"kind":"REWORK","_number":8,"created":"2021-05-27 08:18:25.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/8"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-27 08:17:53.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0157d523a236cf5494b1335e0a2ed8ccf7807dc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0157d523a236cf5494b1335e0a2ed8ccf7807dc1"}]},"branch":"refs/heads/master"},"b11e99a8619ea934c8d3399d7de5dfa60cbf5d03":{"kind":"REWORK","_number":9,"created":"2021-05-27 08:37:34.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/53/792053/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/53/792053/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/53/792053/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/53/792053/9"}}},"commit":{"parents":[{"commit":"6ae294b3e013a964bc22fe4e35681c09e57f15ee","subject":"Merge \"Add support for BIOS Attribute Registry\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6ae294b3e013a964bc22fe4e35681c09e57f15ee"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-16 17:23:49.000000000","tz":120},"committer":{"name":"Iury Gregory Melo Ferreira","email":"iurygregory@gmail.com","date":"2021-05-27 08:37:25.000000000","tz":120},"subject":"EventService support","message":"EventService support\n\nThis commit adds support for the EventService in Redfish\n\nStory: 2008366\nTask: 42486\nChange-Id: I8991bed87cea60e2834b05b1af879a146b12e04c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b11e99a8619ea934c8d3399d7de5dfa60cbf5d03"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b11e99a8619ea934c8d3399d7de5dfa60cbf5d03"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
