)]}'
{"id":"openstack%2Fsushy~795559","triplet_id":"openstack%2Fsushy~stable%2Ftrain~I70f7a6ecbbe72cfa7b0a46206de9cebbd5aef23e","project":"openstack/sushy","branch":"stable/train","hashtags":[],"change_id":"I70f7a6ecbbe72cfa7b0a46206de9cebbd5aef23e","subject":"Don\u0027t log ERROR if GET of /redfish/v1/SessionService fails","status":"MERGED","created":"2021-06-09 14:31:52.000000000","updated":"2021-06-10 08:00:56.000000000","submitted":"2021-06-10 07:59:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"795559","meta_rev_id":"85380ec5db69faea3abc15ebc646f01200f7e23a","_number":795559,"virtual_id_number":795559,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-06-10 07:59:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":2,"date":"2021-06-10 06:59:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":1,"date":"2021-06-10 06:59:58.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-09 14:31:52.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"state":"REVIEWER"},{"updated":"2021-06-09 14:34:40.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-06-09 14:51:06.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-06-09 15:28:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1355b622664ce140b8b8a27b5789bd43884f2363","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-06-09 14:31:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be433fbabb49d587b4bd7144fe156f614476da0b","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-06-09 14:34:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"d50889ae728d8fa65218e1998f03d584d040fe5a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-06-09 14:34:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cad9076fa502be1e4927998ade36b421522c8659","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-09 14:51:06.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/59/795559/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/68d5e66/68d5e66.html : FAILURE in 5m 41s","accounts_in_message":[],"_revision_number":1},{"id":"9c45753b755de21cf9a78f99311ad05203a53f2e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 15:28:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d8873e278114ba39f4a53cc4c68712b : FAILURE in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b8ff6c7bd7041f6a8636b0b3916e5ef : SUCCESS in 3m 48s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/aa8d0d13f3dc485ba1594f7b2331527f : FAILURE in 4m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/82f6434cd5f140b898706a380a39868b : FAILURE in 4m 26s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/641ff335c33f4f6c872bb08b8c8ce526 : FAILURE in 4m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0218098f458f4daabaf47abb7acc58b3 : SUCCESS in 6m 56s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/88dde93623ba42f48b88a278184178a1 : SUCCESS in 54m 53s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/168ab2f5f0aa48abb4cd9610b08c779a : SUCCESS in 55m 02s","accounts_in_message":[],"_revision_number":1},{"id":"ee252189dab7445900b521e69b38990d90476654","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"date":"2021-06-09 18:21:27.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e060a808ca77bc59164a9e69e738105e6f8f2fc1","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-09 18:31:21.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/59/795559/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/625db1d/625db1d.html : FAILURE in 7m 45s","accounts_in_message":[],"_revision_number":2},{"id":"b989568b5e973882875de6faf8b6601b4a25d285","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-09 19:29:56.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5adba015120947b2a911d8ac400ed466 : SUCCESS in 5m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/54d86dc218e74616867bfa8739ae664c : SUCCESS in 4m 22s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/099568eecdb74b528b1ce117b85a7952 : SUCCESS in 4m 43s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/18d72229507849a5a5c8c37807f78fa3 : SUCCESS in 4m 23s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/22f21a2fbaef4ed9ba54b6282b8be8e2 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a0295b14abc41f391af6963e175c69f : SUCCESS in 6m 43s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/f4f1002fd733466888e3b70641e14d07 : SUCCESS in 1h 04m 08s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/361bc2d2b51340bebb8ebd34e1d492e6 : SUCCESS in 54m 01s","accounts_in_message":[],"_revision_number":2},{"id":"4c1b7c9142986b3df4e5e9c17ace456ff84d923c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-06-10 06:59:58.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"38572fa29ae44b27514f5136978348498c159bad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 07:00:22.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"d550b0d1902ff5eb9b29504a71ce1f0bcd571e53","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 07:59:38.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dad3186c855d45eeb62f08725428e7a1 : SUCCESS in 4m 21s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/bec2c7ac6af7429ca53d00c2b35570f9 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ee3d24f614464c2487dec5d864658e4b : SUCCESS in 6m 59s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/e0d92c98dae1457998b10b785b53f1ca : SUCCESS in 12m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d46ee3dff9b4e98985bfbd35cb87d71 : SUCCESS in 6m 02s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/0856c95ad8884deb8a1e0e09c40dd769 : SUCCESS in 55m 19s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/6316758ba2fc44c6ac005f7610c0b327 : SUCCESS in 54m 49s","accounts_in_message":[],"_revision_number":2},{"id":"8dc816b8ec28bc882c5c7f7a1925cfacd5b2f5e8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 07:59:39.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":2},{"id":"85380ec5db69faea3abc15ebc646f01200f7e23a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-10 08:00:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd5f06411c44d419cbedaadc0782e59 : SUCCESS in 59s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"039b44a112b16339a5849f415fad5d01cc2b048d","revisions":{"874961cf2675e6df756e1b53e84a9a350e06eba1":{"kind":"REWORK","_number":1,"created":"2021-06-09 14:31:52.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/59/795559/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/59/795559/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/59/795559/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/59/795559/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/59/795559/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/59/795559/1"}}},"commit":{"parents":[{"commit":"e275e01dc258309cc7d0f9292015377484dce3e8","subject":"Remove auth token header completely when error occurs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e275e01dc258309cc7d0f9292015377484dce3e8"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-02-10 02:29:13.000000000","tz":-300},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-06-09 14:29:06.000000000","tz":-420},"subject":"Don\u0027t log ERROR if GET of /redfish/v1/SessionService fails","message":"Don\u0027t log ERROR if GET of /redfish/v1/SessionService fails\n\nSimilar to https://review.opendev.org/c/openstack/sushy/+/747960,\nan ERROR is logged when the GET of this Redfish endpoint fails\nprior to authentication, which causes concern for operators. Log\na debug message instead.\n\nChange-Id: I70f7a6ecbbe72cfa7b0a46206de9cebbd5aef23e\n(cherry picked from commit 07ca7f458b38654946f5e2e8c59fad24c24d61c1)\n(cherry picked from commit 234df21124dc787659dc68f8500031ddb5d2d915)\n(cherry picked from commit f4515bfc666d75fb31a7ccd71cddc041203f0a0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/874961cf2675e6df756e1b53e84a9a350e06eba1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/874961cf2675e6df756e1b53e84a9a350e06eba1"}]},"branch":"refs/heads/stable/train"},"039b44a112b16339a5849f415fad5d01cc2b048d":{"kind":"REWORK","_number":2,"created":"2021-06-09 18:21:27.000000000","uploader":{"_account_id":21909,"name":"Bob Fournier","email":"bfournie@redhat.com","username":"bfournie"},"ref":"refs/changes/59/795559/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/59/795559/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/59/795559/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/59/795559/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/59/795559/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/59/795559/2"}}},"commit":{"parents":[{"commit":"e275e01dc258309cc7d0f9292015377484dce3e8","subject":"Remove auth token header completely when error occurs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e275e01dc258309cc7d0f9292015377484dce3e8"}]}],"author":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-02-10 02:29:13.000000000","tz":-300},"committer":{"name":"Bob Fournier","email":"bfournie@redhat.com","date":"2021-06-09 18:21:26.000000000","tz":-240},"subject":"Don\u0027t log ERROR if GET of /redfish/v1/SessionService fails","message":"Don\u0027t log ERROR if GET of /redfish/v1/SessionService fails\n\nSimilar to https://review.opendev.org/c/openstack/sushy/+/747960,\nan ERROR is logged when the GET of this Redfish endpoint fails\nprior to authentication, which causes concern for operators. Log\na debug message instead.\n\nNote - backport removed log check in test_connector.py as HTTPStatus\nis not supported in py27.\n\nChange-Id: I70f7a6ecbbe72cfa7b0a46206de9cebbd5aef23e\n(cherry picked from commit 07ca7f458b38654946f5e2e8c59fad24c24d61c1)\n(cherry picked from commit 234df21124dc787659dc68f8500031ddb5d2d915)\n(cherry picked from commit f4515bfc666d75fb31a7ccd71cddc041203f0a0f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/039b44a112b16339a5849f415fad5d01cc2b048d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/039b44a112b16339a5849f415fad5d01cc2b048d"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
