)]}'
{"id":"openstack%2Fsushy~797435","triplet_id":"openstack%2Fsushy~master~I4891b89a65567c9ba2057557462a5d3840808ae1","project":"openstack/sushy","branch":"master","topic":"auth-none","hashtags":[],"change_id":"I4891b89a65567c9ba2057557462a5d3840808ae1","subject":"Protect Connector against empty auth object","status":"MERGED","created":"2021-06-22 11:52:11.000000000","updated":"2021-07-06 10:55:14.000000000","submitted":"2021-07-06 10:53:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"797435-auth-none","meta_rev_id":"c6f61e534cb71038825f38da2ef7158051e7ce24","_number":797435,"virtual_id_number":797435,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-07-06 10:53:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-06-23 08:33:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2021-07-06 10:02:10.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2021-07-06 10:02:10.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-06-22 12:43:27.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-06-22 12:52:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-06-22 16:07:56.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-06-23 08:33:52.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-07-06 10:02:10.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"0a15681bee09ac5a55e1a574b09cac7967fc1534","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-06-22 11:52:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1283821ace56dae3130ca1b2009a8055018ba387","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-06-22 12:43:27.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/35/797435/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/7b070be/7b070be.html : SUCCESS in 48m 13s","accounts_in_message":[],"_revision_number":1},{"id":"790657d06e9d5a85829d4413c872178c7e186161","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-06-22 12:52:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/695ca6dc76f94055afd1baa96af9c027 : SUCCESS in 4m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4ca830baefca49c490f72c1609dec841 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0af070586be74adfbe679203f6259bd0 : SUCCESS in 4m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/178cf3899b9c4d2d95c1935e1a715121 : SUCCESS in 3m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2492c48b6cf64bd182d3dd9db0978322 : SUCCESS in 4m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/857952275f7e4a96ac69039d7b2305ea : SUCCESS in 6m 55s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bad1bf35ea5a482a90cb30a35d9c89d8 : SUCCESS in 49m 31s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ee2873a1b03f4e5d9275d9a4b9805a7e : SUCCESS in 54m 05s","accounts_in_message":[],"_revision_number":1},{"id":"282ad62d4e3ff7471c93c16c19d7daade7443b38","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-06-22 16:07:56.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n- 14G_uefi_sushy https://stash.dellemc-community.org/logs/14G_uefi_sushy/38 : SUCCESS in 4h 15m 25s\n- R6515_bios_sushy https://stash.dellemc-community.org/logs/R6515_bios_sushy/34 : SUCCESS in 3h 44m 52s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"af3222c0a3cc591c2093ffdd9224b3acce8865cc","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-06-23 08:33:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"8826752d341a3ef699289ca7a8921021de990999","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-07-06 10:02:10.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\nThanks, Dmitry!","accounts_in_message":[],"_revision_number":1},{"id":"324aa1838aa19f0b9e87efea82bb72e4173f61d5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 10:02:28.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef44022673247d5e6a46ad9d3b34fe5f9e276798","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 10:53:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d5b2f5f848954ff2bc60dbc59d8f0c10 : SUCCESS in 3m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/63a8cfcd803840a5a4dfa3e5d458568f : SUCCESS in 4m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40d373f613f548af81cf9b2d9e275d08 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97582c8efd434270ab2f7776be8efacb : SUCCESS in 6m 58s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3ff219c4528d4d69a5e2771e4aef76bc : SUCCESS in 45m 05s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ba5a46f5913248f8b83e5892d44a5605 : SUCCESS in 50m 16s","accounts_in_message":[],"_revision_number":1},{"id":"3bf2d027d3ddf9ef00a2400de06391b9d032261e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 10:53:50.000000000","message":"Change has been successfully merged by Zuul","accounts_in_message":[],"_revision_number":1},{"id":"c6f61e534cb71038825f38da2ef7158051e7ce24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-07-06 10:55:14.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/178b1a49ed37472abc8173128fd506eb : SUCCESS in 1m 07s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"bf50d7363888f2333bdf0c36d65b4144b66d37d4","revisions":{"bf50d7363888f2333bdf0c36d65b4144b66d37d4":{"kind":"REWORK","_number":1,"created":"2021-06-22 11:52:11.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/35/797435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/797435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/797435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/797435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/797435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/797435/1"}}},"commit":{"parents":[{"commit":"5b38ec591e2ea8df8f49fc956a2aecb7776a6656","subject":"Fix incorrect formatting and a Python 3.10 failure","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5b38ec591e2ea8df8f49fc956a2aecb7776a6656"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-22 11:49:34.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-22 11:49:34.000000000","tz":120},"subject":"Protect Connector against empty auth object","message":"Protect Connector against empty auth object\n\nAuth uses Connector, Connector uses Auth. This convoluted relationship\nresults in connector._auth being unset when an authentication request is\nmade. If it results in AccessError, sushy tries to access non-existing\nconnector._auth. This change protects against it and allows the actual\nerror to bubble up.\n\nChange-Id: I4891b89a65567c9ba2057557462a5d3840808ae1\nStory: #2008661\nTask: #41935\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bf50d7363888f2333bdf0c36d65b4144b66d37d4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bf50d7363888f2333bdf0c36d65b4144b66d37d4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
