)]}'
{"id":"openstack%2Fsushy~815479","triplet_id":"openstack%2Fsushy~master~I4ea7f084334a9125c737d0d5f594f1f8c926f377","project":"openstack/sushy","branch":"master","topic":"certificate-collection","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2021-11-17 11:33:52.000000000","reason":"Removed by Dmitry Tantsur by clicking the attention icon"}},"hashtags":["ironic-week-prio"],"change_id":"I4ea7f084334a9125c737d0d5f594f1f8c926f377","subject":"Add basic support for CertificateService","status":"MERGED","created":"2021-10-26 12:04:18.000000000","updated":"2021-11-18 07:46:38.000000000","submitted":"2021-11-18 07:45:00.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"815479-certificate-collection","meta_rev_id":"55645a778a447c60d1592b68d57750c482c3465a","_number":815479,"virtual_id_number":815479,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-18 07:45:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-11-16 10:40:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2021-11-18 06:44:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-18 06:44:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-10-26 12:47:13.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-10-26 13:11:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-10-26 13:59:16.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-11-16 10:40:53.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-11-18 06:44:17.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"e54807edef6a3e8afee4b077231967b9ec023600","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-10-26 12:04:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"50803059870914349a7cfc2ce92c2001f4ce83b1","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-10-26 12:47:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/815479/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/ba5d97a/ba5d97a.html : FAILURE in 38m 48s","accounts_in_message":[],"_revision_number":1},{"id":"63ea5bcc98c654fed816308d803471184ddc0cbe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-26 13:11:57.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5239c430d3384e668b82880793a5a788 : SUCCESS in 4m 16s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/898bd02654474de2898a3650d9a4ae1d : SUCCESS in 4m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/288ccd01600143adb3330e9c642554e2 : FAILURE in 4m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/716674d9be264b8a947a26fafa55924c : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5611a2af026340dba36cba6107c1a2c4 : SUCCESS in 4m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bfb3a426fd814af2939709483fabaf12 : SUCCESS in 4m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d6bfad1c398f4071b67effa27266b185 : SUCCESS in 6m 41s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/428abfe190254805930ff4feb3e4eedd : SUCCESS in 51m 04s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/baaf1d4cad5d4266b5fa173b14d9a41f : SUCCESS in 1h 02m 26s","accounts_in_message":[],"_revision_number":1},{"id":"6eb454ac738bde46155dd6a997ab479c6f120ef7","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-10-26 13:59:16.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/f343f8b6cc4f4f81afee6af6e7a19bfd : SUCCESS in 44m 02s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/875f822f879e4d4ba8c8e415f04b7bb6 : SUCCESS in 1h 53m 34s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"57a3ea74a39fc9fa9e30f6001d0df482c06f205d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-10-29 12:54:57.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95566545b9a74e4519936bdfbefabf9f8f12c002","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-10-29 13:38:12.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/815479/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/4347570/4347570.html : FAILURE in 30m 46s","accounts_in_message":[],"_revision_number":2},{"id":"6a6b02c0ffa0d8706368843fe4eadcb37d635d68","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-10-29 13:41:13.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/a6ace705e6fe43068aa10e4b1f1872f6 : SUCCESS in 42m 32s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/79b5ad133fe24cac91b18cecbecac422 : SUCCESS in 44m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"3a400df56213f378665dd778ff6069fd2949abfa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-10-29 13:54:46.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87bcb744a0c74a889867da41ae8bc9d5 : SUCCESS in 3m 52s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/2f8aec48fc774b9da4cfbd029d9efb8d : SUCCESS in 4m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87b1d3a0c3754f6a8aca9e7a66a160d9 : SUCCESS in 3m 31s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e5a83476cb7648f5b7fe55aa4fbf7dd4 : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/10cca1c250864f039d00447e88988b8a : SUCCESS in 3m 48s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/219f258ec3e94b8e81562b95f5cc3504 : SUCCESS in 4m 12s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/293ca1f541c0442c85afa152f730bf9e : SUCCESS in 6m 11s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ad02a27f83e6427195f1ad7a3d0e3600 : SUCCESS in 49m 38s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/15891d380516420781122dfd615de868 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"e31ddd77f37016b92174213ea3ebbfdbb1edbdb3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-10 14:43:21.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"66c1140b8d5bd440759c020659b8a39ae67debe3","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-10 15:14:30.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/815479/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/a6df23d/a6df23d.html : FAILURE in 29m 40s","accounts_in_message":[],"_revision_number":3},{"id":"935a1243f34d9c78e5a10ac048438fe59d81ef6f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-10 16:01:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/58a0ff2ff4a443908ce079a87a632647 : SUCCESS in 42m 28s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/d37079a955e7432a89a4a003cffa1fac : SUCCESS in 1h 15m 48s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"3382915c10f765048e8ffd06097ee9245bdf777e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-10 16:53:49.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2899cea6772d459386e9cd57019c0898 : SUCCESS in 6m 25s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/7375da2893f343b5a76b34b0fcb42fa8 : SUCCESS in 6m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c457f2673a82442eb008bdbb54bd3983 : SUCCESS in 5m 08s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/16492612c2e54022ba1d7f63d028ad79 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ee16d3a26894a0481806f7efe097666 : SUCCESS in 7m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/530aed0734934300b554c1c08affd990 : SUCCESS in 6m 36s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/75e84093fbf84c50b97a55f27a1908f4 : SUCCESS in 7m 35s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e1e7c75c7899424fad08f8c419a83749 : SUCCESS in 44m 36s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/aad4fc466f8041f8b7192bc7a3f71d22 : SUCCESS in 59m 04s","accounts_in_message":[],"_revision_number":3},{"id":"2aeb9689b78eb7709e1610f5f331c5f47f872bdf","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-11 13:52:49.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"45c57f6226f21c74a3492e50185ba79c44d488be","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-11 14:24:01.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/815479/4/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/2987d2a/2987d2a.html : FAILURE in 29m 42s","accounts_in_message":[],"_revision_number":4},{"id":"b011dc2e7347d84d7f8c257826fa3cde0f88291f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-11 14:38:37.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/14589aad70dc43359fae8951d9211cdb : SUCCESS in 42m 11s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/b23a52936e3d4134929cfc79e7319e76 : SUCCESS in 44m 08s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"a06d00a6cabc23db0c1d62d6526ba801eb0d4ef9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 14:49:48.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e2fa3712f14419bad7a7d52f42aee4f : SUCCESS in 3m 50s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c807b72001e541349f032b287d4e5fbb : SUCCESS in 4m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dc9b3b46924c4ea096be509e8fd2cc0a : SUCCESS in 3m 20s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/dd0870482eba44659b47e810d376ff7c : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4612d553ea0540db93859c0efa1e0f40 : SUCCESS in 4m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/289b4c9c2b544cf6b8fdbc80b2224257 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/decf409390434abbb8327ea1782786d8 : SUCCESS in 5m 48s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c7a2e965985b4231a33f91d89d78592a : SUCCESS in 50m 38s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e089b6bf92654f2ba97d2290f76b5b7c : SUCCESS in 50m 55s","accounts_in_message":[],"_revision_number":4},{"id":"43d3ec36df4416a2fa0c867b8b0f001ef0f4f797","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-11 16:11:14.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"9fbcfbf7e4fd9db7063a88f4c5809a1c39bc909d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-11 16:15:24.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b3460fb2b817e23d68446a1e7b081fa646a154e4","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-11 16:46:54.000000000","message":"Patch Set 6:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/79/815479/6/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/1e1bfb3/1e1bfb3.html : FAILURE in 30m 10s","accounts_in_message":[],"_revision_number":6},{"id":"f445c27e651b09c1ea621773963458a9c5abb618","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-11 17:03:28.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/7f95ffd087774b75ae88492c15fe12e9 : SUCCESS in 42m 48s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/5aae80420b464f809b3f6c869c828845 : SUCCESS in 46m 11s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"3a6d465e46cf26f205d379b093d29f8bc7d68335","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-11 17:18:23.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/021ded36fd644c7187a136906696cc1e : SUCCESS in 6m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/1d13f666c9cc4b5494b7013b7f639e11 : SUCCESS in 6m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b467504d400747e29a01ca59c7b014ea : SUCCESS in 5m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2b32c7cd17964026bd4668e522175cfd : SUCCESS in 6m 13s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/55fafc8e417b40e3986eb0a8fa9f69ef : SUCCESS in 5m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2e6838aae19c471ea7c791c2d6f5fd6d : SUCCESS in 6m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82d6eb15b3674abcbb91a5a09056283b : SUCCESS in 9m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f02de2baecca4cafa1dc9f75343352f8 : SUCCESS in 7m 37s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bc9d5a03ae8e40c1adffd0750559bd1e : SUCCESS in 54m 34s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b1c280f6d18e4fa68a160bf074364003 : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":6},{"id":"130740a29eb787a9cef42f5bd461ac89a2c7ec74","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-15 15:14:11.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":6},{"id":"fe7ab62fc7676ca12b9efe4a8da4aff1b03bf95b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-11-16 10:40:53.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"f37e4f8fec9d8959e5b3dcb6c64de90090ae7eed","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2021-11-18 06:44:17.000000000","message":"Patch Set 6: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":6},{"id":"3f9bd864d044176ff8d18f740418469306557719","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 06:45:07.000000000","message":"Patch Set 6: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":6},{"id":"14aea299023099618c645edacee1be63ad220fe3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 07:45:00.000000000","message":"Patch Set 6: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/99b3ef14835c4c60943c611a7e35e013 : SUCCESS in 5m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9190044399b430388df46eadb68696f : SUCCESS in 2m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b55053a511814554843b7a6419020777 : SUCCESS in 3m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/72dcd7d82f604a4cb0f66db336ccfc80 : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5d8bdc9ed824db7805260b5d3b1c967 : SUCCESS in 9m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4690b4e53d54c5087f686b9a4196b52 : SUCCESS in 3m 24s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/720865afa8f94f3fb33654dcc9c99766 : SUCCESS in 58m 11s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ed73956a731c41dbb0dc6b6724c689e7 : SUCCESS in 56m 05s","accounts_in_message":[],"_revision_number":6},{"id":"1b06ee692caf7d1c8210ba250a4a95281ab251ad","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 07:45:00.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":6},{"id":"55645a778a447c60d1592b68d57750c482c3465a","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 07:46:38.000000000","message":"Patch Set 6:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46bac0c4bb994ac58d38ff10a4b45efb : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31d29009029d4f59add497ac7a08aac5 : SUCCESS in 1m 01s","accounts_in_message":[],"_revision_number":6}],"current_revision_number":6,"current_revision":"fa2e0015854b2885e6fbc532a0836a01f2ae329d","revisions":{"a882b8e40b3d47fe9d14e9be0026a274802712ce":{"kind":"REWORK","_number":1,"created":"2021-10-26 12:04:18.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/815479/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/79/815479/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/79/815479/1"}}},"commit":{"parents":[{"commit":"b516eb69b83e814e2630ec01961c3e136acb214d","subject":"Add some VirtualMedia fields from 1.3.0 and 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b516eb69b83e814e2630ec01961c3e136acb214d"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-22 09:33:22.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-26 11:58:34.000000000","tz":120},"subject":"WIP certificate collection","message":"WIP certificate collection\n\nChange-Id: I4ea7f084334a9125c737d0d5f594f1f8c926f377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a882b8e40b3d47fe9d14e9be0026a274802712ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a882b8e40b3d47fe9d14e9be0026a274802712ce"}]},"branch":"refs/heads/master"},"f612c0f3d186a66106358ae7c73591a51ce34b1c":{"kind":"REWORK","_number":2,"created":"2021-10-29 12:54:57.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/815479/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/79/815479/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/79/815479/2"}}},"commit":{"parents":[{"commit":"ff11a204dac0df5a7c0e9d8e71b773e2774a3f76","subject":"Add some VirtualMedia fields from 1.3.0 and 1.4.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ff11a204dac0df5a7c0e9d8e71b773e2774a3f76"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-22 09:33:22.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-29 12:54:36.000000000","tz":120},"subject":"WIP certificate collection","message":"WIP certificate collection\n\nChange-Id: I4ea7f084334a9125c737d0d5f594f1f8c926f377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f612c0f3d186a66106358ae7c73591a51ce34b1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f612c0f3d186a66106358ae7c73591a51ce34b1c"}]},"branch":"refs/heads/master"},"4455aa7d08de4d1d678dda9fe406e625cb5034f2":{"kind":"REWORK","_number":3,"created":"2021-11-10 14:43:21.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/815479/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/79/815479/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/79/815479/3"}}},"commit":{"parents":[{"commit":"1a68ff6d598214d1a99525d0231d83e2f0195a1e","subject":"Merge \"Add some VirtualMedia fields from 1.3.0 and 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1a68ff6d598214d1a99525d0231d83e2f0195a1e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-22 09:33:22.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-11-10 14:42:24.000000000","tz":60},"subject":"WIP certificate collection","message":"WIP certificate collection\n\nChange-Id: I4ea7f084334a9125c737d0d5f594f1f8c926f377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4455aa7d08de4d1d678dda9fe406e625cb5034f2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4455aa7d08de4d1d678dda9fe406e625cb5034f2"}]},"branch":"refs/heads/master"},"2203f359b348562d3fe3cf129f5152c8ad45f8ad":{"kind":"REWORK","_number":4,"created":"2021-11-11 13:52:49.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/815479/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/79/815479/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/79/815479/4"}}},"commit":{"parents":[{"commit":"1a68ff6d598214d1a99525d0231d83e2f0195a1e","subject":"Merge \"Add some VirtualMedia fields from 1.3.0 and 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1a68ff6d598214d1a99525d0231d83e2f0195a1e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-22 09:33:22.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-11-11 13:52:18.000000000","tz":60},"subject":"WIP certificate collection","message":"WIP certificate collection\n\nChange-Id: I4ea7f084334a9125c737d0d5f594f1f8c926f377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2203f359b348562d3fe3cf129f5152c8ad45f8ad"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2203f359b348562d3fe3cf129f5152c8ad45f8ad"}]},"branch":"refs/heads/master"},"0a85f637482a6da2082f1a2941b25bacc952a885":{"kind":"REWORK","_number":5,"created":"2021-11-11 16:11:14.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/815479/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/79/815479/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/79/815479/5"}}},"commit":{"parents":[{"commit":"1a68ff6d598214d1a99525d0231d83e2f0195a1e","subject":"Merge \"Add some VirtualMedia fields from 1.3.0 and 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1a68ff6d598214d1a99525d0231d83e2f0195a1e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-22 09:33:22.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-11-11 16:10:16.000000000","tz":60},"subject":"Add basic support for CertificateService","message":"Add basic support for CertificateService\n\nAdds the related data structures and enums, as well as support for\nreplacing certificates.\n\nChange-Id: I4ea7f084334a9125c737d0d5f594f1f8c926f377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0a85f637482a6da2082f1a2941b25bacc952a885"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0a85f637482a6da2082f1a2941b25bacc952a885"}]},"branch":"refs/heads/master"},"fa2e0015854b2885e6fbc532a0836a01f2ae329d":{"kind":"REWORK","_number":6,"created":"2021-11-11 16:15:24.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/79/815479/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/79/815479/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/79/815479/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/79/815479/6"}}},"commit":{"parents":[{"commit":"1a68ff6d598214d1a99525d0231d83e2f0195a1e","subject":"Merge \"Add some VirtualMedia fields from 1.3.0 and 1.4.0\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1a68ff6d598214d1a99525d0231d83e2f0195a1e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-10-22 09:33:22.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-11-11 16:15:16.000000000","tz":60},"subject":"Add basic support for CertificateService","message":"Add basic support for CertificateService\n\nAdds the related data structures and enums, as well as support for\nreplacing certificates.\n\nChange-Id: I4ea7f084334a9125c737d0d5f594f1f8c926f377\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fa2e0015854b2885e6fbc532a0836a01f2ae329d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fa2e0015854b2885e6fbc532a0836a01f2ae329d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
