)]}'
{"id":"openstack%2Fsushy~818364","triplet_id":"openstack%2Fsushy~stable%2Fwallaby~I384baf5c27a531be0c4b78e11e1214d6371fa8be","project":"openstack/sushy","branch":"stable/wallaby","topic":"story-2009663","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2021-11-19 08:13:20.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I384baf5c27a531be0c4b78e11e1214d6371fa8be","subject":"Changing boot device string for vmedia on SuperMicro","status":"MERGED","created":"2021-11-18 11:01:55.000000000","updated":"2021-11-19 08:14:54.000000000","submitted":"2021-11-19 08:13:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818364-story-2009663","meta_rev_id":"65ae20dbae4c383883081d62a1c694542373da48","_number":818364,"virtual_id_number":818364,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-19 08:13:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-11-19 06:02:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2021-11-19 07:06:28.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":1,"date":"2021-11-19 07:06:28.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-18 11:34:05.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-11-18 12:01:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 00:25:39.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-11-19 06:02:28.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-11-19 07:06:28.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"e5a4bdf84dc2bc9417604fa866eca211ff049f2b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2021-11-18 11:01:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bd6fcdf9cb70e351e149a19111e8e0646c048f8a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-18 11:34:05.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/64/818364/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/99b7ee4/99b7ee4.html : FAILURE in 30m 35s","accounts_in_message":[],"_revision_number":1},{"id":"3e8ef07372df12736bf39fdf02710bff16d5cd0d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-18 12:01:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9f8830923e94adc81915ac016c26ecb : SUCCESS in 3m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d69f0712d24541daaa5f60b4e267bce0 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ed5440b746574efc98432545545e3d56 : SUCCESS in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f7f3f467d51f48e3b8a21b4555a0830f : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/99af975b6aed45648375fefe21867ffe : SUCCESS in 4m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a833f37e8e4e4dcbab5406c837e0a6bb : SUCCESS in 6m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/60b2b752f9a244a593ee50cf9924b85e : SUCCESS in 3m 38s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/44ce78720c254642b032114cad7519bf : SUCCESS in 53m 59s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/63f9020a60274987af848a0943a9499a : SUCCESS in 57m 13s","accounts_in_message":[],"_revision_number":1},{"id":"a552c2c24bdd32d134cfe3a38c10fc313c72918f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-19 00:25:39.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/f9b40a5b2fe641a0a2d99c58af6496f4 : RETRY_LIMIT in 3h 01m 14s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/46dc0f960f4a49dc993761d3a75bf22f : SUCCESS in 1h 17m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"dfc14972aa467f9b5abbe5fbfad364be5a8de848","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2021-11-19 04:50:57.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"d8077b5c0c50f1376962b631d09fdd807c025307","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-11-19 06:02:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6256b816429863e71bd3c3ea184a6d2d8ecdd411","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-11-19 07:06:28.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6e5f47198b209bdaa87e097402b14affba115a2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 07:07:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a72cbfff0367e50aed1064f752d17f104d434c82","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 08:13:20.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32c473e21352402cac50585a9a318e26 : SUCCESS in 3m 16s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5e79dbdd90aa46e1891b3d949fe8d18e : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2ea5053aed1c48b6a133cc1ef9cb359a : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/118f1c31cee04deaa298fbafd73577c0 : SUCCESS in 6m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d0cfbf31857947e8930482743e8b01b9 : SUCCESS in 3m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1e90f5164df048ccac12f501b8e8c460 : SUCCESS in 46m 36s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/472e337e6bcd4e0988c1e3563d56fa2a : SUCCESS in 55m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9aadbce20a733cae039af6d3b4b1dfd8adbe1012","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 08:13:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"65ae20dbae4c383883081d62a1c694542373da48","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 08:14:54.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8359aae6f16c44e491cf5848f156a327 : SUCCESS in 1m 03s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d3ded25e892d4f498eb5cd65debc1924 : FAILURE in 56s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2cc60dc7a5d2282bde3b733a26de24fceb1dd90f","revisions":{"2cc60dc7a5d2282bde3b733a26de24fceb1dd90f":{"kind":"REWORK","_number":1,"created":"2021-11-18 11:01:55.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/64/818364/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/64/818364/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/64/818364/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/64/818364/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/64/818364/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/64/818364/1"}}},"commit":{"parents":[{"commit":"b76050c5c162e78572976dd4b199cab389d1d2d5","subject":"Removing optional fields from insert_media payload","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b76050c5c162e78572976dd4b199cab389d1d2d5"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2021-11-09 05:31:22.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2021-11-18 10:35:57.000000000","tz":600},"subject":"Changing boot device string for vmedia on SuperMicro","message":"Changing boot device string for vmedia on SuperMicro\n\nOn SuperMicro machines (e.g. X11 and X12 models) the device name string\nfor CD based virtual media is \"UsbCd\" as opposed to \"CD\" used by other\nvendors.\n\nThis change adjust the boot device target string to \"UsbCd\" if a\nSuperMicro machine is detected\n\nStory: 2009663\nTask: 43913\nChange-Id: I384baf5c27a531be0c4b78e11e1214d6371fa8be\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2cc60dc7a5d2282bde3b733a26de24fceb1dd90f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2cc60dc7a5d2282bde3b733a26de24fceb1dd90f"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
