)]}'
{"id":"openstack%2Fsushy~818554","triplet_id":"openstack%2Fsushy~master~I0940bf7c28fc6e853e71204767018b80f342da20","project":"openstack/sushy","branch":"master","topic":"cross-sushy","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2021-11-26 11:27:38.000000000","reason":"Dmitry Tantsur replied on the change"}},"hashtags":[],"change_id":"I0940bf7c28fc6e853e71204767018b80f342da20","subject":"Run Ironic unit tests to avoid regressions","status":"MERGED","created":"2021-11-19 13:34:13.000000000","updated":"2021-11-26 12:26:28.000000000","submitted":"2021-11-26 12:21:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"818554-cross-sushy","meta_rev_id":"0d2f9c270dc55ef88ba3784285c49ae5bf597624","_number":818554,"virtual_id_number":818554,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-11-26 12:21:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2021-11-23 15:10:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2021-11-26 09:53:29.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2021-11-26 09:53:29.000000000","_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2021-11-26 11:27:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-11-19 13:34:29.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2021-11-19 13:35:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-11-19 13:35:35.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-11-23 15:10:41.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2021-11-26 09:53:29.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"}],"messages":[{"id":"40f437ca2a6c0966f6dd3e6c8457bb40e847d73e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-19 13:34:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bcf613ca07a66686be1a7d5a6de1e673ae2fcbf6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2021-11-19 13:34:29.000000000","message":"Patch Set 1:\n\nThis change depends on a change with an invalid configuration.\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"a01b995e0518620eece7a5b261991e0b62de6d47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 13:35:18.000000000","message":"Patch Set 1: Verified-1\n\nUnable to freeze job graph: Job ironic-cross-sushy is abstract and may not be directly run","accounts_in_message":[],"_revision_number":1},{"id":"b3a258954c54994263f98b403657a0c9f205e5b1","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-11-19 13:35:35.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"0c51b294e453bd19f2c7987008ca88b9cb225c7b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-19 13:36:46.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d431c9f743da72578f6032e1e70b16a9a0b9378e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 13:57:16.000000000","message":"Patch Set 1:\n\nThis change depends on a change that failed to merge.","accounts_in_message":[],"_revision_number":1},{"id":"f62cb16bc65815b9dbef444ace0d59f76a71c892","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-19 13:57:32.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2d35c9e95ba18981cdffc10b718777dc757586a1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-19 15:01:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d1d21ce4ff5b4e57a50ba2b6fc646a0c : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/73391975e1e14b8c8af52eb5d55d7fd3 : SUCCESS in 4m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c975aaaf76e6461e880b81cdb1c37b90 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0e124eaf2a2a417ca8c528f0c1dcd5af : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b59f1215bb7b4fdb99ddccbbea732d74 : SUCCESS in 5m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8e47c76fd4884aa5b5a575084dded9de : SUCCESS in 4m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0510d8279424f26a8e86936d90bfa04 : SUCCESS in 6m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1613371bca944c4d96a20f55c330b558 : FAILURE in 9m 40s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b894e465c5054167a1d3e759087c6bf0 : SUCCESS in 47m 21s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/74a332f6b8b840f9bfd3d01ca8e55f8d : SUCCESS in 59m 09s","accounts_in_message":[],"_revision_number":1},{"id":"7516860de4f34f632678df981c77f0d9d9aab80a","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-23 13:50:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"68c9c0192d144ea1cd4715accd6ade6c915e181c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-23 15:03:37.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1869aa8bc636448a8868fc1dff55f288 : SUCCESS in 5m 33s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/4103882c074e4f9c981a2b540de0ff9f : SUCCESS in 5m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4ff4b3bff1042c782c97d94bf448fa2 : SUCCESS in 4m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bdda24419c924461b6b15b6dbf4f3902 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8bff56c7555140c38328873053e09dc4 : SUCCESS in 4m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d06de68c3fd8473795ce025ab6700b50 : SUCCESS in 4m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e0a674017a64b09ae0e2f910d83f2a4 : SUCCESS in 6m 44s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/75984909be6944f0a0f7bcde4f551464 : SUCCESS in 9m 23s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4ce0200a079d4576864442524e85c66f : SUCCESS in 55m 40s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8e12c3cde63848aea4919a5a1a4c205d : SUCCESS in 1h 05m 56s","accounts_in_message":[],"_revision_number":1},{"id":"fea6f410a527cf2680a501aea450d17c6d895b11","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-11-23 15:10:41.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"063852e6a0c91c3178e45e1163c0396c03eb782a","author":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"date":"2021-11-26 09:53:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"35fe63aeefa612bb499214f4959a2a916c39d7c7","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2021-11-26 11:27:38.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b2f3d0b08bc211ea7d8824a73e143b0d8effc6e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 11:31:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d0829bd1bb55f88a0dfd4aa293737b8e54d77edc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 12:21:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/07f972245a3d4971bd72497b25658f12 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/934bf55063a54d54ba0dbaf5a98f0d9b : SUCCESS in 3m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/23e389cb867d4de9a9344bad5f941f27 : SUCCESS in 3m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea6b92aa113a4c9786602a822ef266fc : SUCCESS in 3m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/311d2e4bcbf942118d6bb05a62f80c7a : SUCCESS in 5m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e37b26c062324c418433115a60e19632 : SUCCESS in 8m 01s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/44e062fb48914fdf8f37037d4a063614 : SUCCESS in 40m 36s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e1a5e8dabd55415a9db1151a5edb0cb3 : SUCCESS in 49m 32s","accounts_in_message":[],"_revision_number":1},{"id":"f7852de05538c05ea61957c9d3270b56b239461a","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 12:21:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"0d2f9c270dc55ef88ba3784285c49ae5bf597624","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-11-26 12:26:28.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ba85010bfef2462f8d1b7d6d87ad157e : SUCCESS in 1m 12s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9ec278672cec68437c1c4ca40502b73d53221e88","revisions":{"9ec278672cec68437c1c4ca40502b73d53221e88":{"kind":"REWORK","_number":1,"created":"2021-11-19 13:34:13.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/54/818554/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/54/818554/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/54/818554/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/54/818554/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/54/818554/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/54/818554/1"}}},"commit":{"parents":[{"commit":"eda4ea115cc1df6b57bd5851b34f58b8a0118906","subject":"Merge \"Changing boot device string for vmedia on SuperMicro\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eda4ea115cc1df6b57bd5851b34f58b8a0118906"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-11-19 13:33:55.000000000","tz":60},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-11-19 13:33:57.000000000","tz":60},"subject":"Run Ironic unit tests to avoid regressions","message":"Run Ironic unit tests to avoid regressions\n\nDepends-On: https://review.opendev.org/c/openstack/ironic/+/818553\nChange-Id: I0940bf7c28fc6e853e71204767018b80f342da20\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9ec278672cec68437c1c4ca40502b73d53221e88"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9ec278672cec68437c1c4ca40502b73d53221e88"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
