)]}'
{"id":"openstack%2Fsushy~820598","triplet_id":"openstack%2Fsushy~stable%2Fxena~If06fc6a52ce92753b50280a859d799f70e5799dc","project":"openstack/sushy","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2021-12-07 10:51:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If06fc6a52ce92753b50280a859d799f70e5799dc","subject":"Raise an AccessError with SessionService init","status":"MERGED","created":"2021-12-06 21:51:24.000000000","updated":"2021-12-09 17:54:00.000000000","submitted":"2021-12-07 10:51:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"820598","meta_rev_id":"0a70982666a63a6dd663162b0796d55d4b2aa608","_number":820598,"virtual_id_number":820598,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-07 10:51:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2021-12-07 09:47:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2021-12-07 09:47:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-06 22:08:28.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-12-06 22:50:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2021-12-07 09:47:38.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"ea5cfe47ad620e68f513381b96ff9d7379fec6c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-12-06 21:51:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"2e2f3c239625ca32d0044fd3c75ccc05da152422","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-12-06 22:08:28.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/98/820598/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/7f16508/7f16508.html : FAILURE in 4m 41s","accounts_in_message":[],"_revision_number":1},{"id":"10b5e14677d03ce495f6863d8ddfbe59249229ed","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-06 22:50:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b51a285edd2145019c432df157844a97 : SUCCESS in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d7a2cbca88f04a1c9fc17190170ff00d : SUCCESS in 3m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5343ab89faa7404ea9a9612907a9f883 : SUCCESS in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bddf0fad9f2d4a319f46416c54b430ca : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc7314bfd6e14e1aaa3eb994230eb2e6 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e85578c64aaf403f9a6e6f2430193deb : SUCCESS in 6m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68f72d6287eb46358974f342249bbbb2 : SUCCESS in 3m 50s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c2796b8b819e4851afdd43d65dd2913c : SUCCESS in 49m 59s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6a2bed3812954282be87b168e94b4c31 : SUCCESS in 52m 03s","accounts_in_message":[],"_revision_number":1},{"id":"310470f1ea3453c8c12295fec082b7928c3181ae","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2021-12-07 09:47:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"f3bc670e2967abddfdb8378f1a83ce22f5c7fcfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 09:48:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c21d48f77edaf53f56cf65265c85aa498df14ec3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 10:51:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/729081b188b149d483c3422501864f3b : SUCCESS in 4m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/223da2d1f43c45a5add94e40ce7cd55f : SUCCESS in 3m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e46b9e8f71984dbe974ea4819fabbd1b : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2acc361243c1442baffe11d4a1808102 : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3d4ff5b94504332a568c58f04b997b3 : SUCCESS in 4m 03s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1704edceeb334db4bae09080c26fae8c : SUCCESS in 50m 02s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/487f44ea612e4b64b754fa1ad3a9c1e1 : SUCCESS in 52m 55s","accounts_in_message":[],"_revision_number":1},{"id":"fd0d81b38c87689d1b57ced0b62ab5899208d72c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 10:51:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2053c01652776995cb548e5ee1e780ea01453665","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-07 12:12:00.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b9ecb700149432b9f3358f7aebf83f0 : SUCCESS in 1m 12s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93622f5ec1074429a34d7200af00ddb7 : SUCCESS in 1m 20s","accounts_in_message":[],"_revision_number":1},{"id":"0a70982666a63a6dd663162b0796d55d4b2aa608","tag":"autogenerated:gerrit:revert","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-12-09 17:54:00.000000000","message":"Created a revert of this change as Ib0005879c69b1cea4435ec04e0d1208e4a33f1fe","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d6afaf53cce774b8117110d005bf43e6456b9d8b","revisions":{"d6afaf53cce774b8117110d005bf43e6456b9d8b":{"kind":"REWORK","_number":1,"created":"2021-12-06 21:51:24.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/98/820598/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/98/820598/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/98/820598/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/98/820598/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/98/820598/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/98/820598/1"}}},"commit":{"parents":[{"commit":"806622c54de0363d4103dcca8f87bbf73c7b0048","subject":"Changing boot device string for vmedia on SuperMicro","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/806622c54de0363d4103dcca8f87bbf73c7b0048"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-12-01 22:03:10.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-12-06 21:50:09.000000000","tz":-480},"subject":"Raise an AccessError with SessionService init","message":"Raise an AccessError with SessionService init\n\nSessionService init previously had a check where it looked\nfor an AccessError exception, but actually did nothing besides\nlog the incident as a result.\n\nPost-intiailization of a client, this can occur as attempts to\nre-authenticate *also* force new sessions service data to be\ncollected and depending on session service behavior, the path\nof the session session service may not be able to be retrieved,\nas an access error would be raised as the refresh would include\nexpired session information.\n\nPreviously, there was no way to know this was happening short of an\nerror such as AttributError from surfacing, as this case could result\nin the code short circuiting, or just ultimately the service would\nlog an error an the universe would attempt to move on, without any\nunderstanding that an AccessError is occuring.\n\nNow the AccessError is raised to the caller.\n\nStory: 2009719\nTask: 44106\nChange-Id: If06fc6a52ce92753b50280a859d799f70e5799dc\n(cherry picked from commit a2096cd302b343c999c6a73a0a162ac0bd8c4a23)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d6afaf53cce774b8117110d005bf43e6456b9d8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d6afaf53cce774b8117110d005bf43e6456b9d8b"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
