)]}'
{"id":"openstack%2Fsushy~821256","triplet_id":"openstack%2Fsushy~stable%2Fxena~Ib0005879c69b1cea4435ec04e0d1208e4a33f1fe","project":"openstack/sushy","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2021-12-09 20:24:41.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2021-12-09 20:24:41.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib0005879c69b1cea4435ec04e0d1208e4a33f1fe","subject":"Revert \"Raise an AccessError with SessionService init\"","status":"MERGED","created":"2021-12-09 17:54:00.000000000","updated":"2021-12-09 20:26:27.000000000","submitted":"2021-12-09 20:24:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"revert_of":820598,"submission_id":"821256","meta_rev_id":"98aa82a1e157c9e0aad3c81acdddd8b9c0379651","_number":821256,"virtual_id_number":821256,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2021-12-09 20:24:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2021-12-09 17:55:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2021-12-09 17:55:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2021-12-09 17:54:00.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2021-12-09 17:54:00.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2021-12-09 17:54:00.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"1960383b6c05bd6e6204d8f36915bc189c6ccc6b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-12-09 17:54:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"cee73020f9ff312fac2956eac1819b4817af1e26","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2021-12-09 17:55:42.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"02c6eb5862499743a96705e2dee656811f60ee1f","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2021-12-09 18:49:45.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/56/821256/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/72ae529/72ae529.html : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":1},{"id":"5aeb12bac3f89123bfe855d84f7568639bb9cf53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 19:03:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c71fa5504295413daeb55a5da3042092 : SUCCESS in 4m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7d448f6e5b2146fda7a080585d375ab5 : SUCCESS in 3m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ffa26b3616c34706942aae6cacc7dc0b : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed91a65a9264cb3979b0fcea34d0a70 : SUCCESS in 3m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/039230e4ff9641e99966f2f350f8969f : SUCCESS in 3m 47s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1c1ede55b67a4c1e83cd2c296073879e : SUCCESS in 6m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/72743ded14d04fab98e89d6aafba934a : SUCCESS in 4m 58s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1961adb0fce64d76a71b7350c4e8f7a0 : SUCCESS in 1h 02m 17s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a49fc3c025fe44a8b00767d7f6fbca74 : SUCCESS in 1h 01m 08s","accounts_in_message":[],"_revision_number":1},{"id":"0d9fd5bd8acf2ad1eb4d376e50c80f1242c694ad","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 19:04:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ac265009208253604447340f133cb717e9d3b3f4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 20:24:41.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96e4e970f7c7448badcab6a658f15aab : SUCCESS in 3m 26s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/91f85d603df4457a96c063d79a2175c1 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c351126cff1e44089bcf365059975e65 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b1d9195c8ecb474ca8aee1e8475f8dc2 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d006a2da74d4f40b8aa7e2f71bde6c0 : SUCCESS in 4m 08s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ea16ea00591d4172adeb762d5fa866ca : SUCCESS in 51m 47s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0e073f32340e4a79bef7850f9b18969e : SUCCESS in 1h 10m 31s","accounts_in_message":[],"_revision_number":1},{"id":"22e84b867cd7d0213971c0f0907450439bbae9d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 20:24:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"98aa82a1e157c9e0aad3c81acdddd8b9c0379651","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2021-12-09 20:26:27.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e50acb4033a475c81c7b73ed36ccbea : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eee5fd0cf0164632abe707a9d6fb2a30 : FAILURE in 1m 04s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9a04713169de24ed3081de3a71d0d099918d00bd","revisions":{"9a04713169de24ed3081de3a71d0d099918d00bd":{"kind":"REWORK","_number":1,"created":"2021-12-09 17:54:00.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/56/821256/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/56/821256/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/56/821256/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/56/821256/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/56/821256/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/56/821256/1"}}},"commit":{"parents":[{"commit":"d6afaf53cce774b8117110d005bf43e6456b9d8b","subject":"Raise an AccessError with SessionService init","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d6afaf53cce774b8117110d005bf43e6456b9d8b"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-12-09 17:54:00.000000000","tz":0},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-12-09 17:54:00.000000000","tz":0},"subject":"Revert \"Raise an AccessError with SessionService init\"","message":"Revert \"Raise an AccessError with SessionService init\"\n\nThis reverts commit d6afaf53cce774b8117110d005bf43e6456b9d8b.\n\nReason for revert: This creates a regression in fallback to guessing a path for creating session initialization, and the underlying failure was not actually causing an AttributeError to be raised but was still silently being consumed. It seems the right fix is a bit more involved and for the sake of cleanliness, reverting this patch.\n\nChange-Id: Ib0005879c69b1cea4435ec04e0d1208e4a33f1fe\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9a04713169de24ed3081de3a71d0d099918d00bd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9a04713169de24ed3081de3a71d0d099918d00bd"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
