)]}'
{"id":"openstack%2Fsushy~830322","triplet_id":"openstack%2Fsushy~master~I608a776f1849621f00e17d1edca16a54906d05fd","project":"openstack/sushy","branch":"master","topic":"nokia-vmedia","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2022-03-02 16:46:50.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-03-02 15:33:49.000000000","reason":"Riccardo Pittau replied on the change"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2022-03-02 15:19:28.000000000","reason":"Iury Gregory replied on the change"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-02-23 11:09:16.000000000","reason":"removed on reply"}},"hashtags":["ironic-week-prio"],"change_id":"I608a776f1849621f00e17d1edca16a54906d05fd","subject":"Re-trying InsertMedia call with TransferProtocolType if required","status":"MERGED","created":"2022-02-22 08:25:38.000000000","updated":"2022-03-02 16:48:51.000000000","submitted":"2022-03-02 16:46:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":32,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"830322-nokia-vmedia","meta_rev_id":"c66e24903cbd61000462e5de241c579704c1d02d","_number":830322,"virtual_id_number":830322,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-02 16:46:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-03-02 15:33:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":2,"date":"2022-03-02 15:19:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2022-03-02 15:33:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-02-22 08:35:07.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-02-22 09:23:31.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-02-22 09:27:42.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-02-22 09:44:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-02-22 16:05:42.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-03-01 14:52:13.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"}],"messages":[{"id":"b576a8c08e2187e22af8555935d7f3138513ada6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-22 08:25:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0c43ce2b5d1bf9f1a3dfd05ca768f7eb25b353f2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-22 08:28:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"279bcdc809d95edfd10aba8d7c38a8116b2a5f05","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-22 08:35:07.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/22/830322/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/f80050f/f80050f.html : FAILURE in 5m 14s","accounts_in_message":[],"_revision_number":2},{"id":"3757be6785368fcaa128b209e4460930c8fb20c3","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-22 09:23:31.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/eb4aba54a41b45dabe0cbc93aaab7c9e : SUCCESS in 53m 29s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/c4d9eaeebcd24024993d7b4cd7c47927 : FAILURE in 5m 05s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"e7434aa27c22b7c40d73104481e02529b4b3ba87","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-22 09:27:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"8cae4e5182f52651f193290c6f8795b2a8979fd1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 09:44:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/963afb8cd02e43dbaa65392bef33a6ca : SUCCESS in 3m 35s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/323fd648da554f7ab31b4da37672340e : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fea69246d4ef4a8796d6b0b607fe7d49 : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3574f59adfc545afb609a69e425af886 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4169e16a67084b6aac2f1a024aa8bb4e : SUCCESS in 4m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/867476d0ea254f7b91fcb107e833f61e : SUCCESS in 5m 15s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2596344bd61e46dc91fdeab2785f9811 : SUCCESS in 8m 47s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3b4452731d04432dabbd98668792c4b8 : SUCCESS in 48m 13s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/43712c7d04ae465784f7dcebb288d8af : SUCCESS in 1h 12m 09s","accounts_in_message":[],"_revision_number":2},{"id":"e385c8691cd6083ec862a65fca3a7e6d208d285c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-22 12:14:46.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"e3e6044162281e50b1c5f13817a098d8b25e10b7","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-22 12:16:12.000000000","message":"Patch Set 2:\n\n(4 comments)","accounts_in_message":[],"_revision_number":2},{"id":"b94b8a13591efa35cf138c3e9cc9e9d13b2c6f71","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-22 12:56:08.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/22/830322/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/3200850/3200850.html : SUCCESS in 40m 11s","accounts_in_message":[],"_revision_number":3},{"id":"64818b2b0bd4a75ab87aeb2b949e89d54f10e742","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-22 13:13:54.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c711e3e08f8b461cac8068080b74ed3f : SUCCESS in 6m 58s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/821a38e0871b4cc98fe1663b8cccb9ce : SUCCESS in 5m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b398e634e2e6446e871676de1cd96abc : SUCCESS in 3m 34s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d0fa870c03b345afb3996a0a0ae2a54a : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/acb645e4a62d4c49b7c03f4aedb09bf6 : SUCCESS in 4m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02eb975772dd4d2f94e8ca879c5ac681 : SUCCESS in 7m 28s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/eb0d119ea4034cc3baa12d24527f0cb3 : SUCCESS in 8m 18s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b7c62ad48e9a4df3907eeb1c8331aa08 : SUCCESS in 54m 46s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0bee066784bc4ec9b5f2e3a219e77968 : SUCCESS in 49m 56s","accounts_in_message":[],"_revision_number":3},{"id":"23b669ef964dad70bbf019bded42297d85980c6d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-22 16:05:42.000000000","message":"Patch Set 3: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"9cca8bccc38f7a6d77a008fd81d71d559ffcb28f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-22 18:11:16.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/179bfc1854fe4038b6b6c244435a335a : SUCCESS in 48m 51s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/f68cb6dd04b74d2aa24739d5f19f7108 : SUCCESS in 1h 26m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"d99634125c63a0bcc36d39c8bb2b5a5979f7749a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-22 23:17:30.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"3906380891c49503e5b5b98444a143e7e41c6edd","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-22 23:17:49.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"db655bf38696202ce7fbd56a033828d53bf5ff59","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-23 00:07:31.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/e92bc0ed37e74837a6da1e38eefdc672 : SUCCESS in 44m 35s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/51f1b839592b4150b1419cd7bae8a266 : SUCCESS in 48m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"63707c84a367c610e10c442b94901d1c453239b1","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-23 00:10:48.000000000","message":"Patch Set 4:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/22/830322/4/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/90a35a9/90a35a9.html : SUCCESS in 52m 02s","accounts_in_message":[],"_revision_number":4},{"id":"c7dccf3065dd57beb229d03ba6af0a9ba054882f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 00:16:37.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40419914237a4467ab145050ddc6e874 : SUCCESS in 3m 24s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/cf148580c34943b2b07d836259a68259 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02dd694314094e3187ec07538df3d83d : SUCCESS in 3m 35s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2f00cd5d5cf24ebfa9a1556cc0562cdc : SUCCESS in 3m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3fa4d73aa9af449e8b37c395612b76f3 : SUCCESS in 3m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4ffefb8d18774df39bca7f7964d8fb61 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27a3c75fc6644c608a526ddea30521c5 : SUCCESS in 3m 18s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/57aa6944d900469b914f448a93fd80f1 : SUCCESS in 8m 34s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/966bff60a6934f19aee2c3c877a7234d : SUCCESS in 36m 35s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/45b64ef58b2e4eba93445856377e3b53 : SUCCESS in 54m 22s","accounts_in_message":[],"_revision_number":4},{"id":"0b24f47525ea0c10131013c4a59228630ce486b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-23 00:28:13.000000000","message":"Uploaded patch set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"f77cbf8c557dcfb6c2f8098918ed5034debc63b7","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-23 01:18:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/6693012aad3847ecac5ac9a8b94b583f : SUCCESS in 49m 07s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/59da9cfc52134bc0adc146d0a15e5f8b : SUCCESS in 47m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"a706ba2ca7c946cea16b9118330e1d0007d4e24c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-23 01:26:32.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/22/830322/5/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/a5deae8/a5deae8.html : SUCCESS in 57m 07s","accounts_in_message":[],"_revision_number":5},{"id":"edb3b444604554c928c8d9eb12c3138b942ec25a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-23 01:40:32.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24882208665641a4976fbe5d515546db : SUCCESS in 3m 28s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/c3a691cf4e3f404c9a83117efa3c4927 : SUCCESS in 3m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b7fbd47bd58b46d5b912fbd013ad5cd5 : SUCCESS in 3m 24s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/e34e7349ed9c4b21b16dbeb03c3f3a0f : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c928a971fd864a7c86dbea7c75cd3572 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0aeafb53545a45028d8ab1d8c934421d : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ca4d91d6b0074fedb842c5fd2fb4bf1b : SUCCESS in 3m 23s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/48be73cf52d249178f0eb3790131cd9a : SUCCESS in 10m 47s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/103a552d10264c8c88f2e98efb092722 : SUCCESS in 1h 06m 12s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7c2a92f72b51484497f83fa2d9361c8d : SUCCESS in 1h 02m 52s","accounts_in_message":[],"_revision_number":5},{"id":"40bd17e83b18c94e27b301c48a4c2f37e600c0a9","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-02-23 11:09:16.000000000","message":"Patch Set 5: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c90b3c446b395edf44f1c572ae4922c5a4f03272","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-28 11:59:19.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"df7880147c562162535816cce4cb1a2c380173df","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-28 12:00:35.000000000","message":"Patch Set 6:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":6},{"id":"e84d51d420d631d2b16d555aaaf6fd31af1c6961","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-02-28 12:55:35.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6b0db3f31a8d428f9f3604966361e881 : SUCCESS in 5m 08s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/0b366d1136164121b76effc059ad2fa3 : SUCCESS in 5m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c6dc1f37c66548e19489c61071ebc882 : SUCCESS in 4m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6520084320b24c7fb4f566294ff84d52 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ddb99a83a074e57a8d46bde926b9868 : SUCCESS in 5m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e9b251b448bf488e8d21583dea5e59b7 : SUCCESS in 6m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/50db886709554d648b711aafc27fdd8d : SUCCESS in 4m 56s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/fad0cfd18fa742d8bd3cf9c9177989f4 : SUCCESS in 8m 37s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8848bd5a1d99450482ce8d07328edbfb : SUCCESS in 43m 51s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/cbef67ca5228464b92b2386e90162b42 : SUCCESS in 53m 59s","accounts_in_message":[],"_revision_number":6},{"id":"3075561c4746d43a2267bd71ad3255420148d58a","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-02-28 13:41:51.000000000","message":"Patch Set 6: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"8f484d439edcb169a29899ea218912fc124c6a3b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-02-28 18:12:10.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/b3ac311047674223965df45a542940e5 : SUCCESS in 46m 16s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/8175fe303107439cb9015b51f3b997d1 : SUCCESS in 1h 44m 39s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"280c3c03735749d098bae4b2bebc4e3f48fe44bd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-28 23:12:56.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"b3c49e3e35f5f5aece842284e6965cb7928c73af","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-02-28 23:13:35.000000000","message":"Patch Set 6:\n\n(3 comments)","accounts_in_message":[],"_revision_number":6},{"id":"ce5fa06dff1ff0734e458908ddd3d9c9007a29df","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-02-28 23:14:13.000000000","message":"Patch Set 7:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":7},{"id":"df480e61961c3c5c329463e573bcf301809e814c","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-01 00:05:40.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/d97673fc20fd4433a46d21a5dbfa1217 : SUCCESS in 46m 25s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/690d76aadb67457185de38bcadca74e9 : SUCCESS in 49m 45s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":7},{"id":"0bc9c9a00380e6d48a7bad341e1965fc94bc0c4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 00:23:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/92e6fdec27f4402a9aecccbff79051bd : SUCCESS in 5m 59s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/324792870b264668a7bffd4a102cdfb5 : SUCCESS in 5m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1b3a600a934947a2bc1575a2550e3229 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0d2aa0de225d4d1dbcc771fdf11b8783 : SUCCESS in 5m 15s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a2eb41990ab541a5b8f819ddac2bffb7 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ec8fac804e304044844390f3d2c23bcc : SUCCESS in 7m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e9ac99417a0c41fd8d129880ad7c0569 : SUCCESS in 4m 48s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a8d5b370abdf4419a075e83d4c688c7b : SUCCESS in 10m 19s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5175aa53527e4565a43b7113d7166dc7 : SUCCESS in 41m 12s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/36e9368815274acfa8b31157ece23908 : SUCCESS in 1h 02m 49s","accounts_in_message":[],"_revision_number":7},{"id":"aa44f9cf143ce77afd604f7aa0be8b4909b289eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-01 06:20:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"23d92b35e3aff49fec358f215e1ad1e2d61ce27b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-01 06:21:28.000000000","message":"Patch Set 8:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":8},{"id":"c83c8f74c27f758ebd13eb9606d6a3a4815bbaf1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-01 07:11:09.000000000","message":"Patch Set 8:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/dccfd1f094bc4a508cea2f308ebd6ee0 : SUCCESS in 45m 57s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/8ce63632104847e5874bab848d9ccbe3 : SUCCESS in 49m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":8},{"id":"b8000dc8646e3c8e2f98b0f44b87c3f6f6797c4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 07:21:59.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0622952e542f49a29ac656afab2b4112 : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/dea359e29d5041f993bb82cd66ee7689 : SUCCESS in 5m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4bbf5529138f4fd0b8b407ac77070e76 : SUCCESS in 4m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/57466a0ff00f4a5fa4361cb469f5c0b8 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e4eef157ffd949f98e0e5054a054034d : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff781ddf3a5495293f8c453dc819c89 : SUCCESS in 7m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a982c8c0946d4295bfac65831902fc2a : SUCCESS in 4m 55s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2969df4f88514c20a75e547af64697bc : SUCCESS in 9m 18s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4c41cedfa6c347b0af8cda8c2bc28c3e : SUCCESS in 1h 00m 39s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7169d73c13f948b1951b3479f1d6ab4b : SUCCESS in 49m 13s","accounts_in_message":[],"_revision_number":8},{"id":"11d995e078a5a923ef74604ce9e594e562b640ac","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-03-01 08:08:30.000000000","message":"Patch Set 8: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"0659f3f98dffd1821aaca29cecc9e85e7eea2e83","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-01 09:49:47.000000000","message":"Patch Set 8:\n\n(3 comments)","accounts_in_message":[],"_revision_number":8},{"id":"9ca7040c1caa697b30d810a356391c676c061661","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-01 09:53:18.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"20b27077f30d2186038004e889cce1623237484a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-01 09:54:35.000000000","message":"Patch Set 9:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":9},{"id":"5c54e4f42601a036de73ac2e61cffd7101764582","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-01 09:55:37.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"f0ce85943990e3589d240f02a56f0e476837cb9c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-01 09:56:54.000000000","message":"Patch Set 10:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":10},{"id":"948da40ec867f563b743ae3f961bbd02c5cd32bd","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-01 10:15:37.000000000","message":"Patch Set 10:\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"ec65d104fc648f36378a179d3a5b17297742d332","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-01 10:46:03.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/d3e45321da654fd3bd48adeb8db013a5 : SUCCESS in 44m 49s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/13221f74aa0c4816af588a42a86ea3a5 : SUCCESS in 49m 04s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":10},{"id":"851ebb033fc51ec9122c3a272d8e8466c8f5a671","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-01 11:06:53.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/010e4fe66c494aee9fcc4d295bab6d12 : SUCCESS in 5m 36s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/e2a0324c55cb4adda619a72878b837f0 : SUCCESS in 5m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/804a9366185d4abfa5d843d6c2b3b3ea : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/79b17aa73b5c4c0b89d366beaed387ac : SUCCESS in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/924134c16de74cbfa4568e8a91a06f9c : SUCCESS in 5m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2459cdf33314b58b13ee203e03cdad2 : SUCCESS in 7m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0703c07cdfb64f3c9430067b64b7876a : SUCCESS in 4m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4c8084b04c7846d5b18f267e5161a39d : SUCCESS in 11m 04s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3ca5a51e22e343ebbe084939b01b9fe4 : SUCCESS in 48m 17s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/13bc015b52ea433293b5d5559ae9bb79 : SUCCESS in 1h 09m 00s","accounts_in_message":[],"_revision_number":10},{"id":"1c3e7666e29bca1c228358b69c67fa2e3778b70f","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-03-01 14:52:13.000000000","message":"Patch Set 10: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":10},{"id":"450372f80407d838ca684cf136e394fb7d8caaa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-02 07:31:50.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"ded9d43aad6876221d84caf737e31529b94e4698","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-02 07:32:26.000000000","message":"Patch Set 11:\n\n(3 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0f3df0d909afd525b5dea2bed1eff8eb2dcfa92f","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-02 07:33:07.000000000","message":"Patch Set 11:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":11},{"id":"884f652119de5cb04fdd5832571a07632c0dfe1b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-02 08:26:21.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/3a6597325d7644399fae9345e18807bd : SUCCESS in 44m 27s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/e888a111f87e4ba2b5c74f63772aef98 : SUCCESS in 53m 16s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":11},{"id":"2d4d7c49cf20e9aa2c592f6eaaaa72ed32a7e7ec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 08:33:32.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d17c2049c0c34e4dbfdb685242c9870e : SUCCESS in 5m 19s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/378131fe8a29492cb2196a706cb91f06 : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a2b7013ddbe4f9692f7bd1543d57f53 : SUCCESS in 5m 04s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bb86822f3a6843f799fea0b3b36a8237 : SUCCESS in 4m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8a3baa68a0214224a3dd1e0e6b4a482f : SUCCESS in 5m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b03b015aa5584ba7a3de7ca9b6683aae : SUCCESS in 7m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/82a37b67ddbe4a8289d4c36530795afb : SUCCESS in 5m 12s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d29cea80f9224542aa2663a0a759bac5 : SUCCESS in 10m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d3da520761a34cd1a846e7189d4b894d : SUCCESS in 41m 14s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/086931eaa6db470d86dc8acbf4a4d877 : SUCCESS in 1h 00m 13s","accounts_in_message":[],"_revision_number":11},{"id":"52cf1a59c7265c42d1464e8ca3a56f0e9ca14bd3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-03-02 12:19:15.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"9175c68c9700f7ab879c7c1a5b64e6cf3736b55c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-02 12:20:33.000000000","message":"Patch Set 12:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":12},{"id":"8f2cdb09ac126374e11a1f19f2a3d0baa7d6965c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 13:22:15.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d67aeff5a34c4188b8ee7a9b4a7047b8 : SUCCESS in 5m 27s\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/47bf1f8fd9ec453f99799bd4194187f0 : SUCCESS in 5m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fa1079d770234de2a64bdbe795594b9d : SUCCESS in 4m 50s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/4a1205994b9640a38ddf79b2a79b05e8 : SUCCESS in 5m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3a44318fa74f4b87a9da59999ebddf88 : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cea995adc9e4c50ba60ae1c8b18d6cb : SUCCESS in 7m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6317df1b1a514b85a69fba54ffa49173 : SUCCESS in 5m 17s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/52504443744a4f14aca4cfff0d9d1ce9 : SUCCESS in 10m 13s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a59320e3b67148da9456a1b17254a2eb : SUCCESS in 56m 59s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/01b30c58f89c49a09513be9c1cabc497 : SUCCESS in 55m 44s","accounts_in_message":[],"_revision_number":12},{"id":"e90ed154159c14013c55830e8ccc19a2f056df48","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-02 13:28:06.000000000","message":"Patch Set 12:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/fe81b31756c9437bbe9fc902ace3120c : SUCCESS in 45m 51s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/4564385faed443b8af0392bd8178abbe : SUCCESS in 1h 07m 30s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":12},{"id":"8b24a91aea1949641ce0c922a38c768ac4cb0304","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-03-02 15:19:21.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":12},{"id":"73e4408d7d53fd441a3b3552c6a90f80efc5dc1d","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-03-02 15:19:28.000000000","message":"Patch Set 12: Code-Review+2","accounts_in_message":[],"_revision_number":12},{"id":"ad1130050d8e33ebae89ddc752826dd7fb3072c4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-03-02 15:33:49.000000000","message":"Patch Set 12: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":12},{"id":"c4f939d1639065840cdce93c1bbb4cee57dff13f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 15:35:00.000000000","message":"Patch Set 12: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":12},{"id":"2e5aaf6ef01f40c245ab06b88367c5d47d08d3f0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:46:49.000000000","message":"Patch Set 12: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-lower-constraints https://zuul.opendev.org/t/openstack/build/39cc74ddd0e347d487312931e16456c9 : SUCCESS in 5m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/32faa42f3de34bb88951ad121e976185 : SUCCESS in 5m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d18511197f8b4be4bf0a09a85e1872d9 : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c93adecc243b46cfa30a8cf1c5758c17 : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3a35124196f24c6dac446291c462acef : SUCCESS in 8m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a9cc8ee8723434592c53f0feae72618 : SUCCESS in 5m 16s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a4c76f5d299d446691715ee132b55aa4 : SUCCESS in 10m 39s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/58a72d073404435ea4c8ff21950c0c3a : SUCCESS in 47m 14s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/29c635cbea044b7a83834c0eefd303fc : SUCCESS in 1h 11m 03s","accounts_in_message":[],"_revision_number":12},{"id":"6a5f92d7d6b74c44d018f821eb8a33c94efb47c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:46:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":12},{"id":"c66e24903cbd61000462e5de241c579704c1d02d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-02 16:48:51.000000000","message":"Patch Set 12:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6efa1b83edf74735bfd828a88dc948fb : SUCCESS in 1m 33s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2ca8a15d459a43b08ab4ce17d60d5b40 : SUCCESS in 1m 18s","accounts_in_message":[],"_revision_number":12}],"current_revision_number":12,"current_revision":"ecf1bcc80bd14a1836d015c3dbdb4fd88f2bbd75","revisions":{"58e9147a2ced9c3ba4bcc81d5266e42634ba3374":{"kind":"REWORK","_number":1,"created":"2022-02-22 08:25:38.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/1"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter required on some Nokia servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/58e9147a2ced9c3ba4bcc81d5266e42634ba3374"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/58e9147a2ced9c3ba4bcc81d5266e42634ba3374"}]},"branch":"refs/heads/master"},"928d4c36877f017dca64cef1ba96802841bb27aa":{"kind":"REWORK","_number":2,"created":"2022-02-22 08:28:39.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/2"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:28:21.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter required on some Nokia servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/928d4c36877f017dca64cef1ba96802841bb27aa"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/928d4c36877f017dca64cef1ba96802841bb27aa"}]},"branch":"refs/heads/master"},"878295592b503ce68184871fb6f84a1820bce2d0":{"kind":"REWORK","_number":3,"created":"2022-02-22 12:14:46.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/3"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 12:14:24.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter required on some Nokia servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/878295592b503ce68184871fb6f84a1820bce2d0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/878295592b503ce68184871fb6f84a1820bce2d0"}]},"branch":"refs/heads/master"},"221de3a14f90dfe25fd0412b95c897bb16355e6d":{"kind":"REWORK","_number":4,"created":"2022-02-22 23:17:30.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/4"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 23:16:42.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/221de3a14f90dfe25fd0412b95c897bb16355e6d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/221de3a14f90dfe25fd0412b95c897bb16355e6d"}]},"branch":"refs/heads/master"},"d20edceef54ddfd78c8c715c8ab0d621924e6b15":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2022-02-23 00:28:13.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/5"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-23 00:26:36.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d20edceef54ddfd78c8c715c8ab0d621924e6b15"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d20edceef54ddfd78c8c715c8ab0d621924e6b15"}]},"branch":"refs/heads/master"},"9e4d2eafa1a0d6d2e4c5d3a9664960c5c443660a":{"kind":"REWORK","_number":6,"created":"2022-02-28 11:59:19.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/6"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-28 11:59:06.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9e4d2eafa1a0d6d2e4c5d3a9664960c5c443660a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9e4d2eafa1a0d6d2e4c5d3a9664960c5c443660a"}]},"branch":"refs/heads/master"},"855fdbc0d8d0d2bf1e3f55ea773910be98232f5f":{"kind":"REWORK","_number":7,"created":"2022-02-28 23:12:56.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/7"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-28 23:12:33.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/855fdbc0d8d0d2bf1e3f55ea773910be98232f5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/855fdbc0d8d0d2bf1e3f55ea773910be98232f5f"}]},"branch":"refs/heads/master"},"bf7ddaba503256b6d2ee10cce579aed185aabb7f":{"kind":"REWORK","_number":8,"created":"2022-03-01 06:20:11.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/8"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-03-01 06:20:01.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bf7ddaba503256b6d2ee10cce579aed185aabb7f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bf7ddaba503256b6d2ee10cce579aed185aabb7f"}]},"branch":"refs/heads/master"},"e9b372f97ab1b67a4e0d7c65cbfb96e516a377cb":{"kind":"REWORK","_number":9,"created":"2022-03-01 09:53:18.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/9"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-03-01 09:53:09.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e9b372f97ab1b67a4e0d7c65cbfb96e516a377cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e9b372f97ab1b67a4e0d7c65cbfb96e516a377cb"}]},"branch":"refs/heads/master"},"c749cbd8c9a1058836fcd78033a2d347ed14e85f":{"kind":"REWORK","_number":10,"created":"2022-03-01 09:55:37.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/10"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-03-01 09:55:29.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c749cbd8c9a1058836fcd78033a2d347ed14e85f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c749cbd8c9a1058836fcd78033a2d347ed14e85f"}]},"branch":"refs/heads/master"},"eb006e4a273e163fe3239ca687c3cc6e65fd652c":{"kind":"REWORK","_number":11,"created":"2022-03-02 07:31:50.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/11"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-03-02 07:31:42.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eb006e4a273e163fe3239ca687c3cc6e65fd652c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eb006e4a273e163fe3239ca687c3cc6e65fd652c"}]},"branch":"refs/heads/master"},"ecf1bcc80bd14a1836d015c3dbdb4fd88f2bbd75":{"kind":"REWORK","_number":12,"created":"2022-03-02 12:19:15.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/22/830322/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/22/830322/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/22/830322/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/22/830322/12"}}},"commit":{"parents":[{"commit":"4f45f13dd7bc15cf2a20240fa0454bcce0f55d58","subject":"Stop logging MessageRegistry objects","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4f45f13dd7bc15cf2a20240fa0454bcce0f55d58"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-02-22 08:24:39.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-03-02 12:19:08.000000000","tz":600},"subject":"Re-trying InsertMedia call with TransferProtocolType if required","message":"Re-trying InsertMedia call with TransferProtocolType if required\n\nAdding extra logic to handle InsertMedia API call failure due to\nTransferProtocolType parameter missing. This is a non-standard\nparameter (rarely) required on some servers. If such failure mode\nis detected, TransferProtocolType value is automatically added\nto the request payload prior to re-trying the InsertMedia call.\n\nStory: 2009871\nTask: 44574\nChange-Id: I608a776f1849621f00e17d1edca16a54906d05fd\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ecf1bcc80bd14a1836d015c3dbdb4fd88f2bbd75"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ecf1bcc80bd14a1836d015c3dbdb4fd88f2bbd75"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
