)]}'
{"id":"openstack%2Fsushy~832827","triplet_id":"openstack%2Fsushy~stable%2Ftrain~I4891b89a65567c9ba2057557462a5d3840808ae1","project":"openstack/sushy","branch":"stable/train","topic":"auth-none-stable/victoria-stable/ussuri-stable/train","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-03-10 17:58:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4891b89a65567c9ba2057557462a5d3840808ae1","subject":"Protect Connector against empty auth object","status":"MERGED","created":"2022-03-09 22:12:18.000000000","updated":"2022-03-10 18:01:48.000000000","submitted":"2022-03-10 17:58:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832827-auth-none-stable/victoria-stable/ussuri-stable/train","cherry_pick_of_change":832826,"cherry_pick_of_patch_set":1,"meta_rev_id":"77ca38f661dbaf22156fc081087423aeedbdec13","_number":832827,"virtual_id_number":832827,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-10 17:58:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-03-10 16:39:19.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2022-03-10 16:39:19.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-09 22:12:18.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2022-03-09 22:21:48.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-03-09 23:30:18.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 00:38:12.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-03-10 16:39:19.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"23c48f28d4897d5a166ada414729251cf082b5eb","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-03-09 22:12:18.000000000","message":"Patch Set 1: Cherry Picked from branch stable/ussuri.\n\nThe following files contain Git conflicts:\n* sushy/connector.py","accounts_in_message":[],"_revision_number":1},{"id":"7629fab214f0c25a68a3e65edaf7f0baf47d5243","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-03-09 22:16:34.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"95f3fc6e2802b473b0ca64e5a5042725ad761d4d","tag":"autogenerated:gerrit:setReadyForReview","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-03-09 22:16:44.000000000","message":"Set Ready For Review","accounts_in_message":[],"_revision_number":2},{"id":"a8147fb75bfa73b67997f9c5084f40cf660e65d4","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-09 22:21:48.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/27/832827/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/fb936e3/fb936e3.html : FAILURE in 4m 13s","accounts_in_message":[],"_revision_number":2},{"id":"0617b370768228da95a0e1b5b32f0d2420bb4d55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 23:30:18.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c44ad32034de48258c1e9acbb7313387 : SUCCESS in 5m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f3cfc5e3973b4328b30ede8e8bc8323f : SUCCESS in 5m 00s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/1e166690d845460186e84e2f2b606f45 : SUCCESS in 4m 44s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3693b2712c02453d8ae9f018fe978208 : SUCCESS in 5m 27s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/294723d98ce944aa860191a51f0e83f5 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b06c1f0e8533455d98f05488db4ea171 : SUCCESS in 7m 15s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/b9d806bd17684069972a148eb58ae9c7 : SUCCESS in 1h 06m 49s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/6544e6ff59e643618e71ae8cd667d97c : SUCCESS in 45m 37s","accounts_in_message":[],"_revision_number":2},{"id":"8087af5232832d23358c30265c0c57707fee86c9","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-10 00:38:12.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/4bee46a5e19b4786a1f5f2b16004ae5d : FAILURE in 39m 36s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/6f4cf1de7f624de1aec816c3e1e28ef4 : FAILURE in 1h 19m 11s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"e2180978c6ba3cb5d9c323f34831c675ddb17ae5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-03-10 16:39:19.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"2e2a4b8f7ad0cbe4ce3d4a221b8b6260f72dbada","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 16:43:26.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"cb117bfead3c20bd80277b46ca2b37cf6314e20b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:58:15.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a253e269e2284813b3753b629f56d493 : SUCCESS in 5m 56s\n- openstack-tox-py27 https://zuul.opendev.org/t/openstack/build/652d320a67694ad69fea27ce2443d36d : SUCCESS in 5m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/cd14cf19b71f4d7dab2aaf39d8cfe4df : SUCCESS in 6m 05s\n- openstack-tox-py37 https://zuul.opendev.org/t/openstack/build/86bdb37c82564148a8c00a94bd43490a : SUCCESS in 6m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fdea4532397844c0832b02656db169c7 : SUCCESS in 8m 57s\n- sushy-tempest-ironic-partition-redfish-src https://zuul.opendev.org/t/openstack/build/db7dd1952b034b9fa39eee5713e26219 : SUCCESS in 48m 43s\n- sushy-tempest-ironic-partition-redfish-src-python2 https://zuul.opendev.org/t/openstack/build/5c2d0aced849497d997201fdb1f691c9 : SUCCESS in 47m 56s","accounts_in_message":[],"_revision_number":2},{"id":"e28bb898d64d4cbb22ad87c82cab90ba3be9ea18","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:58:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"77ca38f661dbaf22156fc081087423aeedbdec13","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 18:01:48.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a64644258c824e73b9d4bd6e4b77c7b6 : SUCCESS in 1m 08s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"060c74c911bc520b9a84dec640d39cf1acd78ddf","revisions":{"31a4c0dc740bedb6eb170f0c575c1ddece2a87ed":{"kind":"REWORK","_number":1,"created":"2022-03-09 22:12:18.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/27/832827/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/27/832827/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/832827/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/832827/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/832827/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/27/832827/1"}}},"commit":{"parents":[{"commit":"84c8bb548a75a591cb4fd5b49fbc239eba49c369","subject":"Merge \"Revert \"Raise an AccessError with SessionService init\"\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/84c8bb548a75a591cb4fd5b49fbc239eba49c369"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-22 11:49:34.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-03-09 22:12:18.000000000","tz":0},"subject":"Protect Connector against empty auth object","message":"Protect Connector against empty auth object\n\nAuth uses Connector, Connector uses Auth. This convoluted relationship\nresults in connector._auth being unset when an authentication request is\nmade. If it results in AccessError, sushy tries to access non-existing\nconnector._auth. This change protects against it and allows the actual\nerror to bubble up.\n\nChange-Id: I4891b89a65567c9ba2057557462a5d3840808ae1\nStory: #2008661\nTask: #41935\n(cherry picked from commit bf50d7363888f2333bdf0c36d65b4144b66d37d4)\n(cherry picked from commit efd47e36737538a5143c164e360dad2207c0d4fb)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/31a4c0dc740bedb6eb170f0c575c1ddece2a87ed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/31a4c0dc740bedb6eb170f0c575c1ddece2a87ed"}]},"branch":"refs/heads/stable/train"},"060c74c911bc520b9a84dec640d39cf1acd78ddf":{"kind":"REWORK","_number":2,"created":"2022-03-09 22:16:34.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/27/832827/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/27/832827/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/832827/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/832827/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/27/832827/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/27/832827/2"}}},"commit":{"parents":[{"commit":"84c8bb548a75a591cb4fd5b49fbc239eba49c369","subject":"Merge \"Revert \"Raise an AccessError with SessionService init\"\" into stable/train","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/84c8bb548a75a591cb4fd5b49fbc239eba49c369"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2021-06-22 11:49:34.000000000","tz":120},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-03-09 22:16:29.000000000","tz":-480},"subject":"Protect Connector against empty auth object","message":"Protect Connector against empty auth object\n\nAuth uses Connector, Connector uses Auth. This convoluted relationship\nresults in connector._auth being unset when an authentication request is\nmade. If it results in AccessError, sushy tries to access non-existing\nconnector._auth. This change protects against it and allows the actual\nerror to bubble up.\n\nChange-Id: I4891b89a65567c9ba2057557462a5d3840808ae1\nStory: #2008661\nTask: #41935\n(cherry picked from commit bf50d7363888f2333bdf0c36d65b4144b66d37d4)\n(cherry picked from commit efd47e36737538a5143c164e360dad2207c0d4fb)\n(cherry picked from commit 781159e2f3d55cf063801577c2f4b3cf538a66cd)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/060c74c911bc520b9a84dec640d39cf1acd78ddf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/060c74c911bc520b9a84dec640d39cf1acd78ddf"}]},"branch":"refs/heads/stable/train"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
