)]}'
{"id":"openstack%2Fsushy~832864","triplet_id":"openstack%2Fsushy~stable%2Fvictoria~Id9a32ebcc04f4f3fecb4e10cf157a97322707c94","project":"openstack/sushy","branch":"stable/victoria","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-03-10 17:58:33.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id9a32ebcc04f4f3fecb4e10cf157a97322707c94","subject":"Fix session authentication issues","status":"MERGED","created":"2022-03-09 22:00:02.000000000","updated":"2022-03-10 18:02:07.000000000","submitted":"2022-03-10 17:58:33.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"832864","meta_rev_id":"046f4dcc696d8bd801b86b11b6269ce87240a267","_number":832864,"virtual_id_number":832864,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-03-10 17:58:33.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-03-10 16:42:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2022-03-10 16:42:49.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-03-09 22:05:13.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-03-09 22:59:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-03-10 00:05:04.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-03-10 16:42:49.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"de2cb1de0f28f2881168ceea7798dbf485733b5c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-03-09 22:00:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e2a684375f32aaf5f6db3ec464e4984028e5036","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-03-09 22:05:13.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/64/832864/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/e92a61a/e92a61a.html : FAILURE in 4m 10s","accounts_in_message":[],"_revision_number":1},{"id":"f38cf5ced57844a08c1a26bd8317effb30ffbf14","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-09 22:59:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a45b64f00cfd4bfea8924e6eff65928c : SUCCESS in 5m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b0b6e9fd06354b0bb427d5748f2ff313 : SUCCESS in 4m 59s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/80e27a4548fd46809d96e76aee9c93f7 : SUCCESS in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/52d87639acf742069d7a52eb351315b6 : SUCCESS in 5m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b95deded920467983da4ef7ab2f30eb : SUCCESS in 7m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fa2db513774b49d78be8ea1a84ed6101 : SUCCESS in 5m 38s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a8f4c8c696614d269fd0a8b36c42e37f : SUCCESS in 52m 47s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/46e281a680c54569a10a2acccef95bce : SUCCESS in 57m 18s","accounts_in_message":[],"_revision_number":1},{"id":"cb6c1a1f2467930f6650b47079a83b686ad77680","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-03-10 00:05:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://100.82.32.82/t/openstack/build/22e7c9b6675042ec90b3e8b51fcd8a9e : SUCCESS in 1h 10m 59s\n- R6515-bios-redfish https://100.82.32.82/t/openstack/build/9a3aff5d81134b839e4e088d9a80eadb : SUCCESS in 1h 42m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"547c798b0110928573aeab2f32cd6ceca6840218","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-03-10 16:42:49.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4562eb73cdd579e9533a54ce81985509a5a016b6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 16:44:27.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6df4ec534f19c482f1535cc313b3d20101cdd328","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:58:33.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0726e6ed00a24ce7b228753463f710cb : SUCCESS in 6m 30s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/be643e9f4da44df2bc6c2fa210815812 : SUCCESS in 6m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2e3107d811274ca0adf33fbe5872113f : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72f9d08735bd43f3996e88ed5207998b : SUCCESS in 9m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7e5324a9e30b44ada13dcf9d2588bc21 : SUCCESS in 6m 36s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/05b2b35c1ec14811a1fa1cd48e771be2 : SUCCESS in 1h 00m 43s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4d08c8d388444816889bc6a1f5877687 : SUCCESS in 1h 04m 30s","accounts_in_message":[],"_revision_number":1},{"id":"a1edf4bc8b22fbfbc872a734d484028ef86e62c9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 17:58:33.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"046f4dcc696d8bd801b86b11b6269ce87240a267","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-03-10 18:02:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93f6308ff3f34376bcfb20986183d561 : SUCCESS in 1m 14s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc3a5db1ade545fea712ff777ca9c1f8 : SUCCESS in 1m 24s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1acff0608019ad6c6a363b2f600c03400fd80670","revisions":{"1acff0608019ad6c6a363b2f600c03400fd80670":{"kind":"REWORK","_number":1,"created":"2022-03-09 22:00:02.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/64/832864/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/64/832864/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/64/832864/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/64/832864/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/64/832864/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/64/832864/1"}}},"commit":{"parents":[{"commit":"ef9bc98f0870d405fbfc8318ebe7952d06d5240d","subject":"Revert \"Raise an AccessError with SessionService init\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ef9bc98f0870d405fbfc8318ebe7952d06d5240d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2021-12-09 21:04:42.000000000","tz":-480},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-03-09 21:59:11.000000000","tz":-480},"subject":"Fix session authentication issues","message":"Fix session authentication issues\n\nSession authentication issues can occur when the client\nhas been initialized and some intermediate failure has occurred\nsuch as a transport failure to the remote BMC or even with setting\nchanges on the remote BMC.\n\nPreviously, these sorts of issues could result in cases where\nthe client does not properly respond. Either by falling back\nto \"basic\" authentication, or getting stuck in a loop where\nthe client was unable to re-negotiate a new session with the BMC.\n\nIn the event of the BMC no longer being able to support Session\nauthentication, after a session initialization, the client will\nraise an AccessError exception to signify to the consumer that\nsomething has occurred with credentials preventing further client\nuse without potentially re-initializing the client.\n\nShould the re-authentication fail, the exception will be directly\nraised such that the client can no longer fall into a recursive\nloop.\n\nShould the state in the client be unable to be determined,\nand an AccessError is raised, the client will now attempt to\nre-authenticate.\n\nUltimately moves from using the SessionService to using a new\nbase class method ``create_session``. The SessionService\nclass ``create_session`` method is still usable as well, however\nnew authentication attempts will not utilize it in order to bypass\npotential failure cases.\n\nFunctionally, this change requires change\nI9eb238d1bbaf522d03ee3fce12fc0ab80c1b69b4 to be present\nas it removes X-Auth-Token headers which can lead to spurious\nauthentication failures when sessions expire. Prior to any\nauthentication action, this patch does *also* do the needful\nto prevent a poisoned client which would have potentially\nresulted in an AccessError previously.\n\nIn backporting, also minor conflict with\nhttps://review.opendev.org/c/openstack/sushy/+/799504\nresulted in a minor change, specifically around the argument\nbeing passed which was added in that changeset. Additional\nminor changes around ConnectionError exceptions which\npreviously did not include the ability to pass a message back\nas part of the error.\n\nChange-Id: Id9a32ebcc04f4f3fecb4e10cf157a97322707c94\nStory: 2009719\nTask: 44106\n(cherry picked from commit 8d55d51284288ef7cc3dad4dce00fe0155b1e6f1)\n(cherry picked from commit 38e62dd16a971d75e85fcea53612da900f19f606)\n(cherry picked from commit 078707439a67f60d6626c36faa096713a88097ac)\n(cherry picked from commit 8427417e208d7d7b48348b7eb9c36ff78d87613f)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1acff0608019ad6c6a363b2f600c03400fd80670"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1acff0608019ad6c6a363b2f600c03400fd80670"}]},"branch":"refs/heads/stable/victoria"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
