)]}'
{"id":"openstack%2Fsushy~840069","triplet_id":"openstack%2Fsushy~master~Ic3be2ae9dd65991606f42c77acaf7470cd6a436e","project":"openstack/sushy","branch":"master","topic":"drop-l-c","attention_set":{},"removed_from_attention_set":{"8556":{"account":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"last_update":"2022-05-02 16:15:12.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ic3be2ae9dd65991606f42c77acaf7470cd6a436e","subject":"Drop lower-constraints.txt and its testing","status":"MERGED","created":"2022-05-01 01:20:54.000000000","updated":"2022-05-02 16:17:55.000000000","submitted":"2022-05-02 16:15:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840069-drop-l-c","meta_rev_id":"c6ade5e8d28e28ba970f5fa397378df0bd53b688","_number":840069,"virtual_id_number":840069,"owner":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-02 16:15:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":2,"date":"2022-05-02 15:19:02.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":2,"date":"2022-05-02 14:15:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"all":[{"value":1,"date":"2022-05-02 15:19:02.000000000","_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-01 01:22:11.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-05-01 02:32:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-01 10:30:04.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-05-02 14:15:57.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-05-02 15:19:02.000000000","updated_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"reviewer":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"state":"REVIEWER"}],"messages":[{"id":"7afa43ec3420e3758930a511d59d763cf53b253d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"date":"2022-05-01 01:20:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"42722b916ae373f8fd9b6dc1a21e49c875c52e90","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-05-01 01:22:11.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"16c5a2f593a62a264069d93597df6a5b51e461d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-01 02:32:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/3a298a1e862943fdaf69da86fe1b7d58 : SUCCESS in 2m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6daa7d62b0c546dc92ebb7b93dad89f1 : SUCCESS in 4m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8ca20c35eca440b98558a573452a5e66 : SUCCESS in 3m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf87bc75fce34bef9c9952cc2e4ac0e9 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bf4d63fc64e6413cb3299d52faac1dd8 : SUCCESS in 3m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3075f48be0fa473e98dcd1fee46e1c57 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9baa30f96aa34db4a4a6d410e321b621 : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9108667d3a43495687f24882b770ce9f : SUCCESS in 3m 55s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/afa977f3226b4d2cbdcec17b92dad8ad : SUCCESS in 8m 39s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0117cdbacd8d4bb8b27f9fd2e520b0a8 : SUCCESS in 44m 27s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/443d2eccb51c4c96a1efd2b45b354fd4 : SUCCESS in 1h 04m 50s","accounts_in_message":[],"_revision_number":1},{"id":"7703967552f0dcdb19012fa50f305ae5bd512ac8","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-05-01 10:30:04.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/d7465fa09c2a4968b825efcce137330e : SUCCESS in 51m 29s\n- R6515-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/39b121a1989148a584038bb366368575 : RETRY_LIMIT in 3h 01m 12s\n- R750-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/afe2ce91f3d24d8e9777460271876208 : SUCCESS in 52m 50s\n- XR11-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/1a3e1b08da3548aa813cc51a849bc09b : SUCCESS in 51m 38s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"787ec55c81a9bd2fc79dcc0b7d4764bb26ea79dd","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-05-02 14:15:57.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2d6618201268fa728c7e0ae6065235083377b0ed","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-05-02 14:41:29.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"8fa5f63df73a6d012d8954a5cc8dc221cccfd157","author":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"},"date":"2022-05-02 15:19:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"9e306a44e3f9b66fee804a25574f0af4b2b065e8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 15:19:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eb471f62d8d296678bfcb40b7457c4081e2ac444","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:15:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/db7ec89e285f4c9388c649ca9d6d0975 : SUCCESS in 2m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f2b640749fea49ea91ab3609c4e7687e : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c4232117f4bb40c784c24e5867305ef2 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cd784f42bddf4082a324c9761662d384 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a60ff750135b4e1cb0f03669769bda77 : SUCCESS in 5m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7169fde9d58748a281699a4dbdca247b : SUCCESS in 3m 08s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/5a33680e6158435cbe459f30c21d003a : SUCCESS in 8m 53s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5fe01eb6d91c41a591e47694a64bd034 : SUCCESS in 35m 29s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2b77b81a16044a4388a03320d69ab939 : SUCCESS in 54m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c3017d795e9a6b7abbe3db411134810fd9ee5f76","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:15:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c6ade5e8d28e28ba970f5fa397378df0bd53b688","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-02 16:17:55.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/62b8353a7d2e4e1c800c9c52f19e5e87 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3ed6bf2982f40eea356ab84f8f69051 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"cdeeb8b495bd539dd3aaf2b1d6d08c4636908f54","revisions":{"cdeeb8b495bd539dd3aaf2b1d6d08c4636908f54":{"kind":"REWORK","_number":1,"created":"2022-05-01 01:20:54.000000000","uploader":{"_account_id":8556,"name":"Ghanshyam Maan","display_name":"Ghanshyam Maan","email":"gmaan.os14@gmail.com","username":"ghanshyam"},"ref":"refs/changes/69/840069/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/69/840069/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/840069/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/840069/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/840069/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/69/840069/1"}}},"commit":{"parents":[{"commit":"5e0974dc9608e35ce35627c3dce8635614765ad8","subject":"Merge \"Add Python3 zed unit tests\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5e0974dc9608e35ce35627c3dce8635614765ad8"}]}],"author":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 22:54:55.000000000","tz":-300},"committer":{"name":"Ghanshyam Mann","email":"gmann@ghanshyammann.com","date":"2022-04-30 22:54:55.000000000","tz":-300},"subject":"Drop lower-constraints.txt and its testing","message":"Drop lower-constraints.txt and its testing\n\nAs discussed in TC PTG[1] and TC resolution[2], we are\ndropping the lower-constraints.txt file and its testing.\nWe will keep lower bounds in the requirements.txt file but\nwith a note that these are not tested lower bounds and we\ntry our best to keep them updated.\n\n[1] https://etherpad.opendev.org/p/tc-zed-ptg#L326\n[2] https://governance.openstack.org/tc/resolutions/20220414-drop-lower-constraints.html#proposal\n\nChange-Id: Ic3be2ae9dd65991606f42c77acaf7470cd6a436e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cdeeb8b495bd539dd3aaf2b1d6d08c4636908f54"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cdeeb8b495bd539dd3aaf2b1d6d08c4636908f54"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":6618,"name":"Ruby Loo","email":"opensrloo@gmail.com","username":"rloo"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
