)]}'
{"id":"openstack%2Fsushy~840652","triplet_id":"openstack%2Fsushy~stable%2Fwallaby~I021d1116613e2b35352ca9749f8bc8a84bf655a1","project":"openstack/sushy","branch":"stable/wallaby","topic":"wallaby-839842-backport","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2022-05-10 10:35:15.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I021d1116613e2b35352ca9749f8bc8a84bf655a1","subject":"Use Etag in precondition header when setting boot options","status":"MERGED","created":"2022-05-05 11:52:14.000000000","updated":"2022-05-10 10:36:29.000000000","submitted":"2022-05-10 10:35:15.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"840652-wallaby-839842-backport","meta_rev_id":"b299f67401947bef86acac81de8db4b841c02d4e","_number":840652,"virtual_id_number":840652,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-05-10 10:35:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2022-05-05 12:21:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-05-10 09:37:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-05-05 11:52:14.000000000","updated_by":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"CC"},{"updated":"2022-05-05 11:53:30.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-05-05 12:21:56.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-05-05 12:57:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-05-05 14:05:21.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-05-10 09:37:34.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"69b478c65d51f6149cd7470df13227247af8b6aa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-05-05 11:52:14.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"181c2537d2120cd91f6c5adea9f6cda11788370e","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-05-05 11:53:30.000000000","message":"Patch Set 1:\n\nMerge Failed.\n\nThis change or one of its cross-repo dependencies was unable to be automatically merged with the current state of its repository. Please rebase the change and upload a new patchset.","accounts_in_message":[],"_revision_number":1},{"id":"ee6f6ff10faa701715af6e6aae3a418b7ae253f4","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-05-05 12:21:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"69aba5eae24b2b5ea46cbf2fba2a2ecf86acb04b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-05 12:57:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/18744757f783432bb76bafe017f8d1e9 : SUCCESS in 3m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8104b0751c524ed395a4c4756e74833d : SUCCESS in 3m 14s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3537f51b51b74902889222e9e099d816 : SUCCESS in 3m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5bfca22d6d674d76a941d1e9174f85ab : SUCCESS in 3m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/53075c658a8b4dacbaa5855ae92c0f56 : SUCCESS in 3m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/100e28ccf8da4f979039f8464e29e333 : SUCCESS in 5m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f78cd1eee2a64da5af766e93ff40f16d : SUCCESS in 3m 22s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/36306dce71134559a3588701440110d4 : SUCCESS in 57m 23s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f485ea21c33d45b88fc0fd396ee7d9bb : SUCCESS in 51m 16s","accounts_in_message":[],"_revision_number":1},{"id":"1154cba4395d512442d8f51f757a83e20e1d8ae3","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-05-05 14:05:21.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/d8b739e9d36f4ddc98baed44dbe7e87a : SUCCESS in 51m 01s\n- R6515-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/58ad3eb55b7f4f2ebe45ffa6960047bd : SUCCESS in 1h 38m 58s\n- R750-uefi-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/0cef4222ac824a11942a9e75e7bd243a : SUCCESS in 1h 27m 46s\n- XR11-bios-redfish https://stash-dellemc-community-org.us.dell.com/t/openstack/build/ff933df0592d4648a4c5ee37ec017077 : SUCCESS in 1h 33m 12s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"bec0edd92da1697781c8b3d50c23655b83f72f25","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-05-10 09:37:34.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"8d9b32ccc2bbea2c5020e721dd4be9cb54347490","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 09:38:22.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"22dec3f02c26a9c904921790b90f003bba4641ee","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:35:14.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5f3816bcef4c461e95e2802ed7e9a927 : SUCCESS in 2m 51s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/3ec763a338a74d9ea86b071e0a1fc698 : SUCCESS in 2m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/795bf95786d84188a055f9eba47e32d1 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a9ad0bd67ac4f729768b694b855cb21 : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e1418671ed9e4408a75d2ab6ba9bb365 : SUCCESS in 3m 31s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e6d7af89aa864020827e22fcb035a75a : SUCCESS in 49m 30s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c0d5b15a67be43a99925a54f4fc34848 : SUCCESS in 55m 42s","accounts_in_message":[],"_revision_number":1},{"id":"fd1484e4911d8b4a986042e65037a18ae16377b6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:35:15.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b299f67401947bef86acac81de8db4b841c02d4e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-05-10 10:36:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50fc7729a4f24ee589e1e7d0b4898652 : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5ee743d89fa041f9931018c5286b2d9e : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4e24f1414c222ea5eeb97c01e4e216bd7f5a285","revisions":{"e4e24f1414c222ea5eeb97c01e4e216bd7f5a285":{"kind":"REWORK","_number":1,"created":"2022-05-05 11:52:14.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/52/840652/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/52/840652/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/840652/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/840652/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/52/840652/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/52/840652/1"}}},"commit":{"parents":[{"commit":"8427417e208d7d7b48348b7eb9c36ff78d87613f","subject":"Fix session authentication issues","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8427417e208d7d7b48348b7eb9c36ff78d87613f"}]}],"author":{"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","date":"2021-11-16 13:48:45.000000000","tz":60},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-05-05 11:51:53.000000000","tz":600},"subject":"Use Etag in precondition header when setting boot options","message":"Use Etag in precondition header when setting boot options\n\nIf provided by the end point, use the Etag in the If-* precondition\nwhen setting the boot option.\n\nChange-Id: I021d1116613e2b35352ca9749f8bc8a84bf655a1\n(cherry picked from commit 7e2db9853baa1a73323e34899fcf5e8ac7b86290)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e4e24f1414c222ea5eeb97c01e4e216bd7f5a285"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e4e24f1414c222ea5eeb97c01e4e216bd7f5a285"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Verified\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Workflow","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Workflow\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","status":"SATISFIED","is_legacy":true,"submittability_expression_result":{"expression":"label:Code-Review\u003dCustom-Rule","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dCustom-Rule"],"failing_atoms":[],"atom_explanations":{}}}]}
