)]}'
{"id":"openstack%2Fsushy~851023","triplet_id":"openstack%2Fsushy~master~Ie42f68032fc94f98e5b7cb1b0ccea86609b1812c","project":"openstack/sushy","branch":"master","topic":"registries","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-23 13:56:08.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ie42f68032fc94f98e5b7cb1b0ccea86609b1812c","subject":"Better logging when registries are not available","status":"MERGED","created":"2022-07-26 11:53:23.000000000","updated":"2022-08-23 13:57:22.000000000","submitted":"2022-08-23 13:56:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":2,"has_review_started":true,"submission_id":"851023-registries","meta_rev_id":"11e1dd6ba1a510fc62286546b48575b1fe76267b","_number":851023,"virtual_id_number":851023,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 13:56:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":2,"date":"2022-08-23 12:52:18.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2022-08-11 17:11:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":1,"date":"2022-08-23 12:52:18.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-07-26 12:18:49.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-07-26 12:58:33.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-07-26 13:07:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-11 17:11:45.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2022-08-18 07:16:41.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"}],"messages":[{"id":"cd7de8ee569cb811ef370cd9eb4e3f9f01965939","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-07-26 11:53:23.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"eff78719d22bb51b5a53903af707a3b1eaec773d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-07-26 12:00:46.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"b35a8d710fd059fe3d5cc5f6582637cc420ed21b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-07-26 12:09:02.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"aa1ead281d616b2f8726c7dafff8fea821f0f3ff","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-07-26 12:18:49.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/23/851023/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/a4eb609/a4eb609.html : FAILURE in 7m 59s","accounts_in_message":[],"_revision_number":3},{"id":"48bf380d9e053ed8928721a3c24f96b326575f6d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-07-26 12:58:33.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d05447e99e5a49fd8e5f373d1ed2a09e : SUCCESS in 45m 20s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6d32876f33554498929363956d754640 : SUCCESS in 48m 13s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"6b53c0500dab0a28c1a2d38f10d1a235cb1ae4b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-07-26 13:07:35.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8de9379855034937b59d6f1f37a02637\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/977068a4cf4d418691b23a57b830a462 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7c4291fd30e4c04bec531604222edba : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5eea10cf693842de82496c5d1b4427d7 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fdfc68ee690a42fcb2112a975fd8d081 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fddc7d593f1c43be817536d24b58c96f : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b9179ab73e7437cb475f1c4af2c6881 : SUCCESS in 4m 36s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c587edf341934282a7c536eec502d32a : SUCCESS in 8m 11s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/827263ec4c31440cb9ece230a520aa62 : SUCCESS in 52m 11s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6275be2f8983415295ce39c6d9e1da82 : FAILURE in 36m 29s","accounts_in_message":[],"_revision_number":3},{"id":"8d3155c1b9d080062c5198780996f9d36a4aa4c5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-03 15:18:31.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"408a13e613eb271dccdb50f779f9b1c42d10be48","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 16:16:47.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73a5bff27a1a44f8a5b566da7bbcd3cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c344414564e4d54a675fe07c94845d0 : SUCCESS in 3m 46s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbc71b28d37b41ac8dec040bfd4f2b60 : SUCCESS in 3m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8137be8d5ac14f0e8aeb919ce79f0bb3 : SUCCESS in 3m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8755d17cf83b4b6d8a54c9d6eee7dff4 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebde5865fbeb4645bd7e61e1393a67c2 : SUCCESS in 3m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e28b66402795431da81589a15b071a76 : SUCCESS in 5m 32s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d42634a9b53e43e290a7d7da35d5f131 : SUCCESS in 10m 49s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/93d8b01e5ad4403fa3716f34d658f571 : SUCCESS in 51m 22s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3c925262179a43388f13623713f53808 : SUCCESS in 50m 15s","accounts_in_message":[],"_revision_number":4},{"id":"1c69731439f6d004cfa4e36ac962985bd411e3c0","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-03 23:54:20.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/0c35578cd6854c8c8b138895b42ab183 : FAILURE in 37m 51s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ca97d7ba868a46d481f69eaa7db87a87 : FAILURE in 1h 33m 51s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b56ba2cc7d3f4ac1ba47fa40f65c14fe : FAILURE in 59m 36s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a206777f2c7c41b28f30c518ca38e8f3 : FAILURE in 43m 36s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"e0f81eddf6b62fda9f4c1d1b11a59ec42a8ed8fb","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-08-11 17:11:45.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"2020e47ff76ead63eb75c1a2cddbf552e9b3823c","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-08-18 07:16:41.000000000","message":"Patch Set 4: Code-Review+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"175d5cc4618a684ffb524dfdf1c788a603f27c39","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-08-22 15:34:44.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":4},{"id":"10e767119b7679208544ddd69276f92d31c5d133","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-08-23 12:52:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b8fe7e7e7081cb04edbc2dfd94a6aa68c404a188","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 12:53:00.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"cc3ecb668a09f5971e3c18cacdcddaecee9e4c05","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 13:56:08.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6979195aa3f4eafa29cff94f1e2f691\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/927b9983ebfd43a7bf380f96de7477f6 : SUCCESS in 3m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b66df2beae9c4c6e954c33c8dd1a8667 : SUCCESS in 3m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c94ad737805a4975977ba57f75dd88ad : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/98b8b3f15abb48139988faad87ee1dc0 : SUCCESS in 5m 42s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/0b88f5ce8d534ea2b9efd1af9c663662 : SUCCESS in 8m 36s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c0eee6464def4c30a15b1d4632c4e7c0 : SUCCESS in 49m 10s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fcd6319aea8f4086abf4e9b46712a351 : SUCCESS in 59m 06s","accounts_in_message":[],"_revision_number":4},{"id":"9999b034b32bad6c2e23dea33383bdb97df3e9a9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 13:56:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"11e1dd6ba1a510fc62286546b48575b1fe76267b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 13:57:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a5a3ff49d5d947829830dc30af0f6dd7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/698dbd6d30b940c09008a897be56b9eb : SUCCESS in 52s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"738f7aa8c991670e94380ab258d60ab20c523c85","revisions":{"41025ae307d94348a68b3a9e6064884172ff3410":{"kind":"REWORK","_number":1,"created":"2022-07-26 11:53:23.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/23/851023/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/851023/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/851023/1"}}},"commit":{"parents":[{"commit":"10578574e8902f52e1cd81f49b007f58f9a5d30e","subject":"Handle AccessError with Basic Auth instead of \"reauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/10578574e8902f52e1cd81f49b007f58f9a5d30e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-26 11:51:12.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-26 11:53:10.000000000","tz":120},"subject":"Better logging when registries are not available","message":"Better logging when registries are not available\n\nLog which registries are available. Log differently when no registries are\navailable at all. Clarify which language does not match what.\n\nAlso move _get_registries to BaseResource since it\u0027s quite generic.\n\nChange-Id: Ie42f68032fc94f98e5b7cb1b0ccea86609b1812c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/41025ae307d94348a68b3a9e6064884172ff3410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/41025ae307d94348a68b3a9e6064884172ff3410"}]},"branch":"refs/heads/master"},"d0f521f20416fbfac07f25c7bbf135f8eb7e90b6":{"kind":"REWORK","_number":2,"created":"2022-07-26 12:00:46.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/23/851023/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/851023/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/851023/2"}}},"commit":{"parents":[{"commit":"10578574e8902f52e1cd81f49b007f58f9a5d30e","subject":"Handle AccessError with Basic Auth instead of \"reauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/10578574e8902f52e1cd81f49b007f58f9a5d30e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-26 11:51:12.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-26 12:00:19.000000000","tz":120},"subject":"Better logging when registries are not available","message":"Better logging when registries are not available\n\nLog which registries are available. Log differently when no registries are\navailable at all. Clarify which language does not match what.\n\nAlso move _get_registries to BaseResource since it\u0027s quite generic.\n\nChange-Id: Ie42f68032fc94f98e5b7cb1b0ccea86609b1812c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d0f521f20416fbfac07f25c7bbf135f8eb7e90b6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d0f521f20416fbfac07f25c7bbf135f8eb7e90b6"}]},"branch":"refs/heads/master"},"ef65b06320fe576317afeab87ab19143d27d3b6b":{"kind":"REWORK","_number":3,"created":"2022-07-26 12:09:02.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/23/851023/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/851023/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/851023/3"}}},"commit":{"parents":[{"commit":"10578574e8902f52e1cd81f49b007f58f9a5d30e","subject":"Handle AccessError with Basic Auth instead of \"reauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/10578574e8902f52e1cd81f49b007f58f9a5d30e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-26 11:51:12.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-26 12:07:31.000000000","tz":120},"subject":"Better logging when registries are not available","message":"Better logging when registries are not available\n\nLog which registries are available. Log differently when no registries are\navailable at all. Clarify which language does not match what.\n\nAlso move _get_registries to BaseResource since it\u0027s quite generic.\n\nChange-Id: Ie42f68032fc94f98e5b7cb1b0ccea86609b1812c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ef65b06320fe576317afeab87ab19143d27d3b6b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ef65b06320fe576317afeab87ab19143d27d3b6b"}]},"branch":"refs/heads/master"},"738f7aa8c991670e94380ab258d60ab20c523c85":{"kind":"REWORK","_number":4,"created":"2022-08-03 15:18:31.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/23/851023/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/851023/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/851023/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/851023/4"}}},"commit":{"parents":[{"commit":"f804e954963446984a57c35d010011449669f67e","subject":"Handle mismatch between registry Identity and the way it\u0027s referred to","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f804e954963446984a57c35d010011449669f67e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-26 11:51:12.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-03 15:18:00.000000000","tz":120},"subject":"Better logging when registries are not available","message":"Better logging when registries are not available\n\nLog which registries are available. Log differently when no registries are\navailable at all. Clarify which language does not match what.\n\nAlso move _get_registries to BaseResource since it\u0027s quite generic.\n\nChange-Id: Ie42f68032fc94f98e5b7cb1b0ccea86609b1812c\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/738f7aa8c991670e94380ab258d60ab20c523c85"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/738f7aa8c991670e94380ab258d60ab20c523c85"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}}]}],"submit_requirements":[]}
