)]}'
{"id":"openstack%2Fsushy~851884","triplet_id":"openstack%2Fsushy~stable%2Fxena~Id070b2e9916ca3102d708d23dc1594df27523e93","project":"openstack/sushy","branch":"stable/xena","topic":"reg-name-stable/xena","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-08 10:10:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id070b2e9916ca3102d708d23dc1594df27523e93","subject":"Handle mismatch between registry Identity and the way it\u0027s referred to","status":"MERGED","created":"2022-08-03 15:15:49.000000000","updated":"2022-08-08 10:11:51.000000000","submitted":"2022-08-08 10:10:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"851884-reg-name-stable/xena","cherry_pick_of_change":851392,"cherry_pick_of_patch_set":1,"meta_rev_id":"b8eaa4fddc6be1c64c76a10bf1bd66a729d57515","_number":851884,"virtual_id_number":851884,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-08 10:10:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":2,"date":"2022-08-04 07:04:47.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2022-08-08 08:58:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2022-08-08 08:58:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-03 16:22:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-03 23:12:02.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-04 07:04:47.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2022-08-08 08:58:17.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"34a1d63a20c2a014e9d73d163523241d290b37fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-03 15:15:49.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"97ca3f3b6224f8f5b75280f4bbb8bb98204e94f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-03 16:22:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6a4f0377e2914b468148b8d5b94b2445\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/956f226753bc490ea55ce5b3b0f7ee7b : SUCCESS in 3m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77c45c4b86b4099924c33d58624aa29 : SUCCESS in 2m 38s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8516945a22294458ab8393601ad18822 : SUCCESS in 3m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3f8a0a3ac5a24cd2ac072deb36bf622c : SUCCESS in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c29422c8c86e4214936c9d7af331c2d9 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6bc6c32eaa7948c38741ab0869f6e642 : SUCCESS in 4m 47s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c660b3267e3d4c21a3856f2d2cf9789a : SUCCESS in 50m 27s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fb5bc80fc21f41e0bd106715b79a0a1a : SUCCESS in 58m 59s","accounts_in_message":[],"_revision_number":1},{"id":"0e034e9c0479e60a3a60738a84f3fed682552da4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-03 23:12:02.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a7c56577d15f46eaa0071dbb2389a263 : FAILURE in 1h 13m 14s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/987528b3f9c94775bd67befac01efa7b : FAILURE in 31m 40s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4e655ba366bb49af9d4b72ebee4388d2 : SUCCESS in 1h 09m 49s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ccfc37a9a5eb4b399c2147e669918b16 : SUCCESS in 57m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"74ff78ea98c7add83992d9ad1c6321fb8cf17e19","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-08-04 07:04:47.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"90dd98751b07f2d613e3b3b23a4dca7041facdb3","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-08 08:58:17.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d93482a87a7f38821ad3c34d0d31af2b1857aeb9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 08:59:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b09d88a2ff5d16b943ce2273bd3399dec888b5a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:10:37.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8722874901274dccac4e8bd853e16ff0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a99179bcaa504c3cb8c052c48bcf3bc3 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/5cc9b69814954239aba9d6c9483dabfb : SUCCESS in 6m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f92c3d7c033f4ee89d98c2c14186bebc : SUCCESS in 6m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f6de3e872b4f3ba6fbc8e2d46804a7 : SUCCESS in 8m 59s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bc93c322d6e64174986138cfd068edbc : SUCCESS in 1h 07m 00s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/40cb4d7e6abe415187bb0fbc704412f0 : SUCCESS in 1h 07m 22s","accounts_in_message":[],"_revision_number":1},{"id":"a65d465616f4004e01f34220132e8e6255815db5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:10:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b8eaa4fddc6be1c64c76a10bf1bd66a729d57515","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-08 10:11:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f4a6d975345456195c5c22df9f2f4cd\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5735b502b9cf4781aad3839450ad014d : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"b32ab9b705acc897fb7dac9f8d88635fe3c81cb7","revisions":{"b32ab9b705acc897fb7dac9f8d88635fe3c81cb7":{"kind":"REWORK","_number":1,"created":"2022-08-03 15:15:49.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/84/851884/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/84/851884/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/84/851884/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/84/851884/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/84/851884/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/84/851884/1"}}},"commit":{"parents":[{"commit":"eb80989675aeceaca6b39e537d31dc17dba05377","subject":"Handle AccessError with Basic Auth instead of \"reauth\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eb80989675aeceaca6b39e537d31dc17dba05377"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-07-28 12:00:08.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-03 15:15:49.000000000","tz":0},"subject":"Handle mismatch between registry Identity and the way it\u0027s referred to","message":"Handle mismatch between registry Identity and the way it\u0027s referred to\n\nSeen on some SuperMicro hardware: the Identity is\nBiosAttributeRegistry.1_0_0, but it\u0027s referred to as\nBiosAttributeRegistry.v1_0_0 (note the \"v\").\n\nIn my opinion, it\u0027s a firmware issue, but it\u0027s easy to work around.\n\nChange-Id: Id070b2e9916ca3102d708d23dc1594df27523e93\n(cherry picked from commit f804e954963446984a57c35d010011449669f67e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b32ab9b705acc897fb7dac9f8d88635fe3c81cb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b32ab9b705acc897fb7dac9f8d88635fe3c81cb7"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
