)]}'
{"id":"openstack%2Fsushy~853431","triplet_id":"openstack%2Fsushy~stable%2Fyoga~Idb2d6bee96bd23cfc4fb8f014661231766998781","project":"openstack/sushy","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-23 09:49:35.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb2d6bee96bd23cfc4fb8f014661231766998781","subject":"Do not send compression as acceptable encoding","status":"MERGED","created":"2022-08-17 10:35:33.000000000","updated":"2022-08-23 09:50:53.000000000","submitted":"2022-08-23 09:49:35.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853431","cherry_pick_of_change":850547,"cherry_pick_of_patch_set":4,"meta_rev_id":"9d650343a738a0a4c5442a8a482f2147afec32d5","_number":853431,"virtual_id_number":853431,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 09:49:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":2,"date":"2022-08-17 12:31:52.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2022-08-23 08:53:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2022-08-23 08:53:12.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 10:35:33.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2022-08-17 10:41:12.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-08-17 11:49:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 12:19:29.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-17 12:31:52.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2022-08-23 08:53:12.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"673b50e103bc8d44a46398782ed58e961a8964d1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-17 10:35:33.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"feb8ce0afff15a5cd20759c3b1fd04cf99da0f0e","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-08-17 10:41:12.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/31/853431/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/175af6c/175af6c.html : FAILURE in 4m 19s","accounts_in_message":[],"_revision_number":1},{"id":"29b332fde697d0ebf1b59ec850191388a3e2b131","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 11:49:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/43e23b22011d4c3bade62e107db9c3cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c46474fa209549eba506f08a22342aa7 : SUCCESS in 6m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bacda6ee4a240df82d550e03d3fec14 : SUCCESS in 5m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1102cc54028b49c390d8d77201043249 : SUCCESS in 7m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c2cfa0c5fe4f41eab1ad8dfded97a430 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/67ba840567a842df88c2849efc6ba564 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b96c8a835f243ac86996a2fccf470ba : SUCCESS in 5m 20s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c61fb8ceed68474db34788ddae7df52f : SUCCESS in 9m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bf884e44e15e44fd8a31fd3731940a00 : SUCCESS in 56m 25s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/038445e65b1e446290c3aed754318274 : SUCCESS in 1h 06m 25s","accounts_in_message":[],"_revision_number":1},{"id":"b5cecde26664e5d83e9aca657fcab6e3f6ff98e0","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-17 12:19:29.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/445739e596bf4c5aaa66d8fa833c5983 : SUCCESS in 51m 50s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e2a5d53befa14aefbbfdb4cf8a01bf1d : SUCCESS in 1h 42m 38s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2bbf0109b6814844ba14ed5db2ba865b : FAILURE in 1h 34m 45s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/91c3275c3b99482896dc7ddf3401cd27 : SUCCESS in 51m 49s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"ec5baa79128bacfe6f2e428f464175dc50621aee","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-08-17 12:31:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"662aa048d4642137ff950ea7f879a529b9681566","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-23 08:53:12.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"187876d56bab58ff8d5cbbadcbcb52d9904ba975","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 08:53:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d938e2d666780883481c2e818054864542ad8708","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:49:35.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8e61cd035d940968b5aa25c98d57c80\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/171c95362b6a496a9ef40f23a94c2168 : SUCCESS in 3m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/fe7a92c6f33b4d07944b33de59ef637a : SUCCESS in 4m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e084c442139d4b86ad40652e6bebde87 : SUCCESS in 3m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f800a52627494dd882fc9eb8d2e3b45e : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c8068fcd8ab749e186ced639c0ede09a : SUCCESS in 3m 53s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a1ee5f9c33804c2993b1446a5fd5ec54 : SUCCESS in 8m 23s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/652c1739961342b9ac4ec52b61e94eee : SUCCESS in 47m 31s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/2b4aee7ea78d46e18b2aea7885617517 : SUCCESS in 54m 40s","accounts_in_message":[],"_revision_number":1},{"id":"675ef61fca330a0d1125222899772920699abed7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:49:35.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9d650343a738a0a4c5442a8a482f2147afec32d5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:50:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54e386454eb74a9689f6eaaa844be4e7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/627ea622adfb4c8b91a1b5a8feae8fa7 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4d3e51b9582041d680ae3a9ea6d2a83c : SUCCESS in 51s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"56182b39cf5748aea599281677dd3ddb07b6d905","revisions":{"56182b39cf5748aea599281677dd3ddb07b6d905":{"kind":"REWORK","_number":1,"created":"2022-08-17 10:35:33.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/31/853431/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/31/853431/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/31/853431/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/31/853431/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/31/853431/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/31/853431/1"}}},"commit":{"parents":[{"commit":"96fc2057dcfb544ed12afe6e5ed2e289e7fbaf76","subject":"Handle mismatch between registry Identity and the way it\u0027s referred to","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/96fc2057dcfb544ed12afe6e5ed2e289e7fbaf76"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-07-20 17:05:22.000000000","tz":-420},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-17 10:35:33.000000000","tz":0},"subject":"Do not send compression as acceptable encoding","message":"Do not send compression as acceptable encoding\n\nWhen a BMC can compress a response, the internal server... or any\nintermediate proxy may change a \"Strong\" ETag to a \"Weak\" ETag, as\nthe strength of the ETag represents the byte-by-byte consistency of the\nresponse body.\n\nCompressed responses may differ slightly, especially if different encoding\nis used or negotiated. This is problematic when a resource expects a\nfollow-up request or action to contain a \"Strong\" ETag as it is the client\ncertifying it understood the prior precise state.\n\nWe no longer request opportunistic response encoding, and expressly request\nresponses are not modified or compressed by the intermediate server, which\nshould have a minimal impact to use and provide a higher level of BMC\ninteraction consistency.\n\nChange-Id: Idb2d6bee96bd23cfc4fb8f014661231766998781\n(cherry picked from commit 51dacf272f4d0f3e346c69e94786da2b7ee7c935)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/56182b39cf5748aea599281677dd3ddb07b6d905"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/56182b39cf5748aea599281677dd3ddb07b6d905"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
