)]}'
{"id":"openstack%2Fsushy~853432","triplet_id":"openstack%2Fsushy~stable%2Fxena~Idb2d6bee96bd23cfc4fb8f014661231766998781","project":"openstack/sushy","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-08-23 09:53:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Idb2d6bee96bd23cfc4fb8f014661231766998781","subject":"Do not send compression as acceptable encoding","status":"MERGED","created":"2022-08-17 10:35:44.000000000","updated":"2022-08-23 09:54:13.000000000","submitted":"2022-08-23 09:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"853432","cherry_pick_of_change":850547,"cherry_pick_of_patch_set":4,"meta_rev_id":"9190e57752d4e4ba2e7ce05756461f1b8a7fd290","_number":853432,"virtual_id_number":853432,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-08-23 09:53:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":2,"date":"2022-08-17 12:32:33.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2022-08-23 08:53:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2022-08-23 08:53:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-08-17 10:35:44.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2022-08-17 10:41:30.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-08-17 11:27:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-08-17 12:15:47.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-08-17 12:32:33.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2022-08-23 08:53:21.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"e1e07d9c55fde6a3c18cfa8990f2ea7ab281247a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-08-17 10:35:44.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"787a8ad5a9aef8c63a4936618b0db98591c9afe4","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-08-17 10:41:30.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/32/853432/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/c17ae2c/c17ae2c.html : FAILURE in 4m 21s","accounts_in_message":[],"_revision_number":1},{"id":"72a752d0644e057e0a31ef914d4f9c4f7f649aad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-17 11:27:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3027e7c84e6444249c6901ef36f628af\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f5dccf9dc7f43aca43aede5434858b6 : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26b4cc65fc15431ab2211eb967f0b040 : SUCCESS in 3m 29s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6c4931319a1a4a61b496ce6b1f465a9e : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/520e8d1e6a8f456dbf898b0c0802663e : SUCCESS in 4m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cfea50389bb4e769cf226d6600bfd91 : SUCCESS in 3m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/faf2ef988ff546769e43f60c9e444989 : SUCCESS in 5m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5cbe756d540d4c5fa9197c90d40f37d1 : SUCCESS in 4m 03s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/390dea46513d4c4486567955caaaf6c2 : SUCCESS in 50m 47s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/99cdf827a95241f18da8f1074862b459 : SUCCESS in 48m 44s","accounts_in_message":[],"_revision_number":1},{"id":"e1285587f3f17cb2c7d4320c30d204f0c13c6fe4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-08-17 12:15:47.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5bb2085f1c6e4e5889e256b971b5076e : SUCCESS in 49m 00s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/747a2d4182c34f76a921bef26400615d : SUCCESS in 52m 51s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a8d1ebc1e21548e982959c5a6976978c : FAILURE in 1h 12m 59s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/dd10c7464b73495aac3528eca4effe76 : SUCCESS in 48m 14s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"1fd09c21e1c9faaab864334f5dc3ce55463be468","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-08-17 12:32:33.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6d3e0ea1093de096fb58440e7300c55485e2629d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-08-23 08:53:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2a4c7aee6b4b2cfbcd2deb9bef77d5c0abfb1547","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 08:54:02.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7ef5b735d8a56c73545aa31455f4ed7aa011a47a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:53:00.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/15a019c3c0494d19b109e6f626694d3c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1598332336f941da8b33c3a43deef144 : SUCCESS in 3m 33s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0ac758e1cb30484a9ea8aa7e659ae53a : SUCCESS in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ad4165c7dcce494486fb275d4a687d5c : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c4c2a566d96147d192993b71cf8bd130 : SUCCESS in 4m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c4d2ec8706b347de87fc499cdc9c4450 : SUCCESS in 3m 48s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8b4c4087fa894cc789d1ddd057f57526 : SUCCESS in 58m 38s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1e35619867684493aaeb700e3bf34b96 : SUCCESS in 54m 50s","accounts_in_message":[],"_revision_number":1},{"id":"b0414add44407c71536646fcabfa8bfa6f1d6b25","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9190e57752d4e4ba2e7ce05756461f1b8a7fd290","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-08-23 09:54:13.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/848f0aa04781461cbd2fed1b625022b7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7741e718c4e474cb668749686be4e70 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/df56dc844c0147e9a30991f71c2e98c0 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"021e0175e2ffc7c31e7052fb3bc39c9e28b83783","revisions":{"021e0175e2ffc7c31e7052fb3bc39c9e28b83783":{"kind":"REWORK","_number":1,"created":"2022-08-17 10:35:44.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/32/853432/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/32/853432/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/32/853432/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/32/853432/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/32/853432/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/32/853432/1"}}},"commit":{"parents":[{"commit":"b32ab9b705acc897fb7dac9f8d88635fe3c81cb7","subject":"Handle mismatch between registry Identity and the way it\u0027s referred to","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b32ab9b705acc897fb7dac9f8d88635fe3c81cb7"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2022-07-20 17:05:22.000000000","tz":-420},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2022-08-17 10:35:44.000000000","tz":0},"subject":"Do not send compression as acceptable encoding","message":"Do not send compression as acceptable encoding\n\nWhen a BMC can compress a response, the internal server... or any\nintermediate proxy may change a \"Strong\" ETag to a \"Weak\" ETag, as\nthe strength of the ETag represents the byte-by-byte consistency of the\nresponse body.\n\nCompressed responses may differ slightly, especially if different encoding\nis used or negotiated. This is problematic when a resource expects a\nfollow-up request or action to contain a \"Strong\" ETag as it is the client\ncertifying it understood the prior precise state.\n\nWe no longer request opportunistic response encoding, and expressly request\nresponses are not modified or compressed by the intermediate server, which\nshould have a minimal impact to use and provide a higher level of BMC\ninteraction consistency.\n\nChange-Id: Idb2d6bee96bd23cfc4fb8f014661231766998781\n(cherry picked from commit 51dacf272f4d0f3e346c69e94786da2b7ee7c935)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/021e0175e2ffc7c31e7052fb3bc39c9e28b83783"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/021e0175e2ffc7c31e7052fb3bc39c9e28b83783"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
