)]}'
{"id":"openstack%2Fsushy~855799","triplet_id":"openstack%2Fsushy~master~Iee4fa1c509dfffb57a0ea871c8a05dbb9f1f2849","project":"openstack/sushy","branch":"master","topic":"release-health-check","attention_set":{},"removed_from_attention_set":{"17685":{"account":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"last_update":"2022-09-07 09:20:25.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Iee4fa1c509dfffb57a0ea871c8a05dbb9f1f2849","subject":"DNM: dummy change to test gate health","status":"ABANDONED","created":"2022-09-03 20:51:43.000000000","updated":"2022-09-07 09:20:25.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"054873dbef889c3baeee1659fb18d2b702d4b32c","_number":855799,"virtual_id_number":855799,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2022-09-03 21:47:24.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-03 21:31:58.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-09-03 21:47:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-03 21:50:03.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"}],"messages":[{"id":"35a5bd12f345d5cbd4d5c6938e3435ccd96dd431","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-03 20:51:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bed3b9a8a22c73d803d517b9ea09b0b7c1ac341c","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-03 21:31:58.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/99/855799/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/8bfc716/8bfc716.html : SUCCESS in 39m 01s","accounts_in_message":[],"_revision_number":1},{"id":"54caf31551db93d6f99179fcd643819308667daf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-03 21:47:24.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6192586ffa834655853f72139c8a9e1f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb19d0a1d2ab493a9cba999414e772b9 : SUCCESS in 4m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e034f0b598f4bd7be9122434f487ea6 : SUCCESS in 3m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d7aad1cd0f714285b1b39aedbe359c86 : SUCCESS in 3m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/de532c74f54643808b3b636d0f0aa232 : SUCCESS in 4m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9dc1342525624b7cbd2599637cd578a4 : SUCCESS in 4m 23s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b582ae2de39741c08ddbbba94992a0bb : SUCCESS in 6m 01s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/66d5f713694449f9a63acfbd88e420c6 : SUCCESS in 8m 59s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/421181114dad41f98e8edb93a311bf8e : SUCCESS in 50m 00s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ca39176bc3c14dabbd2a028d5e6964e7 : SUCCESS in 51m 35s","accounts_in_message":[],"_revision_number":1},{"id":"06d73e0329eb9a7753e156d96ed775f491c636cd","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-03 21:50:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/96eca0bbbe664a508ae8e8ce8790d76d : SUCCESS in 48m 35s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9080a261a2ae4a33abbade67fd68a156 : FAILURE in 55m 12s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/bbc6417f791b42b6a92b01b6f10da15f : SUCCESS in 49m 21s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4eab3e22c3544a72b52da8b3b955ffbe : SUCCESS in 48m 51s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"054873dbef889c3baeee1659fb18d2b702d4b32c","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2022-09-07 09:20:25.000000000","message":"Abandoned\n\nGate check passed, no further need for this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c37ad631a18490f69085d11cdf2c1ea4b640b9f6","revisions":{"c37ad631a18490f69085d11cdf2c1ea4b640b9f6":{"kind":"REWORK","_number":1,"created":"2022-09-03 20:51:43.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/99/855799/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/99/855799/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/855799/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/855799/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/99/855799/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/99/855799/1"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:51:30.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2022-09-03 20:51:41.000000000","tz":120},"subject":"DNM: dummy change to test gate health","message":"DNM: dummy change to test gate health\n\nChange-Id: Iee4fa1c509dfffb57a0ea871c8a05dbb9f1f2849\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c37ad631a18490f69085d11cdf2c1ea4b640b9f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c37ad631a18490f69085d11cdf2c1ea4b640b9f6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[]}
