)]}'
{"id":"openstack%2Fsushy~856123","triplet_id":"openstack%2Fsushy~master~Iadf19b77ff3132b0533d317e46032b9d5a0e4e33","project":"openstack/sushy","branch":"master","topic":"etag-fallbacks","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2026-10-05 01:34:36.000000000","reason":"\u003cGERRIT_ACCOUNT_32177\u003e replied on the change","reason_account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"}},"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2022-10-27 11:58:47.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-10-27 10:53:50.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},"11292":{"account":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"last_update":"2022-10-27 11:58:47.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-10-27 11:58:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Iadf19b77ff3132b0533d317e46032b9d5a0e4e33","subject":"Improve resiliency of eTag handling","status":"MERGED","created":"2022-09-06 13:57:50.000000000","updated":"2026-10-05 01:34:36.000000000","submitted":"2022-10-27 11:58:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":79,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856123-etag-fallbacks","meta_rev_id":"ec95b3381078292901f4b10564cd26bb4e094e8b","_number":856123,"virtual_id_number":856123,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-27 11:58:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2022-10-27 10:53:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-10-26 11:10:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2022-10-27 10:53:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"}],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"}],"CC":[{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-06 14:05:20.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-09-06 14:08:40.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2022-09-06 14:30:33.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2022-09-06 14:51:45.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-06 15:33:03.000000000","updated_by":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"reviewer":{"_account_id":11292,"name":"Arne Wiebalck","email":"Arne.Wiebalck@cern.ch","username":"wiebalck"},"state":"REVIEWER"},{"updated":"2022-09-06 15:52:42.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-10-19 06:45:41.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2022-10-21 10:07:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2025-10-02 04:30:23.000000000","updated_by":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"reviewer":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"state":"CC"}],"messages":[{"id":"c475f5345637ca9f5e13335facef1a02a591934f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-06 13:57:50.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1fd621beca1ad6e669af5ed071470582ff1dee9c","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-06 13:58:58.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e85ea1a96d0de29e39efb2a023c5bf282f33d72a","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-06 14:05:20.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/23/856123/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/5ab723e/5ab723e.html : FAILURE in 5m 52s","accounts_in_message":[],"_revision_number":1},{"id":"18b795a9afef2970bc746e8d5c1c0e70342f33f2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-09-06 14:08:40.000000000","message":"Patch Set 1:\n\n(4 comments)","accounts_in_message":[],"_revision_number":1},{"id":"c3ab3359e0c70ab7aa305e1c8002dab74376d1fa","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-09-06 14:30:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fa1420a36a17f643b512fb096c4c95875046dac3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 14:51:45.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67432e50de4e47ac94206c7a50eec0c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ff67cff41d74d6e969ddf9c78ddb256 : SUCCESS in 7m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86a673146ad840aa9aa6b0dbefc12825 : SUCCESS in 6m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c9144da5df384bb4bbf78a2a4021aea0 : SUCCESS in 6m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7fac6fa20a6462d8f07f6fa6ff0665f : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/19ce2a1bfdb7412aabd375ce48d5965a : SUCCESS in 6m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/59c80942a5f74759818fbb2ba7034b32 : SUCCESS in 9m 27s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/28f52c23917148ccab91cbbdf68176e2 : SUCCESS in 8m 31s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7c968ab736fc4a1296563a33a7684389 : FAILURE in 46m 30s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5519bf8646ec49d2a7e82205dad3c852 : FAILURE in 51m 21s","accounts_in_message":[],"_revision_number":1},{"id":"fba3e3d1fe84a4f7d3cb6eeeefaaeb67eb581a17","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-06 15:52:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9e68fecc67b045b9b459403a019e621a : FAILURE in 36m 53s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b3a9f28594bc42998b94e60b7f18e748 : FAILURE in 1h 22m 27s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e8bcdad8794f462f8d1ca1c36cf9836b : FAILURE in 34m 02s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d1ab35747fdb46428ec6eaf82d43778d : FAILURE in 32m 47s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"47593177f2b31b43160fa3665b8a6c04103e412e","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-06 22:28:36.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"93bc9beb1b4b1f416868c1befea20c9895a7d158","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-06 23:11:58.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/235afa2e13a747cba51d772c69a3ca05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/06b8382919004d2681538e83f7503037 : SUCCESS in 4m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c194d3d1167249d2a134ecf4e23105bc : SUCCESS in 3m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ef768aa0727475da388c0a89f31c58e : SUCCESS in 4m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f28bcb5060db4015954f38a571400fcc : SUCCESS in 4m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b01013d20a174a30a14ea7d9b0054655 : SUCCESS in 4m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bcb2d99e1bf34cfd9840f77055fc19ca : SUCCESS in 5m 41s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/12baac9382ac4650868a8b35508c194e : SUCCESS in 8m 51s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/876a5f03d6c24382a8e42a2b612f2f9a : FAILURE in 36m 40s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e7cc6ba1630b4c3e9f4c41c8078521c6 : FAILURE in 41m 52s","accounts_in_message":[],"_revision_number":1},{"id":"1ac2f675981ae415e197c00117b93eec1dcc9d29","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-09 01:42:15.000000000","message":"Patch Set 1:\n\n(3 comments)","accounts_in_message":[],"_revision_number":1},{"id":"4bb3d2916798d9f5ba3b5887b6a2639b8233b0f6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-09 04:02:33.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"c884bf1a0365ce577355070843c91251193a5ae6","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-09 04:02:49.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"65d952e656148f8f14aa583ea0f54bbcf5ef32f4","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-09 04:02:55.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"2aab8453fa92a5ec17b3407789750c34586c0560","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-09 04:10:22.000000000","message":"Patch Set 2:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/23/856123/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/c251509/c251509.html : FAILURE in 5m 54s","accounts_in_message":[],"_revision_number":2},{"id":"6afb66a920f872182f2251975a4840e3a2c25d70","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-09 04:38:50.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/bda5c42a43ab4c779e164b5a170da01e : FAILURE in 34m 26s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e8c7464d98e24a0fbd0f8c9a40b3a34c : FAILURE in 34m 18s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/27c5ded4b1334c039f0dbf64aafdc3e1 : FAILURE in 34m 22s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/fd8250f295b2422d9d41e1aba4b4cd18 : FAILURE in 34m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"53996c511b1cf4447841d5c3ebcabe5072ca5a0a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 04:46:56.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/89edc34a85634ae48093cb373d871a8f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f06c51fab474975b627a130414951c0 : SUCCESS in 3m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b71487dace34fe28e03d8bf3c0c529e : SUCCESS in 3m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b67318b985c44285a1e56af4afc9ea9b : SUCCESS in 3m 34s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42c0a8b8150b4acc855c6c5fba7645ef : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02cb17108e6b4f7e948b6a758ae4e4c6 : SUCCESS in 5m 19s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1abb9529266342f7bfe7b574a070af82 : SUCCESS in 7m 36s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7482f23b4d2c47fd843027bb9a170146 : FAILURE in 39m 15s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/eedc9515bb9641af94f59263d19087df : FAILURE in 33m 07s","accounts_in_message":[],"_revision_number":2},{"id":"64c8e1f965ab339436b582ec4c9ec192066f198d","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-12 05:01:27.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c459c24518ff6ec22d3abb195142990e80d71d31","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 05:55:07.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f0b5f72672c4276bb6d4c7f58250c6a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9cd2a46679074b71842b950f74776cc9 : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dfa35fb1cf884fca8da3a065c4516e96 : SUCCESS in 2m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8c34b6f8262e49058a4bc0572452f44f : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a07a2f5786b8409ca6dfff1326be3f81 : SUCCESS in 3m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4321711ca9234ab68f10506fe45c4a3f : SUCCESS in 5m 59s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/bf0e5f17c628447a96c27e481ecdab91 : SUCCESS in 9m 23s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0634f6f6003f4809b1cbdfb2adc5a0bc : FAILURE in 46m 33s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/845dd257776a479c98657356e8fdb51c : FAILURE in 46m 50s","accounts_in_message":[],"_revision_number":2},{"id":"bd465b9395c974969c5083001d0340bfe93a2d64","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-12 21:52:56.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"dcac5ed06001eaa8075ff3775b48944d4f9db2b5","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-12 22:16:09.000000000","message":"Patch Set 3:\n\n(3 comments)","accounts_in_message":[],"_revision_number":3},{"id":"526786dc3332c8a67fdb63a5da20f38c8e960ab2","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-12 22:17:44.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"066ae529796ef07dd556f854e495f2eb9ce49bb2","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-12 22:34:47.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/23/856123/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/ef6956f/ef6956f.html : SUCCESS in 40m 29s","accounts_in_message":[],"_revision_number":3},{"id":"a8a08c97c1020319e5b6904d452327c4223c48b8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-12 22:38:05.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58a4aa5a8e924d26b70b9c7dc07da019\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/56a9814f130a4f7d8141b84d8081468c : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea37b429bc24408ebe85fb1baa4b201e : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/67af3e57fdc043a7a19bb284e665d28e : SUCCESS in 3m 50s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1476a1f633e44b699d8f1a295eba875b : SUCCESS in 3m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/161d4f3636114b61a6d881a5f2523eec : SUCCESS in 5m 36s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9d340ac2a0b54bcd87491a5ba6024278 : SUCCESS in 7m 04s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ce718ceb7bfe46969850a32fcaadb412 : SUCCESS in 33m 22s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a67b0cfb85d04374a07691d2cbb744b8 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":3},{"id":"3b3da97a13bbc2f5a4a6290ee0ea4ef836b68922","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-12 22:47:06.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/0eb9668cebb74d7c9adbd721cc3fe85f : SUCCESS in 50m 34s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3e2c616ae2a34bbebe6fded4965b40d7 : SUCCESS in 52m 31s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/795441844f154327aa0831165761af0c : SUCCESS in 51m 58s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/0ed9260ba22b4135baebd182a5fee2e0 : SUCCESS in 51m 27s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"45f41bc4187ad29464ce368885309c60c4f5a4bc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-09-13 11:00:57.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"4f858ad0f7ea02eaa439dcaf6da96766c82aa2b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-13 11:36:05.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"da9771ccda4a7ae0c9453abb2fc89c0f8ae3a396","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-13 11:37:14.000000000","message":"Patch Set 4:\n\n(3 comments)","accounts_in_message":[],"_revision_number":4},{"id":"cd4ab77468fda528ea1d8bd5a26d453e9455dd2e","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-13 11:42:36.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/23/856123/4/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/09cdf53/09cdf53.html : FAILURE in 5m 10s","accounts_in_message":[],"_revision_number":4},{"id":"83fc33867abe8552a7da1c460e6b4de440e13210","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-13 12:28:16.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3cfade0f7ef34015aee277987bb6be7e : FAILURE in 23m 35s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c327d0383ac24bb8b425e0730865bc0f : SUCCESS in 50m 54s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6a17bee84ae541efa41ecb5617cab155 : SUCCESS in 49m 10s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3e07a129632645dd80d51c2bc6a25fda : SUCCESS in 48m 49s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"7d2ad244c35cf40ea6512b92cf7898afc481f47b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-13 12:31:15.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/473f5bac1e00430e847bd3b8695b2594\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/be416f3001da4f79bdcabfc1df83c00f : SUCCESS in 4m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6504468d4a0b40bdb29f57a907bf8633 : SUCCESS in 3m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/799d112e951c42eabeac4db2dc38efd0 : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ad6ac01e7e5e42949173bdeb8df98a63 : SUCCESS in 4m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c5646aca90342dfb3d8984d0e85cf5f : SUCCESS in 6m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ba7dddb825a465daee59c978df3aac4 : SUCCESS in 4m 26s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e78e45d9200e49a4ad83d497c28c2546 : SUCCESS in 8m 19s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6e9ce1569e7d40ad9007d3c1c3122969 : SUCCESS in 39m 21s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8031183a54cb4cb3a1b2ee345944cf16 : SUCCESS in 53m 11s","accounts_in_message":[],"_revision_number":4},{"id":"efd09d88a5725e40cc649b342d712141aa85c048","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-14 00:00:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"d966ccc98b229c1520b619620d59489f42e0109b","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-14 00:41:51.000000000","message":"Patch Set 5:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/23/856123/5/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/5ec262a/5ec262a.html : SUCCESS in 40m 19s","accounts_in_message":[],"_revision_number":5},{"id":"e01c7ea383f06a6c475922e23d19b350971ccfea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-14 00:44:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f532b0a6793d4fd9b30c8afc7cfa34a0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fcb8ff76f0f54716ba45d760496a86bd : SUCCESS in 4m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0952d5bc476443ceb695f7d732cf3240 : SUCCESS in 3m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39945c39a2a1408c9ee4905cf861d30d : SUCCESS in 4m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3440cfa8a7004375ba224bec08b8a1e1 : SUCCESS in 5m 45s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39dd077e78064f4ba77342eadcf3e539 : SUCCESS in 6m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a3c5f574c06243dc8d3aa33df60c0176 : SUCCESS in 4m 18s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c6929479b0f54c8e8f20b6bd76fb3f43 : SUCCESS in 9m 10s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ce9d5a4d423f45a7b9bf96c1cabc0546 : SUCCESS in 33m 26s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b9bd0c3ec613446cb1db3af1b9c9c72b : SUCCESS in 43m 20s","accounts_in_message":[],"_revision_number":5},{"id":"1b71234cfc3112a695c562321a06695e93df989f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-14 01:18:03.000000000","message":"Patch Set 5:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6d9ee0937229461d9e1c9d3eef3c3e42 : SUCCESS in 47m 56s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f3dba66eb9ae4a4a8a807b33b6eb252f : SUCCESS in 1h 13m 17s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/163f1cbdce8e43b0944a789aba901446 : SUCCESS in 48m 03s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b144a7904db14e52bc7fc7d98b45c38f : SUCCESS in 47m 18s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":5},{"id":"54f79fb444355cc97d5cd205321967491540a3b7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-14 05:45:38.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"629e1db6b9bf61789bb76ecfdd57fb13bf995179","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-14 06:38:31.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e4810650f68a46b893dfea72330b90d1 : SUCCESS in 48m 46s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/44ed723e67c4460dbf113f55e9013a72 : SUCCESS in 51m 33s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/03f3ce7d761c4d18bc26cbb42dea7029 : SUCCESS in 49m 28s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8f45f3b8d506432c8263ae58593f803b : SUCCESS in 49m 00s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"9563c789c958436b61652e5cf2ae0752ec48d98b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-14 06:51:36.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/33002f3e42ba4fa1b88626d3d991e717\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8f495fe9d0d742b8972857d7e4c356b3 : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/293007e319284385a5799ea75db57bb8 : SUCCESS in 2m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/29d46bcc367b403ba362c3c59aafd456 : FAILURE in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eabc3af7ec8c4a22a0787332312eabcf : FAILURE in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56df5381981a4691b0a4faef76a63e1c : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b967637ee7ad45eba63d9fdea71e07fa : SUCCESS in 3m 03s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/64fa534459da4ae2aead0100975d4850 : SUCCESS in 8m 41s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a8d8bbb8dd8d4f6cbbfe77de43bade14 : SUCCESS in 42m 41s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c6f18c48dffa421093e972dd03a44f8b : SUCCESS in 59m 34s","accounts_in_message":[],"_revision_number":6},{"id":"35556386c856507cd9fff4c70a7535cbba9c2af2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-18 09:57:01.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"866ad2011bc4cd75f4804dd03d9178aef95ce1fc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 10:44:34.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d416850c8b7940039d848715bf20e0ed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9a433813eb694b268fa537bf922c08eb : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8b3a8f425fb42e6a1570822f713f147 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a031a3a5191346c0a3b099d14ef9aa63 : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1f39f0d818c14bf09c100a838315f77c : FAILURE in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a5dd43bed8984f49812b1c3c6ccdf96c : SUCCESS in 5m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bbd43d4b0e57421d91b1e9aac41da128 : SUCCESS in 2m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/01121be04c094053a7622ba3888fd582 : SUCCESS in 6m 49s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b52433c7708c4282a8868f59336929d0 : SUCCESS in 29m 25s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c07c65f3d62141f1a55a7510bf9042c8 : SUCCESS in 45m 47s","accounts_in_message":[],"_revision_number":7},{"id":"ed8ccda05db6c1bc0014b80763df45b8b64ef20d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-18 10:50:46.000000000","message":"Patch Set 7:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/df357a4e966645c6b1069df4320526ef : SUCCESS in 48m 16s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ed244aa0e99d40709381219fbb568eb7 : SUCCESS in 50m 37s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/aec0b8afaa6d47f5be970e1b862c6a42 : SUCCESS in 49m 10s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b50a536a5d4b43618f69a34a3404cea7 : SUCCESS in 48m 51s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":7},{"id":"d532c4aeb02f02f44a9a7c9ca03143624ec700fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-18 11:36:06.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"57ffff8145b9cd7329ee7964e53037702f2d8bd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-18 12:10:47.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"91731cf23c6e9765ed02c6b8a5e2c013dcb696b4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-18 12:54:18.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b719b41e222e45f7bb1e58ac8a2a1454\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/57c3e5eb0ec9443394326c6926f19d3f : SUCCESS in 3m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ab07f47b47734eaa8741204f0f578e70 : SUCCESS in 2m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4cec93ec6f5d4fbd960d2f3e16d65313 : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e949080814d47a79eca696d46c6d82d : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/31d79c5b7500456aaf6dd640c6a206aa : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8089e74b56db423da7bd56b641d256e8 : SUCCESS in 3m 13s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/08f0f3efbae141c384655c5c227bcee4 : SUCCESS in 6m 32s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2a49a6720a3f432cacb170ed5073d593 : SUCCESS in 30m 32s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fff81acb884d498cb769a3bd6422cc4f : SUCCESS in 41m 58s","accounts_in_message":[],"_revision_number":9},{"id":"5625c523b251154f381c4babb5b474a882ffc3e2","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-18 13:04:27.000000000","message":"Patch Set 9:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4c8c22e568324f8c84278e4eb5f5ea95 : SUCCESS in 47m 37s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/45ebf80844f94b2eb7f258e03fdc1be7 : SUCCESS in 50m 24s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/62d65dbd56bb4041bfee9789a59e3234 : SUCCESS in 48m 26s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/24064f2c0fb54116b19de248695b4db2 : SUCCESS in 48m 18s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":9},{"id":"3c0a7633bba48a23dac79c867d13bf824c9a7555","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-19 06:45:41.000000000","message":"Patch Set 9: Code-Review-1\n\n(4 comments)","accounts_in_message":[],"_revision_number":9},{"id":"99283aab489b00e1c6e9c8b46ed6c9c3eaeed15b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-19 10:28:38.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"15b59a9a2e00e6db6e2c659b18165f8313b36262","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-19 10:32:18.000000000","message":"Patch Set 10:\n\n(4 comments)","accounts_in_message":[],"_revision_number":10},{"id":"50af85a522485aca50183b8b77632107635acf66","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-19 11:20:17.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ace1407835fa4431b50fbe21f4db59a7 : SUCCESS in 47m 58s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/00222634bcdb4508ac3b485da5bac34d : SUCCESS in 50m 23s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ec970766b891436484a42815ff33a79d : SUCCESS in 48m 54s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b037e86cc86f4c3f8a87d1fdae2a232c : SUCCESS in 48m 40s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":10},{"id":"9b568cb8310b1dad4e72b786ad8408c1cb75ed3a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-19 11:21:40.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/54c4dd38a3094e2ea8e4596bfb707448\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/178a4ce9a3194ef4ad268d3ca796516d : SUCCESS in 3m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d95fe845fd14abf906a24c4b0d2b29b : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d7ce8c4a975240278081fb7c00ee26df : SUCCESS in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f4a94b2cec4947cba0bc104a062116dc : SUCCESS in 2m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a94d841ca5040d98ca41e660310eae0 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/63cec4c74bd14db1b32e63415b6a835b : SUCCESS in 3m 55s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9664885ff9fb4bf6b3fb1550dd1469ed : SUCCESS in 6m 37s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ad17b00f7d2146b0b8bae592c887bce5 : SUCCESS in 34m 38s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9f261626ab50416cba98d0a4bebfff58 : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":10},{"id":"04311f7d54b50193c20e3f5aa5459a9585f56d60","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-20 10:22:56.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"b156d7b4480ce8930cb8c771c1e6b88ca6f76e7c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-20 10:24:33.000000000","message":"Patch Set 11: Code-Review+2","accounts_in_message":[],"_revision_number":11},{"id":"271c7c21c4ee12b7b7b80bbd62669fa3fa15088f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-20 11:10:39.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/80ae0d23eee94fc19c33a44cbef1b245\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc00c32a7f334911804896a964771c66 : SUCCESS in 3m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2a0d15b9042c40c594676101905da7f7 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c3130bc71b2c4dc9b6376dadf469bf0a : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/17f5f54d2bd349d3a354389e68e3a4b5 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0cb4b1706dd04b9c9cac7d5fac7c18ac : SUCCESS in 4m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b060ab359454372a55762e190681b28 : SUCCESS in 2m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/73b1a42dc54343c4ab77374dbbf86552 : SUCCESS in 6m 34s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2d5231a3347043ebbbe527ba5e36eb35 : SUCCESS in 33m 05s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4854c16851424e50bb25f1a23ba796bb : SUCCESS in 44m 17s","accounts_in_message":[],"_revision_number":11},{"id":"01091d5aa7cc1d5da28e2de2a5430642b08b439b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-20 11:14:58.000000000","message":"Patch Set 11:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/485d25feb43346f6b1df9277b52f827c : SUCCESS in 47m 49s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6c185831130c42d38d3268ce62043069 : SUCCESS in 50m 48s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/01f31fc65bfe4bee9179906809619e07 : SUCCESS in 48m 34s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cc271680798b4f29be27e9ef13397612 : SUCCESS in 46m 28s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":11},{"id":"98465454dfd253f93df3efda8d06b4522d43725d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-10-21 10:07:39.000000000","message":"Patch Set 11: Code-Review-1\n\n(7 comments)","accounts_in_message":[],"_revision_number":11},{"id":"15b73ac439a48ecc4c480e9f3aa0277f883b5d3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-21 12:32:22.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"fb2c1a0bc9497d30cf3370e0343fc0861eb8cdd4","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-21 12:33:01.000000000","message":"Patch Set 11:\n\n(8 comments)","accounts_in_message":[],"_revision_number":11},{"id":"0966919dc242cf5a638b8fd0911ee6191df41dee","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-10-21 13:18:54.000000000","message":"Patch Set 12:\n\n(3 comments)","accounts_in_message":[],"_revision_number":12},{"id":"a8e629e13d5728dedac63a63897b8637cc506558","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-21 13:33:35.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fdf62ff5950e40e3aeb398b5fabb0ed9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c14c0c1f17bf4211b06736bd12d0e225 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad00a4e1702047e99aa87d534ceca0df : SUCCESS in 2m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3506fdcf2c04575b0cbcb7b30276d1f : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/595b131a1c464cb49767e39364ca3835 : SUCCESS in 2m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a3858dbf91ad4f6fa0c4dd7415960326 : SUCCESS in 4m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1765bffffa5743cf9af07870a85eebd2 : SUCCESS in 2m 33s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/67d6570e9e524edea2b6432f994799e9 : SUCCESS in 7m 46s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6eee36c023aa4e27893bbdda97dcaff5 : SUCCESS in 43m 36s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f55165506d8f4d5393049bffec2fc5cb : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":12},{"id":"a438e03515216b30365097d1bc2288fb3c33e3db","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-21 14:13:58.000000000","message":"Patch Set 12:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b2ba0222321f490a94faf82e3552e450 : SUCCESS in 49m 01s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/82b8b3890ce84ea1aa6e2e81009b32c9 : FAILURE in 1h 10m 28s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9653cb60d55f48f191bbcf9508f1f49b : SUCCESS in 48m 08s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/299de3458bea4523a8a96f4625167acd : SUCCESS in 46m 10s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":12},{"id":"f80290bbd410fc7ce2a5f1247719e7528e212477","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-24 04:40:05.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"56d74c4806eaa2e4c911f05d8d5befb10ec00a14","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-24 04:43:11.000000000","message":"Patch Set 13: Workflow-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":13},{"id":"4822db4a404a0192cd7ff9cc24100e479c0098b2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 05:26:11.000000000","message":"Patch Set 13: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6c836bde53744a8dae920cd4b97d1dd4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b9b0142f631242c683e56c832e1b318e : FAILURE in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/936583d78e454f479f68cf6c1ef3e1b0 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d4dffb75adce4754906272d9fc67fce1 : FAILURE in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae6a1307012945d3919a3d49253b454e : FAILURE in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/43d4d890c56c4df7b561bc7362505f08 : SUCCESS in 5m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0f6958cd58b4499aa97bccc794d6bdd9 : SUCCESS in 4m 05s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9f030432d81d4a7ea92ffa3ee4f79553 : SUCCESS in 8m 19s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fcd8c6f5885b44aeabfca2754bbe09e5 : SUCCESS in 34m 39s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0f2dc650de444c4ca59e28c41ad84997 : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":13},{"id":"add9adf920c1e25a09adba0bc0ad1bed0bfa65c4","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-24 05:32:17.000000000","message":"Patch Set 13:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cd9ae66c8bf74a7c8d8e21093dd37d50 : SUCCESS in 46m 54s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ea4fc6c7115f48e7ad1db0a41041a43d : SUCCESS in 50m 55s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e839eeffd0324f62ad31b2b7556dd0fa : SUCCESS in 48m 36s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6276d60baba44d6496975dd0094fdac1 : SUCCESS in 48m 20s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":13},{"id":"b37832045c0d3ce72d4be1b788ac04195837b95d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-24 08:28:25.000000000","message":"Patch Set 13: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":13},{"id":"ba5c9926d3b6d7b130ed386cab44fdb2b6e1c240","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-24 10:11:00.000000000","message":"Patch Set 13: -Workflow\n\n(6 comments)","accounts_in_message":[],"_revision_number":13},{"id":"48d8f3a8b93921573869770fd150344d6eb4b803","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-24 10:11:21.000000000","message":"Uploaded patch set 14.","accounts_in_message":[],"_revision_number":14},{"id":"5523b761c4d021f2c41063183b89c576b90a6f20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 11:03:09.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4e080a5c4c9046bb917000df17ee36e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0bc2673a018c49dc80020829ed62a4bd : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/68910e2eb5f64715ac67ca66febdbee0 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5a8ed86664f4ddd95178d2ab9f2b6db : SUCCESS in 3m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ba5209582a264c4f85cb7e78bc6eef02 : SUCCESS in 3m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0fab64da1b5459e99098b5d38af94e7 : SUCCESS in 5m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/feafd2c5ba9c4521817dd1b3d76a90b3 : SUCCESS in 3m 07s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/86d94d7dbae4443281ba6081ef121c12 : SUCCESS in 8m 25s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/60486824ce8f4243927f127521272071 : SUCCESS in 34m 21s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1e95911668ed4c98a9fd0f8078f8f79f : SUCCESS in 47m 34s","accounts_in_message":[],"_revision_number":14},{"id":"84f9cf62e3fcb8af6b8c38d0b56e35820f8f9fb1","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-24 12:18:33.000000000","message":"Patch Set 14:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2369e995ae064812a0053989229b159f : SUCCESS in 49m 14s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d0c4a045b6ca47379e0e2720c40c913d : SUCCESS in 1h 38m 10s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/36e68c56d5ff493f8c917e35b51579c3 : SUCCESS in 48m 36s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9ffec9c2f86c4821aa570b347813fa10 : SUCCESS in 43m 44s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":14},{"id":"28b2111d138c0e1c6a4cfc866c8b41f4a22a3f1c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 05:49:12.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"d47e203e095e127fd968a9597ed4415f1a883513","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 06:38:05.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/411fc8f5fcbb46e297999c439574d627\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/337404ae91164742a3c4df7d1d1411d9 : SUCCESS in 3m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a438d3c233f54e158e39809befab0107 : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c8c5a5598484371aeb38b1d6421fc24 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/248a1caa56134954906495010b15f4b5 : SUCCESS in 2m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b748e9d5893846f2afc58db0ba87febf : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9414102be4d343fdad2baca6bec794e6 : SUCCESS in 2m 44s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/430a5aadb9b14d46acda2173d6e4931a : SUCCESS in 5m 58s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/851590702c724b219d5cea159c34c3a6 : SUCCESS in 30m 34s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bc75b1135c7d47e7b3f28946569bd9d2 : SUCCESS in 47m 36s","accounts_in_message":[],"_revision_number":15},{"id":"8d83c1d3831ad9ed62847bf5a299e49764be8169","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-26 07:13:52.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a89ff13c90a844ca9f9bc640c4aef8e2 : SUCCESS in 48m 19s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a0080d7ad26f4ea2b46afa94a1befac6 : FAILURE in 1h 23m 13s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3d3875c5f6ee4abb84493c7591e43efa : SUCCESS in 1h 10m 15s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ca0d9582de1d42118b106b96021953b8 : FAILURE in 53m 03s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":15},{"id":"db96992d8140c171ca33a2a68e1fcec764b948c3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-10-26 11:10:15.000000000","message":"Patch Set 15: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":15},{"id":"b50ed1915e7878244284dd8520ffa9b0d098f4a6","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 12:52:13.000000000","message":"Patch Set 15:\n\n(7 comments)","accounts_in_message":[],"_revision_number":15},{"id":"82a924dae651e096d9c877329e7649ff8881fdb5","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-10-27 10:53:50.000000000","message":"Patch Set 15: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":15},{"id":"e2194f2b96d61cf855d47a47a77d4393c1af5476","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 10:54:28.000000000","message":"Patch Set 15: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":15},{"id":"4973456c1a27edfd15dcb69930e15bb8c4a50757","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 11:58:46.000000000","message":"Patch Set 15: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/18effcfb53f84e929d2eaed2d504318a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc6767565fbc4856ba300e26daf2a6af : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5ecbffa0bafe4909aa7aafca2e141df5 : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97b5f2fd8b254229a896e92496d9f939 : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e44d1316801543779d8ddd00d6754241 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19672dd814e2432f897789295d43c21b : SUCCESS in 4m 35s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c6d739a26f98406a9f1a0b08dcf92085 : SUCCESS in 8m 28s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b7c49f4b324548ac8b389c91962bb54a : SUCCESS in 32m 49s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b4385d8c75ed4ba8ab2efb5815565ae0 : SUCCESS in 1h 03m 01s","accounts_in_message":[],"_revision_number":15},{"id":"be81e8000620ad5990a9e2e77845833052d7d21e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 11:58:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":15},{"id":"8df3689d6b6d011d0cda25cdefae64dc608edc03","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-27 11:59:47.000000000","message":"Patch Set 15:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eb22ae8331fe40d2908eb28114e1054d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f5077753e23f4595a00c19ab8d748792 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c58dfc1b2f124105aa23fcedf17a394f : SUCCESS in 37s","accounts_in_message":[],"_revision_number":15},{"id":"8f17bac00aaacc86453e4586a2e368c6b4c25f77","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-27 23:26:49.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"cdfca5caeccd5af92c2a9feac01d11757bd0162d","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2025-10-02 04:30:23.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"b6099a13e9c9be622bf93e3fc96a5804a0e8c445","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-10-02 04:35:35.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"d4afd2f80ff31ee306e6588005da4dccddce8f25","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2025-10-02 04:54:08.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"906ab931610d6d4d6bbaab8eb99cfbdee8a40c0f","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2025-10-02 09:22:28.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"af23df9dca9be85b1d92c3b30447d882f4104c44","author":{"_account_id":4257,"name":"Zane Bitter","email":"zbitter@redhat.com","username":"zaneb"},"date":"2025-10-03 04:46:14.000000000","message":"Patch Set 15:\n\n(1 comment)","accounts_in_message":[],"_revision_number":15},{"id":"ec95b3381078292901f4b10564cd26bb4e094e8b","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2026-10-05 01:34:36.000000000","message":"Patch Set 15:\n\n(2 comments)","accounts_in_message":[],"_revision_number":15}],"current_revision_number":15,"current_revision":"3e69fce709765e317cfd0288b7c119e757f94117","revisions":{"2f075218c08fc963f8915633ab07c1e8e8afb507":{"kind":"REWORK","_number":1,"created":"2022-09-06 13:57:50.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/1"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"subject":"[WIP] Improve resiliency of eTag handling","message":"[WIP] Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2f075218c08fc963f8915633ab07c1e8e8afb507"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2f075218c08fc963f8915633ab07c1e8e8afb507"}]},"branch":"refs/heads/master"},"4e9e40f9f7a9c8bbb83dddd000c9410177e43bdb":{"kind":"REWORK","_number":2,"created":"2022-09-09 04:02:33.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/2"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:02:15.000000000","tz":600},"subject":"[WIP] Improve resiliency of eTag handling","message":"[WIP] Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4e9e40f9f7a9c8bbb83dddd000c9410177e43bdb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4e9e40f9f7a9c8bbb83dddd000c9410177e43bdb"}]},"branch":"refs/heads/master"},"073a9ba809bc0108d335eb65ce9fa3a9cd6eec43":{"kind":"REWORK","_number":3,"created":"2022-09-12 21:52:56.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/3"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-12 21:52:43.000000000","tz":600},"subject":"[WIP] Improve resiliency of eTag handling","message":"[WIP] Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/073a9ba809bc0108d335eb65ce9fa3a9cd6eec43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/073a9ba809bc0108d335eb65ce9fa3a9cd6eec43"}]},"branch":"refs/heads/master"},"2e7912473a22530a1728ba1a3824ea6037b6151b":{"kind":"REWORK","_number":4,"created":"2022-09-13 11:36:05.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/4"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-13 11:35:46.000000000","tz":600},"subject":"[WIP] Improve resiliency of eTag handling","message":"[WIP] Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2e7912473a22530a1728ba1a3824ea6037b6151b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2e7912473a22530a1728ba1a3824ea6037b6151b"}]},"branch":"refs/heads/master"},"838791dc58517c0aab0a916887bed07d412a5ab7":{"kind":"REWORK","_number":5,"created":"2022-09-14 00:00:15.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/5"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-13 23:59:58.000000000","tz":600},"subject":"[WIP] Improve resiliency of eTag handling","message":"[WIP] Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/838791dc58517c0aab0a916887bed07d412a5ab7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/838791dc58517c0aab0a916887bed07d412a5ab7"}]},"branch":"refs/heads/master"},"955b12bc4adbfedc4b439101c0343bf302b85f37":{"kind":"REWORK","_number":6,"created":"2022-10-14 05:45:38.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/6"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-14 05:44:31.000000000","tz":600},"subject":"[WIP] Improve resiliency of eTag handling","message":"[WIP] Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/955b12bc4adbfedc4b439101c0343bf302b85f37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/955b12bc4adbfedc4b439101c0343bf302b85f37"}]},"branch":"refs/heads/master"},"3d2f2516ebb49d1531296a0aabd96f975ee1f881":{"kind":"REWORK","_number":7,"created":"2022-10-18 09:57:01.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/7"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-18 09:56:44.000000000","tz":600},"subject":"[WIP] Improve resiliency of eTag handling","message":"[WIP] Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d2f2516ebb49d1531296a0aabd96f975ee1f881"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3d2f2516ebb49d1531296a0aabd96f975ee1f881"}]},"branch":"refs/heads/master"},"76697645db6a6c4b2651c73199e7bf7535ae49c2":{"kind":"REWORK","_number":8,"created":"2022-10-18 11:36:06.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/8"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-18 11:35:46.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/76697645db6a6c4b2651c73199e7bf7535ae49c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/76697645db6a6c4b2651c73199e7bf7535ae49c2"}]},"branch":"refs/heads/master"},"53fb87c9221b996e9b509a85814f17afe260f94d":{"kind":"REWORK","_number":9,"created":"2022-10-18 12:10:47.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/9"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-18 12:10:34.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/53fb87c9221b996e9b509a85814f17afe260f94d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/53fb87c9221b996e9b509a85814f17afe260f94d"}]},"branch":"refs/heads/master"},"ec0f7b1553ca0dc022c37f5a0cb2608467d96738":{"kind":"REWORK","_number":10,"created":"2022-10-19 10:28:38.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/10"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-19 10:28:21.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ec0f7b1553ca0dc022c37f5a0cb2608467d96738"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ec0f7b1553ca0dc022c37f5a0cb2608467d96738"}]},"branch":"refs/heads/master"},"fdb5d9e018bb9d2ae41f46ca2bd8fc245f152793":{"kind":"REWORK","_number":11,"created":"2022-10-20 10:22:56.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/11"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-20 10:22:29.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fdb5d9e018bb9d2ae41f46ca2bd8fc245f152793"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fdb5d9e018bb9d2ae41f46ca2bd8fc245f152793"}]},"branch":"refs/heads/master"},"4cfc795c7aebe42af5d8d0ce78d26d66df04018f":{"kind":"REWORK","_number":12,"created":"2022-10-21 12:32:22.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/12"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-21 12:32:08.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4cfc795c7aebe42af5d8d0ce78d26d66df04018f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4cfc795c7aebe42af5d8d0ce78d26d66df04018f"}]},"branch":"refs/heads/master"},"dbc445a17d425f0a336aee7f4dd3ee9cb49fb613":{"kind":"REWORK","_number":13,"created":"2022-10-24 04:40:05.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/13"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-24 04:39:51.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dbc445a17d425f0a336aee7f4dd3ee9cb49fb613"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dbc445a17d425f0a336aee7f4dd3ee9cb49fb613"}]},"branch":"refs/heads/master"},"c423964ec180a9a0fb387b37061bc9ba964d6329":{"kind":"REWORK","_number":14,"created":"2022-10-24 10:11:21.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/14"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-24 10:11:07.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c423964ec180a9a0fb387b37061bc9ba964d6329"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c423964ec180a9a0fb387b37061bc9ba964d6329"}]},"branch":"refs/heads/master"},"3e69fce709765e317cfd0288b7c119e757f94117":{"kind":"REWORK","_number":15,"created":"2022-10-26 05:49:12.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/23/856123/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/23/856123/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/23/856123/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/23/856123/15"}}},"commit":{"parents":[{"commit":"b619196fdbe374962b2fa0e0958827e247484148","subject":"Fix misuse of assertIsNone","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b619196fdbe374962b2fa0e0958827e247484148"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-26 05:48:58.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3e69fce709765e317cfd0288b7c119e757f94117"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3e69fce709765e317cfd0288b7c119e757f94117"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
