)]}'
{"id":"openstack%2Fsushy~856597","triplet_id":"openstack%2Fsushy~master~If25e30bfb9bf06fdbba6051177947d4d285ae2ab","project":"openstack/sushy","branch":"master","topic":"pending-systemurl-fix","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2022-11-22 09:20:37.000000000","reason":"Change was submitted"},"27909":{"account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"last_update":"2022-11-22 07:45:06.000000000","reason":"\u003cGERRIT_ACCOUNT_27909\u003e replied on the change","reason_account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2022-11-22 09:20:37.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"If25e30bfb9bf06fdbba6051177947d4d285ae2ab","subject":"Fix setting boot related attributes","status":"MERGED","created":"2022-09-09 04:50:55.000000000","updated":"2022-11-22 09:21:49.000000000","submitted":"2022-11-22 09:20:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":44,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"856597-pending-systemurl-fix","meta_rev_id":"c57b80c423968c6a0e0e04036c7630bbc6c271b6","_number":856597,"virtual_id_number":856597,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-11-22 09:20:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":2,"date":"2022-11-22 07:45:06.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":2,"date":"2022-11-22 08:18:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":1,"date":"2022-11-22 08:18:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-09-09 05:38:59.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2022-09-09 05:46:22.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-09-09 05:57:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-09-30 14:05:08.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"CC"},{"updated":"2022-10-26 11:13:40.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2022-11-04 09:04:02.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2022-11-21 18:41:14.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2022-11-22 08:18:01.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"b204c9c502f5446c16725f0f2136e26e9abe2c98","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-09 04:50:55.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"53f77791947f8c22f12c9d1fbb0951c4a42f32bb","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-09 05:38:59.000000000","message":"Patch Set 1:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/856597/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/a8d9236/a8d9236.html : SUCCESS in 46m 00s","accounts_in_message":[],"_revision_number":1},{"id":"04fa24210a91a0b8a874ebc8902807e7f951e470","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-09 05:46:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2900f420b20e4973be824f298b38ff8a : SUCCESS in 50m 28s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ea80c841d3364736879996ef6d6a57fe : SUCCESS in 53m 46s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1b77d172557b40c8901dc75d7afc2014 : SUCCESS in 51m 17s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/dbbb289479fa4dbc844abe098418546b : SUCCESS in 50m 58s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"0f260da6989e3d36abcba36b139f578d8e3684f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 05:57:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e681eabd5ad84021a6f344b3970435f0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45119f295d1741d283a2ad06209ce617 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/96736ce8c793488ebbac6a9579c7c5b4 : SUCCESS in 3m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/44e2ff48d1e04765836d7b0caf48a8d1 : SUCCESS in 3m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4a80d5ed48a54f1381b1f12c9720ceb7 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65d784932f834587a243c4fc86f78787 : SUCCESS in 5m 30s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/cc906bb9189c456cbafe7e7cebe5918a : SUCCESS in 11m 42s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/28c96227d26a46ef9bbedc963f0eff4a : SUCCESS in 51m 31s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/f45b0e181f804f23baac988c5214b3f2 : SUCCESS in 1h 01m 24s","accounts_in_message":[],"_revision_number":1},{"id":"9554a2b5444285ecb994f14f5928179085de162e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-09 11:46:28.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"8a81669050cd77ad309a053d8ddd38589511de54","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-09 11:47:50.000000000","message":"Patch Set 2: Workflow-1","accounts_in_message":[],"_revision_number":2},{"id":"7a401c2ed4477fca962349871434d9669ecbe9df","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-09 12:28:37.000000000","message":"Patch Set 2:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/856597/2/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/7b612b2/7b612b2.html : SUCCESS in 40m 28s","accounts_in_message":[],"_revision_number":2},{"id":"9d59831578dceead7baecf527b587fe6e216fe57","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-09 12:40:01.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da8ceddd2ab54e3b8f98edcacdadd7fb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24e6e26321c247de86ce42b15dae2d15 : SUCCESS in 3m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ef31ea221d7f4615bfbacb832dc1dd3f : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a835c87902104feb9058c3a827751f1b : SUCCESS in 4m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/834aeb49764940d9af6277c9f16e2134 : SUCCESS in 4m 10s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d0097cded9b45f68add3ee63afe2910 : SUCCESS in 6m 27s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/6b80103da0384f8bb62f2c2051f8987e : SUCCESS in 8m 58s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3a1224666e99447c9a0f9db54365d614 : SUCCESS in 41m 20s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0f15d4a626a14556b2943afc98be1bed : SUCCESS in 51m 20s","accounts_in_message":[],"_revision_number":2},{"id":"61e2f77d3a12c417b043019095f7a3b7691df1fe","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-09 12:45:01.000000000","message":"Patch Set 2:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cfdb59fa46dc49fab977838a431393b2 : SUCCESS in 48m 43s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3e9b79dd2a61457096ca17733713df25 : SUCCESS in 57m 13s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a9c4cf65762f45f487cf7ab262977d00 : SUCCESS in 49m 42s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/fe047503e6484396b01cd76ab9514c12 : SUCCESS in 49m 01s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":2},{"id":"d64110db90959380fb9fb91d59b8dfd32ee68cbd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-22 01:11:05.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"5fb8a9c6ceab0d724b4f2289622b79987058ebf9","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-22 01:56:12.000000000","message":"Patch Set 3:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/856597/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/b2a1e13/b2a1e13.html : SUCCESS in 42m 26s","accounts_in_message":[],"_revision_number":3},{"id":"ee6cc59f9cc36824fc46466c8b008d779e32cd5f","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-22 02:09:47.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/b63d2da9592544efb267d3a207e84f8e : SUCCESS in 48m 39s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/40e9c67be19e4de1b71bc5cd0f605ca2 : FAILURE in 55m 32s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/222a2e0fe5e540aa91fc79b4003acb3c : SUCCESS in 50m 08s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f421d8ee69144de8a8649063b0578169 : SUCCESS in 49m 37s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"d2bb286b3ae42e806ad955f3431ee5c21bbfaaf7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 02:17:52.000000000","message":"Patch Set 3: Verified-1\n\n(7 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2826a4ca178940b392079bf83b3f5d0b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f0a496f515c64e7097e4d346acf9bed8 : FAILURE in 2m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ad74b10e4c480f9f31b6b75c24783c : FAILURE in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7a3d35bb8e784def822a3aa7ffbdca73 : FAILURE in 2m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f3e13866dd884ad89aa59b58a4748cf2 : FAILURE in 3m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbd2cf8f475347429a90c8e375021e27 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6953e095f95b44f786186e0ffb4bbd4f : SUCCESS in 2m 52s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f19e1259c98f47f097c4f91fb7049223 : SUCCESS in 7m 28s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d45a10b61572442f9e022c1749c50122 : SUCCESS in 45m 09s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3adefa0de4424af0a3c70202bedea9c5 : SUCCESS in 1h 00m 55s","accounts_in_message":[],"_revision_number":3},{"id":"1a7278353cfa15aa2364179d4f61e9e551cf0404","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-22 11:19:18.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"af7d816cacff27eb370f74385b6841f2dda448ca","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-22 11:28:07.000000000","message":"Patch Set 4:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/856597/4/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/be4a4d2/be4a4d2.html : FAILURE in 6m 37s","accounts_in_message":[],"_revision_number":4},{"id":"b2d73a69040b46838b20dfe255b2a610532ba5ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-22 12:11:56.000000000","message":"Patch Set 4: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2985302ac7f04c8ba12ed64e94b3d00e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fff8373d96604ec0a7b493344c6953bc : FAILURE in 3m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1511912d068f41f4853857f1a3ddbc6e : FAILURE in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7ab5c417262b4f3fb03b7061eea96328 : FAILURE in 3m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e06b7883f504ef881674a853a8387b2 : FAILURE in 3m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4de010fc88794f54b99280f82c5e5ca1 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1f61100678cf49ada2ea023847efa463 : SUCCESS in 4m 10s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/88a789e9f3f54aa8b59f93961a435067 : SUCCESS in 7m 42s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/0e0262c7896c4a42a77c41f4b72bc98a : SUCCESS in 34m 45s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d99ab1af47d1426382560cca8f563c17 : SUCCESS in 51m 12s","accounts_in_message":[],"_revision_number":4},{"id":"b19da94f3c2967a030dc1ae8854e8626c333e50b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-22 12:17:24.000000000","message":"Patch Set 4:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c3428605d05940389c7b1bd7513ef63c : SUCCESS in 50m 30s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3c256d89a6654d109db67024b64bafef : SUCCESS in 54m 30s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/be8e00fefc42429d8dbab86c19587604 : SUCCESS in 51m 37s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e16e4157a6aa4a7590fba528babec0c5 : SUCCESS in 51m 25s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"3417c282ecb803c0115ccce09e5e1297bffc18ca","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-27 12:40:31.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"1b39497650526a03eadb5f973daef6286fb44afc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-27 12:52:15.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"96e33ab73d2705350bd0f68d44d1a49920efab77","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-27 13:37:14.000000000","message":"Patch Set 6:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/856597/6/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/4d75f12/4d75f12.html : SUCCESS in 43m 09s","accounts_in_message":[],"_revision_number":6},{"id":"a9f0fcd82441673e268648184610d75e952ebebf","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-27 13:46:42.000000000","message":"Patch Set 6:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9da3348a7bb749458b9bcc8e984dd02e : SUCCESS in 50m 02s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/abc2b1f1f9c74cd9b174dc11e22b4795 : SUCCESS in 53m 13s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cf384d383b89400c9652ba07f42224d2 : SUCCESS in 51m 34s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/1bd73661e0f74325b2efa8dedf3aebd5 : SUCCESS in 51m 17s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":6},{"id":"b65745d661d2bdf1d90d318edc5382cdd5113181","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 14:01:54.000000000","message":"Patch Set 6: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e1e33739edd0465c81fc58f1ef67c7b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f58b6eaabf943bcab4c83c3741f1fd9 : FAILURE in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da6b1ca6e92c44718899a8d4a9fea106 : FAILURE in 3m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/069d8b1fe39645448554d32f65480c1e : SUCCESS in 3m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1249467ef2744b70be091d41c815abff : SUCCESS in 4m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad5daccf2ac042b8899154ea490bbf7b : SUCCESS in 6m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94fcc62109d349028dbf8a0483d7dce9 : SUCCESS in 3m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3803597b19c7413fbb37a25c7aad5469 : SUCCESS in 9m 31s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5c0be30801e1415580a1122d648282f2 : SUCCESS in 39m 48s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/28fd088999844747b4239f42731611a0 : SUCCESS in 1h 02m 30s","accounts_in_message":[],"_revision_number":6},{"id":"f5500ceaa00006ee4e9a3abd0f5a3d0b25accde9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-27 20:35:49.000000000","message":"Uploaded patch set 7.","accounts_in_message":[],"_revision_number":7},{"id":"e5c52f88e65b83df0532294e7f1a3f48cf81f03d","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-27 20:48:18.000000000","message":"Patch Set 7:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/856597/7/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/d3b51d2/d3b51d2.html : FAILURE in 10m 50s","accounts_in_message":[],"_revision_number":7},{"id":"5f310823c54032af119b14a16d23acf6f8278b20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-27 21:36:18.000000000","message":"Patch Set 7: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/649626ad832b4ba48ca6228ca113f608\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5812a74c626f49b086a6ce3ddf5c77c2 : FAILURE in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e4cc5fd5dd84a29948a5d0bad6063e1 : FAILURE in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4dc55cb33e3e40dda77b4fade8b8c6e9 : SUCCESS in 2m 51s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9662c8108238450eb837bb614763bf71 : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eab17076431c46eab09b75ce807b73ba : SUCCESS in 4m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ab7a16f91764f4c88a8fdffb0cdc0ac : SUCCESS in 2m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3ed4799da5d64714bc56c2cd385d3d0f : SUCCESS in 7m 21s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/645779be94af491d99592b0f9abf1939 : SUCCESS in 45m 52s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4e1ec77392944b498306af7777e3c270 : SUCCESS in 55m 09s","accounts_in_message":[],"_revision_number":7},{"id":"d073a380d002c067ca0c04addc3d53ef0d013d9d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-27 22:14:21.000000000","message":"Patch Set 7:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/28a9c07a16714fa59dc3847f9ce17944 : FAILURE in 39m 23s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3d81aca9043b494194bb46153eb0ca8b : SUCCESS in 1h 37m 04s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5b8b4597e67e4a408ed4ef122f913859 : SUCCESS in 48m 22s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9e7377a84f6c40488fe6dcc406f72443 : SUCCESS in 47m 54s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":7},{"id":"2f990545f61aba389510bc70b36165372c4f75a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-28 01:31:11.000000000","message":"Uploaded patch set 8.","accounts_in_message":[],"_revision_number":8},{"id":"5bfbc03cc9d81f304d71d6a7ab5a9878b5d393a6","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-28 01:35:05.000000000","message":"Patch Set 8:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 finger://HPE-CI/57af20a2ceaf4b97a47a89e8075525f0 : POST_FAILURE in 23s","accounts_in_message":[],"_revision_number":8},{"id":"bda526886c284fbf23b708dd21b6878d802a4fb9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 02:24:47.000000000","message":"Patch Set 8: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/dcc0577d36044ee196199a8cbacec2c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a1b83e851c6458d8031e3d1a6352ae1 : FAILURE in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9edff59755744a4bad460a59455f9b29 : FAILURE in 5m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c7f1ab2b9ee549d591f096589b4ce608 : SUCCESS in 3m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4e1dc006cc144446999086b79ade4584 : SUCCESS in 6m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d3bbe728670d432da605bd1294965825 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/52d32357cdb84b1bbfe94ca88d48e1cd : SUCCESS in 5m 53s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/40554c95990b43c2a35f374b05c23fb4 : SUCCESS in 11m 04s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a293bffd1f344a2cb2d3c23f31aa3ca9 : SUCCESS in 46m 02s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1a327f7354a142fb871af5d2a494b4c6 : SUCCESS in 51m 41s","accounts_in_message":[],"_revision_number":8},{"id":"e8742877c3d6c19c62263d5fac84f574e3c704bc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-09-28 04:57:17.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"5e8f4fde2710de36acdcfa67d90e63b251fcb198","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2022-09-28 05:56:57.000000000","message":"Patch Set 9:\n\nBuild succeeded.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/97/856597/9/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/3de2fa6/3de2fa6.html : SUCCESS in 49m 05s","accounts_in_message":[],"_revision_number":9},{"id":"0a6af1d2cd99f64a3f0a46156d9a92bb3e493dad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-09-28 06:06:01.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/acbbe82f211842ed97164caa1169d7d1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1f0c0900daf144fa9d1641f2b0e6cefc : SUCCESS in 3m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/26f06f3a5fe645c4a5ba4338deade65e : SUCCESS in 3m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/418886e18e944588aa371683f33fecb3 : SUCCESS in 3m 36s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bd3015b8dada4872b11baa8d405262fc : SUCCESS in 3m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d0fdbf29e079424787ab80cd5c7d1bec : SUCCESS in 6m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17a177e5114e46eda05da90c18b75bcc : SUCCESS in 3m 43s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c3a70e4d576d4e86949899b63ee11970 : SUCCESS in 7m 38s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7b7125b9c228463ebccccdce420a2a97 : SUCCESS in 35m 06s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8da07005ad364f6f937251ca82ce6eec : SUCCESS in 1h 03m 38s","accounts_in_message":[],"_revision_number":9},{"id":"3ef4642b8cbae2916549ece91df406b1fa8471f6","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-09-28 11:09:29.000000000","message":"Patch Set 9:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6d39d3137e1b486086c8b013201623e1 : FAILURE in 37m 35s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/fe6a925fcbe64adea2a0133248200d1b : FAILURE in 43m 54s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/13a24f266f244fa4b3d70ceb85c1bbb4 : SUCCESS in 48m 38s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/7a39f3bfbb534794ac000c9c0e6b3f6e : SUCCESS in 48m 16s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":9},{"id":"8fed241b28280cd93b65023f8d00383a609f94c6","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-09-30 14:05:08.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5af3567130d27f6a0feb3e7092f77aec80605f7a","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-03 23:35:16.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"b0a53c8740f5fce626f27d7b89f35a61942ac9dc","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-06 11:30:56.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"9c61721f37fea743517e3d5ebaf025d1cb0b13f7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-24 12:57:53.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"c4430bd87374ad53097161245c19d1395b9bdf6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-24 13:42:31.000000000","message":"Patch Set 10: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/a21b09fdb6684b58a02ff29f33f79eb1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2970a8f840174beab627913bf39edc77 : FAILURE in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0555e1cabcdb45b18260dfded46e4b82 : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d303b73333e24d69ac77af4c90b8b545 : FAILURE in 3m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ae0b888f03a14ffbad092869a895f081 : FAILURE in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7216e4c9893a40e0bdf983c2fb44a69d : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/37b7e9416aa643f58b0ee32d5dfc59e2 : SUCCESS in 2m 55s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1bb8392b43234ca18fede02432c2cc16 : SUCCESS in 7m 54s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/24d1171b77e041528af49b406dd04981 : SUCCESS in 34m 32s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/78b6fbe7c4db4979961247b983fd3991 : SUCCESS in 43m 10s","accounts_in_message":[],"_revision_number":10},{"id":"066e0f03beb7e7b4afed2ae09480332fd1399454","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-24 13:54:46.000000000","message":"Patch Set 10:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/31fe743745db479b811757e0179481a5 : SUCCESS in 47m 57s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a1873bd3fab84340b487c002e92813d2 : SUCCESS in 55m 26s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ec34674792ce46bcb62dfbb6f130eb0a : SUCCESS in 48m 57s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/bea5048ac3934027801211c0f664d833 : SUCCESS in 48m 18s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":10},{"id":"6be1d5c50ae4ebab6e7fdf2fc32868e8db757391","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-25 13:24:35.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"386bb49cad9d7010f52131d3f90385be433ad129","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-25 14:11:44.000000000","message":"Patch Set 11: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5cd67ff4e41348a4a64b529e5e342b97\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f38495c476d040f2ae6fc100f5cc6393 : FAILURE in 3m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3384155ed0d34a1a83017f2462d54699 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/eba91cab0b714090a58a13393402d84e : FAILURE in 3m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/330aa312f5904339a7d6421a5f1ef677 : FAILURE in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e10ebfb857d2472797d57deffd996209 : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/995b9e74bb244d4893d54b942de7c86c : SUCCESS in 3m 01s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8e1fab8104394985a514ce2d823e4748 : SUCCESS in 8m 47s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b4e28251b71e400f8e4f9fbe2aaa6421 : SUCCESS in 33m 44s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a64efa2e6b9240c8a8ea3443d43aa345 : SUCCESS in 45m 17s","accounts_in_message":[],"_revision_number":11},{"id":"2233764f8dc3afadc82a02407abe55995c7e0a9b","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-25 14:21:57.000000000","message":"Patch Set 11:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cdfdb16cb57d44838fc0b9c8c1f18971 : FAILURE in 55m 29s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9041004b7c4a40218189d32c4d085ee4 : SUCCESS in 51m 36s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/67f65c2ac249433abb941c0e558cf559 : SUCCESS in 47m 49s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8089ff8dd00548d38e260d935d6c6b8a : FAILURE in 56m 01s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":11},{"id":"b7599c92b5bd01e7e68884fabcdf59b223fb4013","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 05:14:26.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"840d30a03a48baef502307ca48a120128ca4b701","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 05:26:12.000000000","message":"Uploaded patch set 13.","accounts_in_message":[],"_revision_number":13},{"id":"c97032cec6157af5667e8fa64cb875debd3e2d96","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 05:27:49.000000000","message":"Uploaded patch set 14: Commit message was updated.","accounts_in_message":[],"_revision_number":14},{"id":"590e3088a6b02fbe57348835ffe640923f5f5e00","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 05:37:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"aa3d988805d8ae680be877b715effd8271c9305c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 06:15:13.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c307112191744e628c1d00119d65d5b0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/dc09fad5e3ca43e88943bcd4fac70ab8 : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/142d05380f3248b8bbd97296fd1101ea : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d2eb46143b0e4669abda39dff5ee9d01 : SUCCESS in 3m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/929e875f04044ec0b6e6d3f89a4e28a1 : SUCCESS in 2m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e63e13c1393b4ad292b829c9abc0813f : SUCCESS in 5m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6ea7fb17ffe044b3a218eb833d745cf3 : SUCCESS in 2m 40s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/940ba89757ef481e86ecd2345fb3f80e : SUCCESS in 8m 01s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/00ff9f9d7c664680b7624d8dfa92b930 : SUCCESS in 29m 31s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b1e173fb896a4521a871732806aaa3b7 : SUCCESS in 45m 59s","accounts_in_message":[],"_revision_number":14},{"id":"44035f36554a667d0441bcac8a70b76cbc778d81","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-26 06:25:24.000000000","message":"Patch Set 14:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cf97a23fefcf48cab182ae6db7c25821 : SUCCESS in 49m 17s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c971a390ef0049e7a21f187b0f9a5b4c : FAILURE in 56m 17s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/019b90b720fb4807a6c505e105d05cf4 : SUCCESS in 48m 47s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/39f8877c24ae44589bec7d5a027bf755 : SUCCESS in 49m 31s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":14},{"id":"9a9a3e8c6a796c02264091c94daaa6faed12421f","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-10-26 11:13:40.000000000","message":"Patch Set 14: Code-Review-1\n\n(1 comment)\n\nOne minor but potentially important issue inline","accounts_in_message":[],"_revision_number":14},{"id":"48b67c78b3ede407817d2ab7f2181c388a2761a6","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-10-26 12:53:14.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"320f10a53844a9b6324e2152a3a25cffc33c1a3b","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 13:04:38.000000000","message":"Patch Set 14:\n\n(1 comment)","accounts_in_message":[],"_revision_number":14},{"id":"16dd65df2f0de04245b53f4410be3b509b312815","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 13:08:28.000000000","message":"Uploaded patch set 15.","accounts_in_message":[],"_revision_number":15},{"id":"fa0064a341ed0ad3f40c265fed0d07d18332a673","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-26 13:08:53.000000000","message":"Patch Set 14:\n\n(2 comments)","accounts_in_message":[],"_revision_number":14},{"id":"d073bc4791fb5e78168f543d629e91623c7ee789","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-26 14:00:58.000000000","message":"Patch Set 15: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d3f48c44741f4a89840fe0ecb63e0ff4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/804b3d33ec784002b44327ecb700863a : SUCCESS in 2m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2ee8e958ac87497da22aa0545c29d27e : SUCCESS in 2m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/988d4e76fc034d9c9d7d1ec9f73035cd : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84665fd716164e7d8a25ef6441a7631c : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aabf720874dc4c578adb28194d000315 : SUCCESS in 6m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeb57940f0164e0f8aa52fa1969a1214 : SUCCESS in 3m 01s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c50881725ef043e8ba30d8a02c4d7b19 : SUCCESS in 7m 46s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9237e59a2c6540cb8979d1053bbabdef : SUCCESS in 31m 24s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/460746b2aad5453b95e11d43c9eb0553 : SUCCESS in 48m 18s","accounts_in_message":[],"_revision_number":15},{"id":"e63a6c9f79df6cc2b0a81db31889b7862d37e837","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-26 14:05:43.000000000","message":"Patch Set 15:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/90cb3291c93941d79f525cc37162c1a1 : FAILURE in 54m 51s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/159aea958ee743fdae51a30cf338c229 : FAILURE in 54m 58s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6f4941ffa8a14a6fa9d571c59820ea13 : SUCCESS in 48m 22s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/5742e1ff7faf4003abe24bd8b447e0fd : FAILURE in 54m 48s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":15},{"id":"e29048f35ca56fcdc3ed45bedb5d0c6344bcbe26","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-31 07:38:36.000000000","message":"Uploaded patch set 16.","accounts_in_message":[],"_revision_number":16},{"id":"bdbbc26906fc785ddd2bebbe73ce1fde0bc48ac0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 08:09:36.000000000","message":"Patch Set 16: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9acfceb934a413fbd767b8d62704fab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/30d4b6e8fc234b52a20c6774f464dfe3 : FAILURE in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/161697a4cf214957beeef817454939cf : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbd3e6d8e39d480cb035db73c506ccd6 : FAILURE in 2m 54s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4f4735fa827345d9b1121bc4b6c1a3b9 : FAILURE in 2m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71e2e7cb758c40c0b18b4c4312426ece : SUCCESS in 5m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/503be2c98346421689ed8c6ebd0772cb : SUCCESS in 2m 38s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/fcd8e66de5d642af8738813a394119ba : SUCCESS in 11m 47s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8be6127ed6714854a7a524a5b65cce48 : FAILURE in 22m 39s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c350c4aacd974660a337da2721cc4ec9 : FAILURE in 25m 50s","accounts_in_message":[],"_revision_number":16},{"id":"9fe6386ded32b89078b615c9ad8c727cbabb1b6d","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-31 08:23:15.000000000","message":"Patch Set 16:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/46ac3d6c83a6421ab78d517c43c7f7da : FAILURE in 40m 46s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a067c317995e41bcb7f441eef5ed90aa : FAILURE in 40m 43s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/419b45ec72de4384a86e20198f2036e6 : FAILURE in 28m 49s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/8cbcefe5197643f19a1d31cfffb76541 : FAILURE in 39m 41s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":16},{"id":"b7eba00ffe023effb1cf17ea29fdd83730be32dc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-31 11:22:21.000000000","message":"Uploaded patch set 17.","accounts_in_message":[],"_revision_number":17},{"id":"fc6cdbb83adbde5d0a4eeaea64eaf890c89e8616","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-31 12:11:32.000000000","message":"Patch Set 17: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":17},{"id":"9fc85e90f0ed49e270be0e54a9b5ccfcf1de33f5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-31 12:26:32.000000000","message":"Patch Set 17: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/efd13ef43f3444859658bd14ba5e1eb7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f07866027ec54a7696c717acfc341024 : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d42b30dc1fe44009ac8ac7f735a2e341 : FAILURE in 2m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8b1a569af3d64430bf18b0fad746eeb4 : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a7023d4867df4cb0ae6a069044f8ed5a : SUCCESS in 2m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/982232038c404a7b854edfbe9ad6aa27 : SUCCESS in 4m 49s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/35fb2cd40abc423caa512d50ca960c11 : SUCCESS in 2m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7faa6ede02f040768bc863f70e6e7dce : SUCCESS in 8m 09s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/72a5c06d35274ad6b85556a56519ae3a : SUCCESS in 45m 15s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/95133c6c345e4b7e961a9058060947b3 : SUCCESS in 56m 59s","accounts_in_message":[],"_revision_number":17},{"id":"ce052d787a6640a7af10861af08cefdc435542d7","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-31 12:54:01.000000000","message":"Patch Set 17:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e742736a5cfd4d429f03b7e6c796e829 : FAILURE in 38m 25s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d548f35d2ee9481585ea95dd7d35a7a0 : FAILURE in 1h 24m 23s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e7f646a799e649fdab82b7eb5aea27ae : SUCCESS in 50m 16s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c4498f61fae141cc855881929fc8f9fc : FAILURE in 56m 27s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":17},{"id":"fa465d7d4b24a086a8cad316952d48e3a4ac3a1e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-31 23:25:25.000000000","message":"Uploaded patch set 18.","accounts_in_message":[],"_revision_number":18},{"id":"9f16617cf9f9b5f5530a26187d3e4260739315f2","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-31 23:26:07.000000000","message":"Patch Set 18: Workflow-1","accounts_in_message":[],"_revision_number":18},{"id":"05ab5cd822b526201306d6f5420de2c2bafb7901","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-01 00:10:56.000000000","message":"Patch Set 18: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dd08cd1df924f6d9457ae0a641485d2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ee8884785cea4275b3bc413976aaa665 : SUCCESS in 2m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/394b34255d814fe0bf37ef0fe3ca7ab6 : SUCCESS in 2m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81ae5b0be4414529bfc31e19a587c877 : SUCCESS in 2m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fbfd477af5634d1db6a6a4d52de597a2 : SUCCESS in 3m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c2aecc153a7a4c7395212870bdc2a19a : SUCCESS in 6m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02e3777f77bb4074a466ae60b0e85efa : SUCCESS in 2m 50s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7bcafeb31c624711a852c9e920539a76 : SUCCESS in 9m 13s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3f4826643e124edfbef83ec9cda6096f : SUCCESS in 30m 50s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7953eb1969274912bb2b6138a4c266a5 : SUCCESS in 44m 05s","accounts_in_message":[],"_revision_number":18},{"id":"4ff0d79905b3173eabec7687f98807b3a825c632","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-11-01 00:22:04.000000000","message":"Patch Set 18:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2fe4b10a76ff4ebb98caa4dc81e53a9c : SUCCESS in 48m 40s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/3b7779d5aeb84490aa21abc7ba0ac9f2 : FAILURE in 55m 11s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ccac113f78fe47b6a5942c1d50741c14 : SUCCESS in 48m 54s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/88dc784627924fe6b1530686a5cfae03 : FAILURE in 55m 22s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":18},{"id":"cb41903fc1c96e9e03ee20e761f8b6943a20cbc0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-03 01:55:08.000000000","message":"Uploaded patch set 19.","accounts_in_message":[],"_revision_number":19},{"id":"4e4209b306a669949359d9ff081e71a89afd311c","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-03 02:09:59.000000000","message":"Patch Set 19:\n\n(1 comment)","accounts_in_message":[],"_revision_number":19},{"id":"22ec7c5b02da9fa3a9737f579acb33d637673ed4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 02:43:49.000000000","message":"Patch Set 19: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f830c3d559614a1aa6fd05e1a7c37769\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e1afd6f4aec2417eb3a333681ad59e92 : SUCCESS in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/80b79312fa6841c38d9889c6fdde0e6c : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/da0c19c69a4a451196770e9398a88729 : SUCCESS in 2m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8d3e8fd1ffb9436c83ca66666e732342 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a901f492b8e472aa1bafe21f71557c2 : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/18ccd1b9f8164a63ab8ee4eaaf690b98 : SUCCESS in 3m 08s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/430930ef1c1c45a6a4609b68d3cfc69b : SUCCESS in 8m 03s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ab08e1a91f4f4ed59d3182ebc1a2edb8 : SUCCESS in 36m 16s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/658c3c175542439dbe94230055668e9c : SUCCESS in 47m 06s","accounts_in_message":[],"_revision_number":19},{"id":"cb97db1fa944070e2386b614f08168d63139e502","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-11-03 02:47:51.000000000","message":"Patch Set 19:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/945fe6022ad74ed589a55e1c24ac07ff : SUCCESS in 48m 24s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/307600fab8904e859c349d76cf3ffa0a : SUCCESS in 51m 23s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/de2ede957ae345cca613821bdd977fd9 : SUCCESS in 49m 41s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d5dafa5cdd7f4af8a7ccf29ec1af0a3b : SUCCESS in 49m 15s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":19},{"id":"7613cabd00700e477e33eb1ce1bf2f367ace7387","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-03 12:00:03.000000000","message":"Uploaded patch set 20.","accounts_in_message":[],"_revision_number":20},{"id":"adc7d8434bb97d19b36645c7c1a031d87c46944a","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-11-03 12:56:16.000000000","message":"Patch Set 20:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2b82e051901c4b4cb3d41ef3015db8d2 : SUCCESS in 49m 50s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4516a61719c14b16b7e9aacee03b11fb : SUCCESS in 52m 23s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f16ede8a4dc247b2ae9d52230fd869d6 : SUCCESS in 51m 33s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/cac5feaffb484fe3b6a99593843416da : SUCCESS in 50m 54s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":20},{"id":"f631a53d6e5fce8c6174547511da6e778d3783da","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-03 13:01:27.000000000","message":"Patch Set 20: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d954180e08514163bab3e600859faedc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1795124d0f204cb59508c93598d000aa : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/979183b2fe434ee686f316bfd7722a56 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/12d84afb5a4b40ecab24d684060807f2 : SUCCESS in 2m 55s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/14641c370d514ee1bf7d64db7bb4aa30 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/441fab784623461db0d27c20d95f97fa : SUCCESS in 5m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31e28b1c65c943d2bd6b7674becaac05 : SUCCESS in 3m 19s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8f441492cb694944960856b7f65ccd52 : SUCCESS in 8m 21s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2ed724ad01964b0c9a49855cc8591ddd : SUCCESS in 45m 04s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b96edeacf3bb4cd6b59bb31019a8d703 : SUCCESS in 54m 25s","accounts_in_message":[],"_revision_number":20},{"id":"063aa33ddf930d6720167f836c9f80456e0a2e6c","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-11-04 09:04:02.000000000","message":"Patch Set 20: Code-Review+1\n\n(5 comments)","accounts_in_message":[],"_revision_number":20},{"id":"c3532185185b0c7d6f43303336dadd9d6a35c800","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-07 12:45:09.000000000","message":"Uploaded patch set 21.","accounts_in_message":[],"_revision_number":21},{"id":"bb79324ac5d4770b8f62e9bd9cce7f3ff88bc7aa","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-07 12:45:33.000000000","message":"Patch Set 20:\n\n(4 comments)","accounts_in_message":[],"_revision_number":20},{"id":"2e400e69c69c316e078283bda01c4d0a38019702","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-07 13:46:13.000000000","message":"Patch Set 21: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d1d23029b7c4ba8b690cb1314e81a24\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2d20b9ad6a546a8a9d6de4f066a78ce : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/83998981aaa44de2a2531cf35c1e586d : SUCCESS in 2m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c545aefbe1924fd7bb9962fa1be3b94d : SUCCESS in 2m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2cb0283f28f946558266c8f55f78b633 : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf76367564e34cdf9ef5bcfbc2d15d91 : SUCCESS in 4m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73540d21b08f4974870eef58d5796d55 : SUCCESS in 3m 23s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4838160f76b54acab8f87eb3fcad94c6 : SUCCESS in 8m 51s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b8393e6a5a9b467d891e0c51fe21635f : SUCCESS in 45m 01s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c367aa3cb0f54481b05958c36274483e : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":21},{"id":"aec46beefaeb43daefd4be836cfc4991d0a43e6f","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-11-09 10:14:02.000000000","message":"Patch Set 21: Code-Review+2\n\n(3 comments)","accounts_in_message":[],"_revision_number":21},{"id":"44d80983972b9dd2d439fdb3bac8df1bdb698b11","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-09 11:45:19.000000000","message":"Uploaded patch set 22.","accounts_in_message":[],"_revision_number":22},{"id":"e8cbf058317ba6abcb7bf40fe455136473daa2b4","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-09 11:46:52.000000000","message":"Patch Set 22:\n\n(4 comments)","accounts_in_message":[],"_revision_number":22},{"id":"5784d082c561062a27e84f081ea11e39b3673df5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-09 11:50:11.000000000","message":"Uploaded patch set 23.","accounts_in_message":[],"_revision_number":23},{"id":"87dafb7048b49920c79fc13e05c24f508dfdf13b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-09 12:48:29.000000000","message":"Patch Set 23: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/339ff415b02c4f13be9b90688efaa684\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a945f7340d54e60a8080fbe81b306ea : SUCCESS in 2m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a446e7fda9454ae28b0b6816384d4be9 : SUCCESS in 2m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bc76ff1fdee478b8bf4bffccda267e7 : SUCCESS in 2m 44s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/83ee538b73b249e5b0db2780c41ed8eb : SUCCESS in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d02209fde0bf426cba5f5f0e3c197fe0 : SUCCESS in 4m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd0c00b6ad6e42c0b35f361a79c89ddd : SUCCESS in 2m 44s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4f863751574a44da8ee8fd3eb8773e98 : SUCCESS in 8m 42s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d077d95dea78464b9866040e0bf29873 : SUCCESS in 29m 32s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8f1c66f6f81548df94231cf42a450d2b : SUCCESS in 51m 57s","accounts_in_message":[],"_revision_number":23},{"id":"628da1672c913c39fa8a8c8bdad982663f3b95b6","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-11-09 14:44:14.000000000","message":"Patch Set 23: Code-Review+2\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"d6228ebd4b12607432a8fe3fe627ee89d9e6c2c5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-11-09 15:36:15.000000000","message":"Patch Set 23:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a683bc5de4dd43a685a3bd57d4bea040 : SUCCESS in 51m 26s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/81aab8e8d3234a0fb3d61dc266fcd32e : SUCCESS in 1h 07m 08s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/588873549c294ba299aba2e20d387877 : FAILURE in 32m 02s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c9528124c504403195f47819c238416d : SUCCESS in 51m 40s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":23},{"id":"b856a51a363abeb43ad8f974eb50e272c601e471","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-11-10 08:16:13.000000000","message":"Patch Set 23:\n\n(2 comments)","accounts_in_message":[],"_revision_number":23},{"id":"d9c6ddbd1252aea0c6598c031cca2dc0cf717f27","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-10 10:36:46.000000000","message":"Uploaded patch set 24.","accounts_in_message":[],"_revision_number":24},{"id":"f2007523ff0bd3316105789da4d6207e10611b7d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-10 11:17:42.000000000","message":"Patch Set 24: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a80278506ea948339ad2a10dba28a90d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f7394b3962347a3b2fc41205dccb4b2 : SUCCESS in 3m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3d8372576a2c44ccac380bb0ed1e8db3 : SUCCESS in 2m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/7e8bd042840547f8b92861a98094d3a3 : SUCCESS in 2m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5bcf5c5fb79f4533b1e56b154d2d50bf : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d63e72a9923a4b4b92b9950d867a56ee : SUCCESS in 4m 36s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ecfd101ef77d4606b59a7b9e6b96b3f8 : SUCCESS in 3m 22s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/421368f9aeee49bc9f01bf8a52a9d61c : SUCCESS in 6m 13s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/29e4960054a04ceb9c7273f8490d4f5c : SUCCESS in 36m 56s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/faf1639b27bf48ff89c53d57d2f014aa : SUCCESS in 39m 45s","accounts_in_message":[],"_revision_number":24},{"id":"95f2c352c4d759a250de10840f408e9153b93cd5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-11-10 11:29:04.000000000","message":"Patch Set 24:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/bd07e9a5fb0046a2addac352ee6efc7c : SUCCESS in 48m 31s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/a6d78d029ac24a9bb557f447cb6154b9 : SUCCESS in 51m 27s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/726198e7025b4bbf9341ddf2738b97f8 : SUCCESS in 49m 43s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/84827ff1eb074ad89d367bf91a295ae8 : SUCCESS in 47m 59s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":24},{"id":"1af083d185591112755ef585f3bee916718cac14","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-11 05:13:28.000000000","message":"Patch Set 24:\n\n(4 comments)","accounts_in_message":[],"_revision_number":24},{"id":"a79b1f8a5bd70a1d505864671867810213f9711e","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-11-11 08:34:18.000000000","message":"Patch Set 24: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"89151dc1c621cf78e8ff52558b2fc64ade697f92","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-17 07:11:07.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"f17dd17677e8f7cca71bd7885c6052f7e2f26e27","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-11-17 08:12:50.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"1048552251e58e9dfe86c586adb6e864a0e73b79","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2022-11-17 16:56:50.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"4284fdeffdf6b53f7f4ae8f157d658f2088314fa","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-18 10:04:26.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"e058ae395ac910084c8bf75620f068597cc162a6","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-11-21 11:58:30.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"4408152b6abd15ec166ef3009fe2f042d71cc98f","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-21 12:42:44.000000000","message":"Patch Set 24:\n\n(1 comment)","accounts_in_message":[],"_revision_number":24},{"id":"ae57efb722717b3f8fb590363b867c7e8397ee47","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2022-11-21 18:41:14.000000000","message":"Patch Set 24: Code-Review+2","accounts_in_message":[],"_revision_number":24},{"id":"6e6445d05a8fe14500ae40e860d1a3aaf3f195da","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-22 00:30:01.000000000","message":"Uploaded patch set 25.","accounts_in_message":[],"_revision_number":25},{"id":"e58c6fe06c1822a1741eaa51025c403c84607d3b","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-11-22 00:32:55.000000000","message":"Patch Set 25:\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"d7af9083d6ca7f531d8218e0df96294a0c5f88e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 01:17:00.000000000","message":"Patch Set 25: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5cf3f840e5844de4b5ec91d7a1c74dfa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6a8a294f70374b91a4f792813f1140b4 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bb04e07c7204d34b6751a5e48a2fd7f : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/12c6ddd221b14262a85476f2005270ce : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/849e3d44504540a4bb915ba5f77ee045 : SUCCESS in 3m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5ac69fc2d97c4ff0824c6e7029b3883d : SUCCESS in 4m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/997aa45f948a4868ac905a513913666e : SUCCESS in 2m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/cbce933647df4f7ea7fd1a904bbf7bb2 : SUCCESS in 7m 56s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/663ed9c1b4574b50b40c2fd254b61a08 : SUCCESS in 30m 34s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/33fa7751472c44b9806cce4b68dbc54f : SUCCESS in 43m 28s","accounts_in_message":[],"_revision_number":25},{"id":"dd68aaaa81684c9da332a05667e37567edfab805","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-11-22 01:51:49.000000000","message":"Patch Set 25:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/0563779b7d1047ec94087e920d7200cc : SUCCESS in 1h 04m 21s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f4a00412c7894dc4b15b592ef3aa11f1 : SUCCESS in 1h 10m 07s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c0c3e02648804698956a0fb9b0756f4a : FAILURE in 25m 17s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/fa5e0f0b20414d8e8ab55e98d2f7a17a : SUCCESS in 49m 36s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":25},{"id":"d230ab995d4fc631a8c8b54880fc40d31f383de9","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-11-22 07:45:06.000000000","message":"Patch Set 25: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":25},{"id":"a64a698268ac0ee057a97158a4c6fbc6c6dbd027","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-11-22 08:18:01.000000000","message":"Patch Set 25: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":25},{"id":"858e6e4ef1dc5b53fcbb9832614542ea99a194d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 08:18:37.000000000","message":"Patch Set 25: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":25},{"id":"99cd61dcd66773865bc12550578a477eaa0d1651","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 09:20:37.000000000","message":"Patch Set 25: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bec0bda4123847398e79133640312678\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3741a3311d734d3e808c7d06f0187251 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a69206806e764417abe02a29f520dc8f : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/37ce2ff2875f4de1bd887a29d2e78de4 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c28369907f8b4fc08b45dcebca066ece : SUCCESS in 5m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40b2951645bb4f47bef21af9c67871b6 : SUCCESS in 3m 47s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4a9b972e4a164dd592ac57ab82218a26 : SUCCESS in 8m 20s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b1ab49d5b51b4370a070b163380433c6 : SUCCESS in 50m 53s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4502ab04656b4340bd3ccbd140184ba0 : SUCCESS in 52m 30s","accounts_in_message":[],"_revision_number":25},{"id":"cdbcbdf202a793396c6e5e2c01912370c32ebaf9","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 09:20:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":25},{"id":"c57b80c423968c6a0e0e04036c7630bbc6c271b6","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-11-22 09:21:49.000000000","message":"Patch Set 25:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a3e483b89714bc2b6fcff6d1f3b237d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82ce6b54494b49a7ba9184b4fc4a6dc0 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a7e5a39373104c2aa10c84c87985e4a4 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":25}],"current_revision_number":25,"current_revision":"2c8229046c6ca8aa342e028c96aaa066f517eb9f","revisions":{"196ea2d2552bae2dd35c7fe644c81b614b6b5cb1":{"kind":"REWORK","_number":1,"created":"2022-09-09 04:50:55.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/1"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:50:49.000000000","tz":600},"subject":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only","message":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/196ea2d2552bae2dd35c7fe644c81b614b6b5cb1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/196ea2d2552bae2dd35c7fe644c81b614b6b5cb1"}]},"branch":"refs/heads/master"},"5e79b33a98ee01aaed088d94680a7c7eb4d11a3e":{"kind":"REWORK","_number":2,"created":"2022-09-09 11:46:28.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/2"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 11:46:15.000000000","tz":600},"subject":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only","message":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5e79b33a98ee01aaed088d94680a7c7eb4d11a3e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5e79b33a98ee01aaed088d94680a7c7eb4d11a3e"}]},"branch":"refs/heads/master"},"33ccaeb414c0441f15aa19e2d237fbb21cba875c":{"kind":"REWORK","_number":3,"created":"2022-09-22 01:11:05.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/3"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-22 01:10:17.000000000","tz":600},"subject":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only","message":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33ccaeb414c0441f15aa19e2d237fbb21cba875c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33ccaeb414c0441f15aa19e2d237fbb21cba875c"}]},"branch":"refs/heads/master"},"86e7dd86da3d225c2ed3788faf71ba0ba81b8cc1":{"kind":"REWORK","_number":4,"created":"2022-09-22 11:19:18.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/4"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-22 11:19:04.000000000","tz":600},"subject":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only","message":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/86e7dd86da3d225c2ed3788faf71ba0ba81b8cc1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/86e7dd86da3d225c2ed3788faf71ba0ba81b8cc1"}]},"branch":"refs/heads/master"},"e576b74e85ebce7532c9dcd96d8f482944f4403d":{"kind":"REWORK","_number":5,"created":"2022-09-27 12:40:31.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/5"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-27 12:40:17.000000000","tz":600},"subject":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only","message":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e576b74e85ebce7532c9dcd96d8f482944f4403d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e576b74e85ebce7532c9dcd96d8f482944f4403d"}]},"branch":"refs/heads/master"},"47cadc5b2988aa1f86631f5b46c4fe7bad5bfa30":{"kind":"REWORK","_number":6,"created":"2022-09-27 12:52:15.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/6"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-27 12:51:58.000000000","tz":600},"subject":"Retry BootSourceOverride request when SettingsURI is read-only","message":"Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/47cadc5b2988aa1f86631f5b46c4fe7bad5bfa30"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/47cadc5b2988aa1f86631f5b46c4fe7bad5bfa30"}]},"branch":"refs/heads/master"},"25cf8f117f0bb92884f505ba9dda57c6358bf1df":{"kind":"REWORK","_number":7,"created":"2022-09-27 20:35:49.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/7"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-27 20:35:34.000000000","tz":600},"subject":"Retry BootSourceOverride request when SettingsURI is read-only","message":"Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/25cf8f117f0bb92884f505ba9dda57c6358bf1df"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/25cf8f117f0bb92884f505ba9dda57c6358bf1df"}]},"branch":"refs/heads/master"},"9646718933e4f24283dbf5ba65691cd876c81f28":{"kind":"REWORK","_number":8,"created":"2022-09-28 01:31:11.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/8"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-28 01:30:59.000000000","tz":600},"subject":"Retry BootSourceOverride request when SettingsURI is read-only","message":"Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9646718933e4f24283dbf5ba65691cd876c81f28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9646718933e4f24283dbf5ba65691cd876c81f28"}]},"branch":"refs/heads/master"},"e6d40325c2f533ad580dcfdbd032b9015beed0cf":{"kind":"REWORK","_number":9,"created":"2022-09-28 04:57:17.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/9"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-28 04:56:59.000000000","tz":600},"subject":"Retry BootSourceOverride request when SettingsURI is read-only","message":"Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e6d40325c2f533ad580dcfdbd032b9015beed0cf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e6d40325c2f533ad580dcfdbd032b9015beed0cf"}]},"branch":"refs/heads/master"},"06dcc185c7856437079790c0b64377f8763754ea":{"kind":"REWORK","_number":10,"created":"2022-10-24 12:57:53.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/10"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-24 12:57:31.000000000","tz":600},"subject":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only","message":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/06dcc185c7856437079790c0b64377f8763754ea"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/06dcc185c7856437079790c0b64377f8763754ea"}]},"branch":"refs/heads/master"},"49d67ada603aa1bc0fe1a4ba3d173c918c64cd14":{"kind":"REWORK","_number":11,"created":"2022-10-25 13:24:35.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/11"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-25 13:24:20.000000000","tz":600},"subject":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only","message":"[WIP] Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/49d67ada603aa1bc0fe1a4ba3d173c918c64cd14"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/49d67ada603aa1bc0fe1a4ba3d173c918c64cd14"}]},"branch":"refs/heads/master"},"398a195995173bd32b5078e63dc97c5a9a2f9409":{"kind":"REWORK","_number":12,"created":"2022-10-26 05:14:26.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/12"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-26 05:14:11.000000000","tz":600},"subject":"Retry BootSourceOverride request when SettingsURI is read-only","message":"Retry BootSourceOverride request when SettingsURI is read-only\n\nOn Lenovo SE450, System Settings URI advertised by the BMC is\ncurrently read-only and doesn\u0027t allow setting BootSourceOverride.\nIf this condition is detected, as a work around, re-try the request\ndisregarding Settings URI advertised by the BMC.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/398a195995173bd32b5078e63dc97c5a9a2f9409"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/398a195995173bd32b5078e63dc97c5a9a2f9409"}]},"branch":"refs/heads/master"},"33413d26adca211b949c088ad8528d3d78e4cd29":{"kind":"REWORK","_number":13,"created":"2022-10-26 05:26:12.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/13"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-26 05:23:35.000000000","tz":600},"subject":"Retry BootSourceOverride request when SettingsURI is read-only","message":"Retry BootSourceOverride request when SettingsURI is read-only\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33413d26adca211b949c088ad8528d3d78e4cd29"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/33413d26adca211b949c088ad8528d3d78e4cd29"}]},"branch":"refs/heads/master"},"edf9208025be30933ea7204674552c57b66e1dc7":{"kind":"NO_CODE_CHANGE","_number":14,"created":"2022-10-26 05:27:49.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/14"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-26 05:26:37.000000000","tz":600},"subject":"Only use SettingsURI for BootSourceOverride if the attribute is present in GET response","message":"Only use SettingsURI for BootSourceOverride if the attribute is present\nin GET response\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/edf9208025be30933ea7204674552c57b66e1dc7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/edf9208025be30933ea7204674552c57b66e1dc7"}]},"branch":"refs/heads/master"},"18da924389594ade34e151e2a4201c67789b5f0c":{"kind":"REWORK","_number":15,"created":"2022-10-26 13:08:28.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/15","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/15","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/15 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/15 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/15 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/15"}}},"commit":{"parents":[{"commit":"19a0671fb88ac8c3bc7890bc26e616c0f7be26be","subject":"Add Python3 antelope unit tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/19a0671fb88ac8c3bc7890bc26e616c0f7be26be"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-26 13:08:16.000000000","tz":600},"subject":"Only use SettingsURI for BootSourceOverride if the attribute is present in GET response","message":"Only use SettingsURI for BootSourceOverride if the attribute is present\nin GET response\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/18da924389594ade34e151e2a4201c67789b5f0c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/18da924389594ade34e151e2a4201c67789b5f0c"}]},"branch":"refs/heads/master"},"caf12318429b412752b105be526d7047c9c825c5":{"kind":"REWORK","_number":16,"created":"2022-10-31 07:38:36.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/16","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/16","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/16 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/16 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/16 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/16"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-31 07:38:32.000000000","tz":600},"subject":"Only use SettingsURI for BootSourceOverride if the attribute is present in GET response","message":"Only use SettingsURI for BootSourceOverride if the attribute is present\nin GET response\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/caf12318429b412752b105be526d7047c9c825c5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/caf12318429b412752b105be526d7047c9c825c5"}]},"branch":"refs/heads/master"},"a0a6c56dafd4e857a39026ac1136fef1976c3ab1":{"kind":"REWORK","_number":17,"created":"2022-10-31 11:22:21.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/17","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/17","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/17 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/17 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/17 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/17"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-31 11:22:17.000000000","tz":600},"subject":"Only use SettingsURI for BootSourceOverride if the attribute is present in GET response","message":"Only use SettingsURI for BootSourceOverride if the attribute is present\nin GET response\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a0a6c56dafd4e857a39026ac1136fef1976c3ab1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a0a6c56dafd4e857a39026ac1136fef1976c3ab1"}]},"branch":"refs/heads/master"},"568aacec2fc406aa901d6c9826d50d1c01cbb280":{"kind":"REWORK","_number":18,"created":"2022-10-31 23:25:25.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/18","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/18","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/18 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/18 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/18 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/18"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-31 23:25:13.000000000","tz":600},"subject":"Only use SettingsURI for BootSourceOverride if the attribute is present in GET response","message":"Only use SettingsURI for BootSourceOverride if the attribute is present\nin GET response\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/568aacec2fc406aa901d6c9826d50d1c01cbb280"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/568aacec2fc406aa901d6c9826d50d1c01cbb280"}]},"branch":"refs/heads/master"},"9cb9b50b94be842f215e984c662fee33a618cb25":{"kind":"REWORK","_number":19,"created":"2022-11-03 01:55:08.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/19","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/19","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/19 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/19 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/19 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/19"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-11-03 01:54:48.000000000","tz":600},"subject":"Only use SettingsURI for BootSourceOverride if the attribute is present in GET response","message":"Only use SettingsURI for BootSourceOverride if the attribute is present\nin GET response\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9cb9b50b94be842f215e984c662fee33a618cb25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9cb9b50b94be842f215e984c662fee33a618cb25"}]},"branch":"refs/heads/master"},"649d8079ef4d72212f5d1440b443f35d1f39dab4":{"kind":"REWORK","_number":20,"created":"2022-11-03 12:00:03.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/20","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/20","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/20 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/20 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/20 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/20"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-11-03 11:59:51.000000000","tz":600},"subject":"Only use SettingsURI for BootSourceOverride if the attribute is present in GET response","message":"Only use SettingsURI for BootSourceOverride if the attribute is present\nin GET response\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/649d8079ef4d72212f5d1440b443f35d1f39dab4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/649d8079ef4d72212f5d1440b443f35d1f39dab4"}]},"branch":"refs/heads/master"},"abcd0a21bfa183d51778c398bb2e00e130801c90":{"kind":"REWORK","_number":21,"created":"2022-11-07 12:45:09.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/21","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/21","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/21 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/21 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/21 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/21"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-11-07 12:42:24.000000000","tz":600},"subject":"Fix setting boot related attributes","message":"Fix setting boot related attributes\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/abcd0a21bfa183d51778c398bb2e00e130801c90"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/abcd0a21bfa183d51778c398bb2e00e130801c90"}]},"branch":"refs/heads/master"},"14c4bdae7de20bd2b45f470c5aba2f6e5754bda3":{"kind":"REWORK","_number":22,"created":"2022-11-09 11:45:19.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/22","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/22","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/22 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/22 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/22 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/22"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-11-09 11:45:03.000000000","tz":600},"subject":"Fix setting boot related attributes","message":"Fix setting boot related attributes\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/14c4bdae7de20bd2b45f470c5aba2f6e5754bda3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/14c4bdae7de20bd2b45f470c5aba2f6e5754bda3"}]},"branch":"refs/heads/master"},"97cb69ecdc1b49cddcb4de46bdf233a6445f0d35":{"kind":"REWORK","_number":23,"created":"2022-11-09 11:50:11.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/23","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/23","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/23 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/23 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/23 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/23"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-11-09 11:49:56.000000000","tz":600},"subject":"Fix setting boot related attributes","message":"Fix setting boot related attributes\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/97cb69ecdc1b49cddcb4de46bdf233a6445f0d35"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/97cb69ecdc1b49cddcb4de46bdf233a6445f0d35"}]},"branch":"refs/heads/master"},"d814412087e7e6b2064c8d9aef81230f5ce14202":{"kind":"REWORK","_number":24,"created":"2022-11-10 10:36:46.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/24","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/24","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/24 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/24 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/24 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/24"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-11-10 10:36:32.000000000","tz":600},"subject":"Fix setting boot related attributes","message":"Fix setting boot related attributes\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d814412087e7e6b2064c8d9aef81230f5ce14202"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d814412087e7e6b2064c8d9aef81230f5ce14202"}]},"branch":"refs/heads/master"},"2c8229046c6ca8aa342e028c96aaa066f517eb9f":{"kind":"REWORK","_number":25,"created":"2022-11-22 00:30:01.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/97/856597/25","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/97/856597/25","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/25 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/25 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/97/856597/25 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/97/856597/25"}}},"commit":{"parents":[{"commit":"5d6e5486fab14f6934b4f510f71a304acbf18351","subject":"Merge \"Improve resiliency of eTag handling\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d6e5486fab14f6934b4f510f71a304acbf18351"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-09 04:42:09.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-11-22 00:29:47.000000000","tz":600},"subject":"Fix setting boot related attributes","message":"Fix setting boot related attributes\n\nEnhance handling of System Settings URI so that it is only used\nfor PATCH requests needed for boot sequence overrides if\nthe attributes in question are present in response to a GET\nrequest against SettingsURI. If the attributes are absent, fall\nback to System URI.\n\nChange-Id: If25e30bfb9bf06fdbba6051177947d4d285ae2ab\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2c8229046c6ca8aa342e028c96aaa066f517eb9f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2c8229046c6ca8aa342e028c96aaa066f517eb9f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
