)]}'
{"id":"openstack%2Fsushy~862847","triplet_id":"openstack%2Fsushy~stable%2Fyoga~Iadf19b77ff3132b0533d317e46032b9d5a0e4e33","project":"openstack/sushy","branch":"stable/yoga","topic":"856123-yoga-backport","attention_set":{},"removed_from_attention_set":{"32177":{"account":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"last_update":"2022-10-28 22:20:03.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Iadf19b77ff3132b0533d317e46032b9d5a0e4e33","subject":"Improve resiliency of eTag handling","status":"MERGED","created":"2022-10-28 01:18:35.000000000","updated":"2022-10-28 22:21:04.000000000","submitted":"2022-10-28 22:20:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"862847-856123-yoga-backport","meta_rev_id":"b8b52d36887ebba01780ea164e724bfdbae6f8fe","_number":862847,"virtual_id_number":862847,"owner":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-10-28 22:20:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2022-10-28 21:29:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2022-10-28 21:29:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-10-28 02:19:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-10-28 02:59:52.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-10-28 21:29:16.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"cda1051db8d77e78472d284dede0e8f56e9d5e02","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-28 01:18:35.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"43ca8b112350010e8c99aaeb9f68e80309b0de5f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 02:19:00.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/99b6d84877814accab083ea5feaaaa2e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4a609bb4386f41e2904ab4c10fe597ad : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad50341b90ae4559aa224b81f5629b51 : SUCCESS in 2m 39s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/423c3719c6c144dfa8aad1a20c41c5be : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/40f86bc6b7b24ea5b548fb37571c1af5 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5631aa7e6bfe4d11b092f8a2bd0e0d31 : SUCCESS in 5m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4e808c42e1f44c808a0da8e038d73922 : SUCCESS in 3m 47s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c7e53e90ebe84e398573eeac96647253 : SUCCESS in 8m 11s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9e6f0b79422f4375ad372383949dff8f : SUCCESS in 53m 34s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/019fb862d4334ad88e4e40216f69a476 : SUCCESS in 43m 59s","accounts_in_message":[],"_revision_number":1},{"id":"43a67422f3c6fd179ebb36fd2a61bd543c5d33a0","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-10-28 02:59:52.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/2d8c51cf75f54fedbb7fd428c776ead5 : FAILURE in 58m 39s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/9dc94510da484947a67e7b3aeacf00c9 : FAILURE in 57m 59s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/19e5b0eca5ca4a65a4abc1d55d4f4ae1 : SUCCESS in 1h 37m 36s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ac93ab11ab804371aefb138e6bd87460 : FAILURE in 54m 42s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"0d5b07e63954132c5ae97be0828fc557a57d27b4","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"date":"2022-10-28 12:18:15.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"c11af4cd5631b10925b2aa776b05ed95e483b951","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-10-28 21:29:16.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"2bb5db8fce7777b352ab4a496c5758f118bbcaab","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 21:29:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"0400c6419f776fb100e3379ad1d7b6fcf1898e8e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 22:20:03.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8efc93d87fd7429fb922cc136918c449\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/00d14c7271d7496f8460dcc48c74f7f9 : SUCCESS in 2m 25s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c8e475ea6ec249a3be8f0d37ae92730f : SUCCESS in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9efd38f1f8ab45f5ab335973f62c6c73 : SUCCESS in 2m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a68cd08e16b94986b123d771171faddb : SUCCESS in 4m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/13863fdca77e4294824a2375be55effe : SUCCESS in 2m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/72bfb2df9fc34985b6e840e2e41a78b2 : SUCCESS in 8m 00s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/fdc656069927457fb95da826ee8f6a3f : SUCCESS in 36m 09s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7166e4cf1826452db37a8f224377dd8d : SUCCESS in 49m 33s","accounts_in_message":[],"_revision_number":1},{"id":"6a900b23441c86956d12ac17644a3f4e7a640fec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 22:20:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"b8b52d36887ebba01780ea164e724bfdbae6f8fe","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-10-28 22:21:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/654e9e207eca4af4882de66c3fa464c8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b51d8c3612764884ad23a0c193169b6b : SUCCESS in 41s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2cfcaf163faa426da25f7fdb4828f9c0 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"1da4385cc2e6b4722474df676849f337a8df5b9b","revisions":{"1da4385cc2e6b4722474df676849f337a8df5b9b":{"kind":"REWORK","_number":1,"created":"2022-10-28 01:18:35.000000000","uploader":{"_account_id":32177,"name":"Jacob Anders","email":"jacob-anders-dev@proton.me","username":"janders"},"ref":"refs/changes/47/862847/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/47/862847/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/862847/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/862847/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/47/862847/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/47/862847/1"}}},"commit":{"parents":[{"commit":"1ae8e49d4f63efb13ae75aff5a7d0969de92f02f","subject":"Capture requests errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1ae8e49d4f63efb13ae75aff5a7d0969de92f02f"}]}],"author":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-09-06 13:55:59.000000000","tz":600},"committer":{"name":"Jacob Anders","email":"janders@redhat.com","date":"2022-10-28 01:18:19.000000000","tz":600},"subject":"Improve resiliency of eTag handling","message":"Improve resiliency of eTag handling\n\nSome servers manifest non-standard way of implementing eTags. This\npatch attempts to improve resiliency of handling eTag related ops\nby re-trying failing request with known work arounds.\n\nChange-Id: Iadf19b77ff3132b0533d317e46032b9d5a0e4e33\n(cherry picked from commit 3e69fce709765e317cfd0288b7c119e757f94117)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1da4385cc2e6b4722474df676849f337a8df5b9b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/1da4385cc2e6b4722474df676849f337a8df5b9b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[]}
