)]}'
{"id":"openstack%2Fsushy~867209","triplet_id":"openstack%2Fsushy~stable%2Fzed~I063f5c3a280262f1a1d70b44906a504e2b16134d","project":"openstack/sushy","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2022-12-16 14:05:39.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I063f5c3a280262f1a1d70b44906a504e2b16134d","subject":"Handle proper code_status in unit test","status":"MERGED","created":"2022-12-12 11:26:58.000000000","updated":"2022-12-16 15:14:01.000000000","submitted":"2022-12-16 14:05:39.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867209","meta_rev_id":"e99c659c01b7e342c3ab4caeaf273c9ec1913d41","_number":867209,"virtual_id_number":867209,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2022-12-16 14:05:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":2,"date":"2022-12-15 07:23:33.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2022-12-16 15:14:01.000000000","post_submit":true,"permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2022-12-16 13:52:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2022-12-16 13:52:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-12 11:40:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2022-12-13 17:57:46.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-12-15 07:23:33.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2022-12-16 13:52:18.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2022-12-16 15:14:01.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"6ec00efa58969fa66a98a4628376dff3651a08fe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-12 11:26:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e1f046b26cef844519a45558346d43a9281300a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-12 11:27:45.000000000","message":"Patch Set 2: Commit message was updated.","accounts_in_message":[],"_revision_number":2},{"id":"6cf1d70ca528ab641c94086aacdfb4ef4b871146","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-12 11:40:29.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/73606c2ef4fb4af59b21d322635ca9ea\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/99d6d37be5c74813ba3d5a670079b169 : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/da54a662dfc1400e878370aa0f6a613d : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a83e9012dc904ffdb6cdd978b98ede70 : SUCCESS in 2m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b1c931770c99413582849c3a003e43b4 : SUCCESS in 3m 24s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/cf6e8958274545b8bfc2ad5acf4b6c69 : SUCCESS in 2m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4dd2ca04805447cb9de25ad5176f6aac : SUCCESS in 5m 02s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1cc22e0251104cbc82f162290d7ec12b : SUCCESS in 8m 00s","accounts_in_message":[],"_revision_number":2},{"id":"0c49deaf41d4fbd8853692747c8f6d8806afe469","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-13 16:40:09.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"2eceea866b2e89f60fee64c6e05ee43edc28ad20","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 16:57:47.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/00ac7ec305b8448fbe7814e88e4af166\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f34fb761f004400399e68ea34b248447 : FAILURE in 3m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1d156b1db2ea45649659e34b63b38de0 : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11a0fb6ca3414e159004d35063bf193d : FAILURE in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/37a8795f7a1a4ce09c73c41fda514e6f : FAILURE in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fcdaf13aa5b4f60acaa3edf85761a9b : FAILURE in 2m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/139bfd48fbc54f72a11b0d3ba5386a89 : SUCCESS in 4m 32s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9976930f77e24f78ac23cca6419c6582 : SUCCESS in 7m 22s","accounts_in_message":[],"_revision_number":3},{"id":"acad73a2de6f5447a7c948133a2531e1b0f3027b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-13 17:04:03.000000000","message":"Patch Set 4: Published edit on patch set 3.","accounts_in_message":[],"_revision_number":4},{"id":"73e49224854bc068e52f1f1dd5e60a6eef54e0e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-13 17:13:31.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2f5c502ffdc419cb4b48ec274d139ae\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fc27d75f9104f1e9f2afb203df82b6b : SUCCESS in 3m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f40378e1b434280a4931e435892b9dd : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a7ec96552ebb45e78c5e9257ea15f8ab : SUCCESS in 2m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/07ff6acfce3744e4a9b73dc358c1c123 : SUCCESS in 3m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5179aeda59554e6fbb7acd9979a1ad2b : SUCCESS in 3m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a8ab516e664f7da4a5561d3aeec262 : SUCCESS in 5m 30s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2564fbfaeeff4e489df35b81d4cebf64 : SUCCESS in 8m 06s","accounts_in_message":[],"_revision_number":4},{"id":"b8e33d50b759efe809aee11e7662a0dc44383d12","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-12-13 17:57:46.000000000","message":"Patch Set 4:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/d42db87ebb534db18e80f39075f4fcbe : FAILURE in 26m 47s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/c404459ee4f74ada9452f8b459103fcf : FAILURE in 27m 13s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ac471a8f0e2f43748e7fab7d717a399f : FAILURE in 26m 53s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/6bfddf5764c64161ab629a1009abd109 : SUCCESS in 51m 58s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":4},{"id":"79e3a32f005e5eea1b0f10619d5885ec644d0832","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2022-12-14 08:14:39.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":4},{"id":"e2afba26df7da3e54d098a05eeb98c6e28eab15d","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2022-12-15 07:23:33.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"d52fd6a544eedac3853ffe5dd57bac3b607dab30","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2022-12-16 13:52:18.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"5ce6519012312de88f30d4859798264b978f3281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 13:52:38.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"8780614bb9b1ca65f251e10514898db33686af61","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 14:05:38.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b29acfd4d8764e58bd66202a4ad0eb16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1fa6f5042f01423a9c6c58f8e00b2de0 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c1f11eab8bb04829b719c716e3596737 : SUCCESS in 2m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f0e68790f8844b6cbff996051b7a962e : SUCCESS in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d280390aaebe4f6eb9929bd261ec7d36 : SUCCESS in 5m 19s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2487bb6590324d3091ae3400053adfbd : SUCCESS in 8m 30s","accounts_in_message":[],"_revision_number":4},{"id":"ba56e82765f4df48b59ae435b419002dcf1c06f8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 14:05:39.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"50ffc5b7807986374138cee50c105d9937733fbf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 14:06:39.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5b3d995036e145f1b37eb5e9b1035b5b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/82109c6590f7441f927c3b2f704bac82 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":4},{"id":"e99c659c01b7e342c3ab4caeaf273c9ec1913d41","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2022-12-16 15:14:01.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"5e2807a79a105f4dda68c549c7abf7b02d4cb855","revisions":{"ff695b998c9edf193b48b1418515d42c2ae7e1f9":{"kind":"REWORK","_number":1,"created":"2022-12-12 11:26:58.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/09/867209/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/09/867209/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/09/867209/1"}}},"commit":{"parents":[{"commit":"6489bd2ea158cc3a12c0d15311bbaf62a8503d8a","subject":"Fix setting boot related attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6489bd2ea158cc3a12c0d15311bbaf62a8503d8a"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-06 16:22:42.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-12 11:26:48.000000000","tz":60},"subject":"Handle proper code_status in unit test","message":"Handle proper code_status in unit test\n\nPython 3.11 correctly resolve the code_status with its number.\n\nSee https://bugs.debian.org/cgi-bin/bugreport.cgi?bug\u003d1025124 for\nreference.\n\nChange-Id: I063f5c3a280262f1a1d70b44906a504e2b16134d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ff695b998c9edf193b48b1418515d42c2ae7e1f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ff695b998c9edf193b48b1418515d42c2ae7e1f9"}]},"branch":"refs/heads/stable/zed"},"94f067d18d8f52a32058545db3756936a0f585fe":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2022-12-12 11:27:45.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/09/867209/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/09/867209/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/09/867209/2"}}},"commit":{"parents":[{"commit":"6489bd2ea158cc3a12c0d15311bbaf62a8503d8a","subject":"Fix setting boot related attributes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6489bd2ea158cc3a12c0d15311bbaf62a8503d8a"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-06 16:22:42.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-12 11:27:45.000000000","tz":0},"subject":"Handle proper code_status in unit test","message":"Handle proper code_status in unit test\n\nPython 3.11 correctly resolve the code_status with its number.\n\nSee https://bugs.debian.org/cgi-bin/bugreport.cgi?bug\u003d1025124 for\nreference.\n\nChange-Id: I063f5c3a280262f1a1d70b44906a504e2b16134d\n(cherry picked from commit 24d88f4aec476cde10a18a2f24b50612c9e981ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/94f067d18d8f52a32058545db3756936a0f585fe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/94f067d18d8f52a32058545db3756936a0f585fe"}]},"branch":"refs/heads/stable/zed","description":"Edit commit message"},"aac820bf583668b08b7afaf9aa9b95f486cd241d":{"kind":"REWORK","_number":3,"created":"2022-12-13 16:40:09.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/09/867209/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/09/867209/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/09/867209/3"}}},"commit":{"parents":[{"commit":"69b014c5338e038c418e06984c63b3136910d499","subject":"Handle a different error code for missing TransferProtocolType","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/69b014c5338e038c418e06984c63b3136910d499"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-06 16:22:42.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-13 16:40:02.000000000","tz":60},"subject":"Handle proper code_status in unit test","message":"Handle proper code_status in unit test\n\nPython 3.11 correctly resolve the code_status with its number.\n\nSee https://bugs.debian.org/cgi-bin/bugreport.cgi?bug\u003d1025124 for\nreference.\n\nChange-Id: I063f5c3a280262f1a1d70b44906a504e2b16134d\n(cherry picked from commit 24d88f4aec476cde10a18a2f24b50612c9e981ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/aac820bf583668b08b7afaf9aa9b95f486cd241d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/aac820bf583668b08b7afaf9aa9b95f486cd241d"}]},"branch":"refs/heads/stable/zed"},"5e2807a79a105f4dda68c549c7abf7b02d4cb855":{"kind":"REWORK","_number":4,"created":"2022-12-13 17:04:03.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/09/867209/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/09/867209/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/09/867209/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/09/867209/4"}}},"commit":{"parents":[{"commit":"69b014c5338e038c418e06984c63b3136910d499","subject":"Handle a different error code for missing TransferProtocolType","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/69b014c5338e038c418e06984c63b3136910d499"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-06 16:22:42.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2022-12-13 17:04:03.000000000","tz":0},"subject":"Handle proper code_status in unit test","message":"Handle proper code_status in unit test\n\nPython 3.11 correctly resolve the code_status with its number.\n\nSee https://bugs.debian.org/cgi-bin/bugreport.cgi?bug\u003d1025124 for\nreference.\n\nChange-Id: I063f5c3a280262f1a1d70b44906a504e2b16134d\n(cherry picked from commit 24d88f4aec476cde10a18a2f24b50612c9e981ef)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5e2807a79a105f4dda68c549c7abf7b02d4cb855"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5e2807a79a105f4dda68c549c7abf7b02d4cb855"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}}]}],"submit_requirements":[]}
