)]}'
{"id":"openstack%2Fsushy~867639","triplet_id":"openstack%2Fsushy~stable%2Fzed~Ief49157438813c650fdd34d60dfe7ea8ea2d1702","project":"openstack/sushy","branch":"stable/zed","topic":"retry_on_idrac-sys518-stable/zed","attention_set":{},"removed_from_attention_set":{"27909":{"account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"last_update":"2023-01-05 09:46:37.000000000","reason":"Removed by \u003cGERRIT_ACCOUNT_27909\u003e by clicking the attention icon","reason_account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}},"33244":{"account":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"last_update":"2023-01-18 00:40:51.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ief49157438813c650fdd34d60dfe7ea8ea2d1702","subject":"Retry on iDRAC SYS518 errors for all requests","status":"MERGED","created":"2022-12-15 11:36:52.000000000","updated":"2023-01-18 00:42:34.000000000","submitted":"2023-01-18 00:40:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867639-retry_on_idrac-sys518-stable/zed","cherry_pick_of_change":864911,"cherry_pick_of_patch_set":4,"meta_rev_id":"92137e019d887ba73025a055c7ddb98037e68539","_number":867639,"virtual_id_number":867639,"owner":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-18 00:40:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2023-01-17 23:35:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2023-01-12 10:15:49.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2023-01-17 23:35:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}],"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-15 12:05:45.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-12-15 12:10:44.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-02 07:48:58.000000000","updated_by":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2023-01-12 10:15:49.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2023-01-12 14:28:20.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REMOVED"},{"updated":"2023-01-17 23:35:26.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"1eb677d7f6eb2c459db35c833e669f56454dce7a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2022-12-15 11:36:52.000000000","message":"Patch Set 1: Cherry Picked from branch master.\n\nThe following files contain Git conflicts:\n* sushy/connector.py","accounts_in_message":[],"_revision_number":1},{"id":"915b74ea53be75c51438ec24662f907cd32e2fb5","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-12-15 12:05:45.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/37710154d8534c5d8eaa9221740a8bd7 : FAILURE in 27m 16s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/763b3b6f9440440b96cc51e6d9ce6188 : FAILURE in 27m 33s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/42174dcb24c34d2a98dbeabe7ec405aa : FAILURE in 27m 33s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/45ab9f2490b143bb92dd40bc1688b79c : FAILURE in 27m 16s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":1},{"id":"9b4833e1beaf24ce3bbe7b0d95e42a7c2b704e95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-15 12:10:44.000000000","message":"Patch Set 1: Verified-1\n\n(17 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b0f77bfa0b184aeb82e47425e635cea7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cebcfcaf87f04250b11b0fffdc2ee9a8 : FAILURE in 2m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3efb5bd1946f469ea694fd067f0cc167 : FAILURE in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf99deb9dcad4e38b1cc2c5f02afc1dc : FAILURE in 3m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd6b54b607b549db896a4f22cb11c35f : FAILURE in 2m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2eeb125fac134a18b70f6b19479b28b2 : FAILURE in 3m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/065ebdd62bac429a920e5bf97e3e6ffd : FAILURE in 4m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20fbeb578ec84afea5b5bc3af092db38 : SUCCESS in 2m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/6add7b011d654c41aea2500d55ba1297 : FAILURE in 4m 47s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/28c824f04533434098907c3b65320103 : FAILURE in 21m 28s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3a0e466ea0064e069357eb4ceeee75c6 : FAILURE in 31m 02s","accounts_in_message":[],"_revision_number":1},{"id":"7540d39800361657f4e5c75ac3d775b2af11af7b","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-02 07:30:18.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"01c52c22ce4be98039b2c35d3a6441c9a3abf910","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-02 07:48:58.000000000","message":"Patch Set 2:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"dc00a3691f80b28cd4ac335cf23fecf0a7901e4e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 08:35:25.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/633520cc444f438988481135f4c1f87b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/332f82e76c4046ae972f93fa0c735134 : FAILURE in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/951693e5947747c3bd535b6c1351aa32 : SUCCESS in 2m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/39467dc3847347d5af7c947ff3b95b3b : FAILURE in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/10a390fb641c4180a2c0f3af0e1e1786 : FAILURE in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ebbe9856ec434d5e9af666f8822020d7 : FAILURE in 2m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1e2f999bf94647489608a00d84a35ac2 : SUCCESS in 4m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ba6ff94d0e8a4f3b837de1da9015911a : SUCCESS in 3m 00s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f280c1461b51444fb749bd4efc2be5c1 : FAILURE in 7m 06s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/228ad978d69e435da65cdf5c98c83552 : FAILURE in 43m 46s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/78425d0e4eeb45a8974ef9e93cccddf8 : SUCCESS in 59m 43s","accounts_in_message":[],"_revision_number":2},{"id":"61ec8295525e63ad759f80c4d064314eab0a4f27","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-02 12:59:53.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1f548279fc423b062a32ce0b663c186ec8b92222","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 14:26:21.000000000","message":"Patch Set 2:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/76cfbfb4fa8248de8b1363efa8db4444\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8906ddf1c5694543a02f08524b7ffabc : FAILURE in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a22ff8a8f8044363b46e56f5173f7bba : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/71b9b12e7e264c25a320baaf804b98d9 : FAILURE in 3m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/35522769511c497391d454f18f75b002 : FAILURE in 3m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98227ad77ec04120967ceaa5ce667e99 : FAILURE in 2m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/10597303152041d08c1ce45c773bddd3 : SUCCESS in 5m 17s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/874e031438594e2c836828df16aac4bd : SUCCESS in 3m 17s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/757a37696af844c08f5825e4f7430b90 : FAILURE in 7m 17s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/48b2d5909d49413294e969bd428b7ae1 : FAILURE in 46m 14s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4339721267f94809a5335ef1e150e7ae : FAILURE in 1h 24m 41s","accounts_in_message":[],"_revision_number":2},{"id":"deec09af7e1424770671cefde04284bc82bad3b3","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-12 08:35:11.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b04d150470f784fcc005d1b6916e06771b64ec88","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 09:36:09.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a4c289df62584de8a8163be2570fc4e8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a6913badfb6a4f52976b13e473f6891a : SUCCESS in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5afd87edde0c437a9ee4585fee1922cf : SUCCESS in 2m 40s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5625ac0b5ef648b89781211ad4af1835 : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e275926361194071851a3c07b4d1b168 : SUCCESS in 2m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1c12f017415f430aa7c04a842108a4b2 : SUCCESS in 3m 04s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3431951f4d734cc9a727d60de83ba691 : SUCCESS in 5m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5b244a2904c45ed918f08c59436f906 : SUCCESS in 3m 00s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9cc8113c2a0942339f32d59c2b1df7c1 : SUCCESS in 7m 42s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3643646c098e498f9bec0e11947c1951 : SUCCESS in 37m 01s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/7bf26a476aaa4346b358914170967a10 : SUCCESS in 54m 02s","accounts_in_message":[],"_revision_number":2},{"id":"734562a44d9c15e38d4ca2e74536e199d77b26fa","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-12 10:00:48.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"e0022314d6bf4946e39bb7a52bdf4447592765ed","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-01-12 10:15:49.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"a1a84ae10a46ab3c7994cc1da18aa0671fe689af","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2023-01-12 14:28:20.000000000","message":"Removed reviewer \u003cGERRIT_ACCOUNT_27909\u003e.","accounts_in_message":[{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"_revision_number":2},{"id":"26a394bb26fca49d091a85b3db0de7480644c375","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-01-17 23:35:26.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"af287a624059268b75eea96c93dfbe3975b96f3b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-17 23:36:15.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"fb51ea72fceb566b560b78f36e827d06e407ea2b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:40:51.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b47ab05ef9454a37a9bebb596598d892\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4ec086a4cce4fa0ae76fe4ed0352eb5 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6e0ebf6fced8426290383232d8e81aa9 : SUCCESS in 2m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3c2b358a780642278151b86de8d5a4d3 : SUCCESS in 3m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e205ec9176234769b6752045d67303d8 : SUCCESS in 4m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0ca2b465a57d4b24aef790de16d6f25b : SUCCESS in 3m 18s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/94e393f5a7144f4cae40a17b8d5e2489 : SUCCESS in 8m 00s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/193b07568e984b2ea3100c3ef0e24921 : SUCCESS in 55m 07s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5b6b0a1b8de140f7804c1808bc9afc71 : SUCCESS in 48m 29s","accounts_in_message":[],"_revision_number":2},{"id":"56309f8593f7a3aad9c4466d4be148f996228631","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:40:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"92137e019d887ba73025a055c7ddb98037e68539","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-18 00:42:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7da76800be34d8fb499f58396820b00\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3435c37a36d24cd1a1fda30c262da1ba : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/784f5375399640b2a2b98dc64274f568 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"e39c25fbb59df17b0bac043eb45640c4f272472b","revisions":{"333d90e4803c069b6f666bd5d2f7e5a1f0f9e628":{"kind":"REWORK","_number":1,"created":"2022-12-15 11:36:52.000000000","uploader":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"ref":"refs/changes/39/867639/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/39/867639/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/39/867639/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/39/867639/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/39/867639/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/39/867639/1"}}},"commit":{"parents":[{"commit":"69b014c5338e038c418e06984c63b3136910d499","subject":"Handle a different error code for missing TransferProtocolType","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/69b014c5338e038c418e06984c63b3136910d499"}]}],"author":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2022-11-17 15:06:29.000000000","tz":0},"committer":{"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2022-12-15 11:36:52.000000000","tz":0},"subject":"Retry on iDRAC SYS518 errors for all requests","message":"Retry on iDRAC SYS518 errors for all requests\n\nWhen iDRAC is still loading, some of the components is not available\nfor short period of time and returns HTTP 500 error with SYS518 error\ncode and message \"iDRAC is currently unable to display any information\nbecause data sources are unavailable.\". There is already retry on GET\nerrors with Server side error, but this can happen also for other\nrequests, for example, DELETE.\nThis patch adds retry for all SYS518 failures disregarding request\nmethod. This helps to fix a known intermittent issue when deleting\nset of volumes one after another and iDRAC is reloading after\ndeleting each volume.\n\nCo-Authored-By: Swapnil Machikar \u003cswapnil.machikar@yahoo.com\u003e\nChange-Id: Ief49157438813c650fdd34d60dfe7ea8ea2d1702\n(cherry picked from commit 2b760a6c9ef23ca2a9f736ca6a6c8c146490f1c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/333d90e4803c069b6f666bd5d2f7e5a1f0f9e628"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/333d90e4803c069b6f666bd5d2f7e5a1f0f9e628"}]},"branch":"refs/heads/stable/zed"},"e39c25fbb59df17b0bac043eb45640c4f272472b":{"kind":"REWORK","_number":2,"created":"2023-01-02 07:30:18.000000000","uploader":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"ref":"refs/changes/39/867639/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/39/867639/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/39/867639/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/39/867639/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/39/867639/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/39/867639/2"}}},"commit":{"parents":[{"commit":"69b014c5338e038c418e06984c63b3136910d499","subject":"Handle a different error code for missing TransferProtocolType","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/69b014c5338e038c418e06984c63b3136910d499"}]}],"author":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2022-11-17 15:06:29.000000000","tz":0},"committer":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2023-01-02 07:28:05.000000000","tz":0},"subject":"Retry on iDRAC SYS518 errors for all requests","message":"Retry on iDRAC SYS518 errors for all requests\n\nWhen iDRAC is still loading, some of the components is not available\nfor short period of time and returns HTTP 500 error with SYS518 error\ncode and message \"iDRAC is currently unable to display any information\nbecause data sources are unavailable.\". There is already retry on GET\nerrors with Server side error, but this can happen also for other\nrequests, for example, DELETE.\nThis patch adds retry for all SYS518 failures disregarding request\nmethod. This helps to fix a known intermittent issue when deleting\nset of volumes one after another and iDRAC is reloading after\ndeleting each volume.\n\nCo-Authored-By: Swapnil Machikar \u003cswapnil.machikar@yahoo.com\u003e\nChange-Id: Ief49157438813c650fdd34d60dfe7ea8ea2d1702\n(cherry picked from commit 2b760a6c9ef23ca2a9f736ca6a6c8c146490f1c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e39c25fbb59df17b0bac043eb45640c4f272472b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e39c25fbb59df17b0bac043eb45640c4f272472b"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}]}],"submit_requirements":[]}
