)]}'
{"id":"openstack%2Fsushy~867969","triplet_id":"openstack%2Fsushy~master~I3dbf80f3383c51d3e754516529471698851bf573","project":"openstack/sushy","branch":"master","topic":"handle_session_uri","attention_set":{},"removed_from_attention_set":{"18816":{"account":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"last_update":"2024-02-05 20:55:41.000000000","reason":"Change was submitted"},"24434":{"account":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"last_update":"2024-02-05 20:55:41.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-12-07 10:28:00.000000000","reason":"removed on reply"}},"hashtags":[],"change_id":"I3dbf80f3383c51d3e754516529471698851bf573","subject":"Handle session-uri in body","status":"MERGED","created":"2022-12-16 11:20:51.000000000","updated":"2024-02-05 20:56:38.000000000","submitted":"2024-02-05 20:55:41.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"867969-handle_session_uri","meta_rev_id":"80a290cdd0fd638a2ef86cf01ad56155fc5d7c82","_number":867969,"virtual_id_number":867969,"owner":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-05 20:55:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2024-02-05 20:00:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-07 10:28:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2024-02-05 20:00:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2022-12-16 12:21:04.000000000","updated_by":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"reviewer":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"state":"CC"},{"updated":"2022-12-16 12:22:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-08 11:49:38.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2023-05-17 06:05:12.000000000","updated_by":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"reviewer":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"state":"CC"},{"updated":"2023-12-07 10:28:00.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-02-05 20:00:16.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"7ce484795ea1f928c93196b017fbba25db5753b6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2022-12-16 11:20:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8ab065fb0389eeb654d3053e9bf53b11c28838ce","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2022-12-16 11:24:23.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.","accounts_in_message":[],"_revision_number":2},{"id":"c832c913e4ade31d990025f04afcdf8770a8862f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2022-12-16 11:25:37.000000000","message":"Uploaded patch set 3: Commit message was updated.","accounts_in_message":[],"_revision_number":3},{"id":"5ee16a4c69cc14f357c5db5c800fcf6046c2e888","author":{"_account_id":19339,"name":"Dell EMC Ironic CI","email":"openstack-ironic-ci@dell.com","username":"openstack-ironic-ci"},"date":"2022-12-16 12:21:04.000000000","message":"Patch Set 3:\n\nBuild succeeded (check pipeline).\n\n\n- 14G-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/e6770ccf48764beab726a3cb071fe6e0 : SUCCESS in 51m 33s\n- R6515-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/4eca22c0544b4c8e9890cb16030b1738 : SUCCESS in 54m 30s\n- R750-uefi-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/ac7ac0c72e7e4483bbc82a72ac31254e : SUCCESS in 52m 42s\n- XR11-bios-redfish https://ironic-3rd-party-ci.us.dell.com/t/openstack/build/f7057dbadbe342b3a9d5ec77276a5be5 : SUCCESS in 52m 15s\n\nFor 3rd party CI contact info: https://wiki.openstack.org/wiki/ThirdPartySystems/DellEMC_Ironic_CI","accounts_in_message":[],"_revision_number":3},{"id":"7498a22a654ac6393f85ec84e8c750dd1e8a7a12","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2022-12-16 12:22:53.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/87c96123b5584313bb83b25815231d62\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d162ed3298647548bd980ea8bcae946 : FAILURE in 2m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/47be5de6f346497da7eca8ce3fa9b885 : SUCCESS in 2m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4deffeb52f904946906f895e217ae675 : FAILURE in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ec976f1982f64e44950018eec8f059ff : FAILURE in 2m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7ead3ecb5f74540a48bb86a138b5016 : SUCCESS in 4m 37s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/9d787bf3814d4a2b84de9f9bdca4f347 : SUCCESS in 8m 20s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6cc6df8ea62d4f028b5a52caff88a49c : SUCCESS in 44m 26s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d12815fa8b144d00a7828f0630574772 : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":3},{"id":"44e3fba67f2c5c41d97a729d3a59334f09d830f5","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-08 11:49:38.000000000","message":"Patch Set 3:\n\nLooks like the tests need updating.","accounts_in_message":[],"_revision_number":3},{"id":"90bf4ecbfe21dc9ce3511fd0c84b4fc79f5ef7d9","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-05-17 06:05:12.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"e9ed7f4285703a4d9d5228b2def14d530e68280b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 06:57:28.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/37515eb7b0db44af9723a39837f2ac17\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fa6505c2c08e49c18643da1b1357d0b7 : FAILURE in 3m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/120de296816842ec887a9d7d4be1a39b : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8f325c9fc011498485cb7b015dc6edbd : FAILURE in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/365c10a24b1049569f79ef3f9f499887 : FAILURE in 3m 03s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/934c018964984c958660987ce75008a3 : FAILURE in 2m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/64cf7605bad44241af59613bc327dd2e : SUCCESS in 6m 02s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/40e5e66283134aec8087c0423ea12615 : SUCCESS in 9m 07s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/73ef2125ebf542b3807f16f901c7e33f : SUCCESS in 47m 01s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c5fd07f3d03346a7b443f1eb28279fe2 : SUCCESS in 39m 28s","accounts_in_message":[],"_revision_number":3},{"id":"d24b8968312aa3b424a6b2dad52c63a690ad903b","author":{"_account_id":18816,"name":"Maurice Escher","display_name":"carthaca","email":"maurice.escher@sap.com","username":"mapocace"},"date":"2023-05-17 07:33:30.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"c4929813a233f0b720eb8bdf27b009a937efa6e9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2023-05-17 11:40:30.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"dc925491b3328515c370a711f60bc0b08ccc4d86","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-17 12:28:28.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc992f464d734e9882c1373124e875a9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/68dffa9ecc96433e8409166a51302387 : SUCCESS in 3m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f352052a17ba43b5b1e1b264f45d4e6c : SUCCESS in 3m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebab3338d2f241fbaa8b2daae4dc3a87 : SUCCESS in 3m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/77b6e829b1e1492c925d586308dcc8d1 : SUCCESS in 3m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3bbf8b642ee74024ab90d92821854f25 : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8d084e2780de498c8846b82fdbe4a0cb : SUCCESS in 5m 25s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/762b8a34a34a43ed83f2b8841e06895e : SUCCESS in 7m 39s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/418b5cf99ee8404cb77ba8ab056ac09f : SUCCESS in 47m 13s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/de64ceef0d534117bd9a28c0e5c849b5 : SUCCESS in 41m 12s","accounts_in_message":[],"_revision_number":4},{"id":"572fe3540a3c084f65fac9cee7d6782b205b17e7","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2023-05-18 11:10:04.000000000","message":"Patch Set 4:\n\n(2 comments)","accounts_in_message":[],"_revision_number":4},{"id":"bcc61fbcd344eeea42baed682b5fe5bc33ec8787","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-05-19 08:11:41.000000000","message":"Patch Set 4:\n\nThanks! Could you also add a new unit test to exercise the change?","accounts_in_message":[],"_revision_number":4},{"id":"92b72142cfd33d33b214a82481b190a133397f85","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2023-05-19 09:45:09.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"c2b8f5bc060303f2aee705a12251a404c436093c","author":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"date":"2023-05-19 09:45:52.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"1689896094e34b73c6063803be99e4ce64885e2b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-19 10:35:13.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58dd70f893e1444eabe04692df505aec\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/69d7328c1ce9425087374b3163e3c9e8 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c90d3651dc584652aa15d0b1ddbf7565 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ae64b0cb0cd54c7f98e4a67ba5f4eeb8 : SUCCESS in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/443f5aa56afe40efb1025372fb7760d4 : SUCCESS in 3m 01s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6cd55890804c40ecb6ddee236c0bdac8 : SUCCESS in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a4a97a8eaee245c09708c7f14cd370c0 : SUCCESS in 5m 47s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8920882d61fd4b0caf90223fe8bf74c4 : SUCCESS in 8m 41s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7bf848ecc60743bc9e1d1cfe2bf2d415 : SUCCESS in 44m 25s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b01f5aa2973a4b4d9c7172226b090614 : SUCCESS in 48m 49s","accounts_in_message":[],"_revision_number":5},{"id":"a217f3a0a30ffc9f06052f3c149732d524837cdb","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-12-07 10:28:00.000000000","message":"Patch Set 5: Code-Review+2\n\n(1 comment)\n\n1 comment inline + a release note would be really appreciated.","accounts_in_message":[],"_revision_number":5},{"id":"4d90b89104010445a9c31f743c6090e244a4571b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-02-05 20:00:16.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"3ea0f81e0e71ecdc5f98facdafa67972503271a7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:00:50.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"44697ea47b7aa248b5fccb65da78c616d320a4bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:55:40.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6d422debc0ea4c8caf00df2c05c257f6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28f2c4bbfd5544caa7ebbaf4ade40e08 : SUCCESS in 3m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5ed1db0636b84a8c9b576c39535bc0c2 : SUCCESS in 2m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/640be30c70b44115926d43896c12ee23 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d39521239e8c442fb728c03b621df1b0 : SUCCESS in 4m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e4af8e817ebc41418a87ab52cdfea3f7 : SUCCESS in 8m 45s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c7349d3c6a9d47b7b6077b42bbb05446 : SUCCESS in 35m 42s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/44217c4943a54aa38f3691a9eb25416e : SUCCESS in 53m 33s","accounts_in_message":[],"_revision_number":5},{"id":"6f36a10879b2e41ae6b9fc387c786219b237f6ee","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:55:41.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"80a290cdd0fd638a2ef86cf01ad56155fc5d7c82","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 20:56:38.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c0c390f77f9e475e9caf7ca8dbb8077f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d1f104a53c874691942cf46348e59f78 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"43ed6a683dedbc22a54b0b6dd66a8ddcd6fc97b2","revisions":{"4ab54ab81434b3ceaba5fdaf093dd8e7212ca72b":{"kind":"REWORK","_number":1,"created":"2022-12-16 11:20:51.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/69/867969/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/69/867969/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/69/867969/1"}}},"commit":{"parents":[{"commit":"9427360ccbe0cfc1340ecd2e79bbc77663cdeff4","subject":"Merge \"Handle proper code_status in unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9427360ccbe0cfc1340ecd2e79bbc77663cdeff4"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:20:32.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:20:49.000000000","tz":60},"subject":"Handle session-uri in body","message":"Handle session-uri in body\n\nThe Cisco CIMC 4.1(2b) and potentially others do not\nset the Location header as required by the Redfish\nstandard on a successful login, but returns a session\nobject as a body.\n\nBy taking the session-uri out of the body,\nwe avoid leaking sessions in that situation.\n\nChange-Id: I3dbf80f3383c51d3e754516529471698851bf573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4ab54ab81434b3ceaba5fdaf093dd8e7212ca72b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4ab54ab81434b3ceaba5fdaf093dd8e7212ca72b"}]},"branch":"refs/heads/master"},"90d86c868bf615157959d4f86a83b8615d360d5f":{"kind":"NO_CHANGE","_number":2,"created":"2022-12-16 11:24:23.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/69/867969/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/69/867969/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/69/867969/2"}}},"commit":{"parents":[{"commit":"9427360ccbe0cfc1340ecd2e79bbc77663cdeff4","subject":"Merge \"Handle proper code_status in unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9427360ccbe0cfc1340ecd2e79bbc77663cdeff4"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:20:32.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:24:07.000000000","tz":60},"subject":"Handle session-uri in body","message":"Handle session-uri in body\n\nThe Cisco CIMC 4.1(2b) and potentially others do not\nset the Location header as required by the Redfish\nstandard on a successful login, but returns a session\nobject as a body.\n\nBy taking the session-uri out of the body,\nwe avoid leaking sessions in that situation.\n\nChange-Id: I3dbf80f3383c51d3e754516529471698851bf573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/90d86c868bf615157959d4f86a83b8615d360d5f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/90d86c868bf615157959d4f86a83b8615d360d5f"}]},"branch":"refs/heads/master"},"87a940697960ac1813455e09691a734f0f9326c2":{"kind":"NO_CODE_CHANGE","_number":3,"created":"2022-12-16 11:25:37.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/69/867969/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/69/867969/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/69/867969/3"}}},"commit":{"parents":[{"commit":"9427360ccbe0cfc1340ecd2e79bbc77663cdeff4","subject":"Merge \"Handle proper code_status in unit test\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9427360ccbe0cfc1340ecd2e79bbc77663cdeff4"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:20:32.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:24:57.000000000","tz":60},"subject":"Handle session-uri in body","message":"Handle session-uri in body\n\nThe Cisco CIMC 4.1(2b) and potentially others do not\nset the Location header as required by the Redfish\nstandard on a successful login, but return a session\nobject as a body.\n\nBy taking the session-uri out of the body,\nwe avoid leaking sessions in that situation.\n\nChange-Id: I3dbf80f3383c51d3e754516529471698851bf573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/87a940697960ac1813455e09691a734f0f9326c2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/87a940697960ac1813455e09691a734f0f9326c2"}]},"branch":"refs/heads/master"},"7e01d872a7e9c31239174e67249f5c9e6e047d45":{"kind":"REWORK","_number":4,"created":"2023-05-17 11:40:30.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/69/867969/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/69/867969/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/69/867969/4"}}},"commit":{"parents":[{"commit":"f03d77ebcb7199fa08a82d5b34d7a6d677d86e0f","subject":"Merge \"Retry on ilo state error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f03d77ebcb7199fa08a82d5b34d7a6d677d86e0f"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:20:32.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2023-05-17 11:40:04.000000000","tz":120},"subject":"Handle session-uri in body","message":"Handle session-uri in body\n\nThe Cisco CIMC 4.1(2b) and potentially others do not\nset the Location header as required by the Redfish\nstandard on a successful login, but return a session\nobject as a body.\n\nBy taking the session-uri out of the body,\nwe avoid leaking sessions in that situation.\n\nStory: 2010477\nTask: 47038\nChange-Id: I3dbf80f3383c51d3e754516529471698851bf573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7e01d872a7e9c31239174e67249f5c9e6e047d45"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/7e01d872a7e9c31239174e67249f5c9e6e047d45"}]},"branch":"refs/heads/master"},"43ed6a683dedbc22a54b0b6dd66a8ddcd6fc97b2":{"kind":"REWORK","_number":5,"created":"2023-05-19 09:45:09.000000000","uploader":{"_account_id":24434,"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","username":"fwiesel"},"ref":"refs/changes/69/867969/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/69/867969/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/69/867969/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/69/867969/5"}}},"commit":{"parents":[{"commit":"f03d77ebcb7199fa08a82d5b34d7a6d677d86e0f","subject":"Merge \"Retry on ilo state error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f03d77ebcb7199fa08a82d5b34d7a6d677d86e0f"}]}],"author":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2022-12-16 11:20:32.000000000","tz":60},"committer":{"name":"Fabian Wiesel","email":"fabian.wiesel@sap.com","date":"2023-05-19 09:44:52.000000000","tz":120},"subject":"Handle session-uri in body","message":"Handle session-uri in body\n\nThe Cisco CIMC 4.1(2b) and potentially others do not\nset the Location header as required by the Redfish\nstandard on a successful login, but return a session\nobject as a body.\n\nBy taking the session-uri out of the body,\nwe avoid leaking sessions in that situation.\n\nStory: 2010477\nTask: 47038\nChange-Id: I3dbf80f3383c51d3e754516529471698851bf573\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/43ed6a683dedbc22a54b0b6dd66a8ddcd6fc97b2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/43ed6a683dedbc22a54b0b6dd66a8ddcd6fc97b2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
