)]}'
{"id":"openstack%2Fsushy~868778","triplet_id":"openstack%2Fsushy~stable%2Fxena~Ief49157438813c650fdd34d60dfe7ea8ea2d1702","project":"openstack/sushy","branch":"stable/xena","topic":"retry_on_idrac-sys518-stable/zed-stable/yoga-stable/xena","attention_set":{},"removed_from_attention_set":{"27909":{"account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"last_update":"2023-01-04 08:16:46.000000000","reason":"removed on reply"},"33244":{"account":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"last_update":"2023-01-12 11:25:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ief49157438813c650fdd34d60dfe7ea8ea2d1702","subject":"Retry on iDRAC SYS518 errors for all requests","status":"MERGED","created":"2023-01-02 07:44:17.000000000","updated":"2023-01-12 11:26:46.000000000","submitted":"2023-01-12 11:25:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"868778-retry_on_idrac-sys518-stable/zed-stable/yoga-stable/xena","cherry_pick_of_change":868777,"cherry_pick_of_patch_set":1,"meta_rev_id":"5fdbf38dca82a7bfea36f0b3f9607543d3062b66","_number":868778,"virtual_id_number":868778,"owner":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-12 11:25:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"all":[{"value":2,"date":"2023-01-04 08:16:46.000000000","_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":2,"date":"2023-01-12 10:15:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":1,"date":"2023-01-12 10:15:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-02 07:49:51.000000000","updated_by":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REVIEWER"},{"updated":"2023-01-02 08:47:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-12 10:15:38.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"32de209e48e93b8f309038f98597df60b5998c72","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-02 07:44:17.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"9fa253ca8e1b11aa1b00a40d83b6a87f58406947","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-02 07:49:51.000000000","message":"Patch Set 1:\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":1},{"id":"3a5bdbcf9df6e4500a164c4814756f97081d5fb7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 08:47:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f9c97e930df648e8af92dcfd3e0c7ef7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eeb1182bf6634affa8e341b5cbdab660 : FAILURE in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9aae6a3c7f14bf5b4323ac8b7459e1d : SUCCESS in 2m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/9e9572d9fae0460e96aa9a072b0e9950 : SUCCESS in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c0d5803ca644b4f8145d777c89c697f : FAILURE in 2m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/47a2d62cae1b421c8141655ebe97b0d5 : FAILURE in 2m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f838c89574a4050b18aae9816541b69 : SUCCESS in 4m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/225da6fd07444ce3884fb999ece5d252 : SUCCESS in 2m 44s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e587473cf6b44ecd9c290dda52019f0a : SUCCESS in 38m 20s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1f7d2d50465d4dcd9404dd4fdb6ff60a : SUCCESS in 57m 52s","accounts_in_message":[],"_revision_number":1},{"id":"f04d4535055f089fceff677c1413549734063b48","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-02 13:00:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c35d1bff1d32652862f45ddb69065b48f9ae29ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-02 13:42:05.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c25a61fda1244deab7cc46982c8adc4d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2ca0ff5ac4c741da95d7b3fecc44f7b2 : FAILURE in 2m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbfa23825c124d5cad534c6b23b20255 : SUCCESS in 2m 01s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/bed372265b5b416e841ab8909b28b3ec : SUCCESS in 3m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/40b4e27ea11e4623a949c10cbbdf9504 : FAILURE in 2m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fc1fc0f173b64bf398bcd613d8cebd90 : FAILURE in 2m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e3ec9adeed8d432aafd0f96cd32f90d2 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8c1cf41e823045b589220db903a619ad : SUCCESS in 2m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ff544bc853a44fbc9ad5628e7fa58c91 : SUCCESS in 32m 30s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e874f49a49c4423fb4266a96bb7456dd : SUCCESS in 39m 37s","accounts_in_message":[],"_revision_number":1},{"id":"06c82d59441f817c437366d8e4f032b4fe02f036","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-03 15:50:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"49419321f1d78679e4c659c892e4f83d1c3cefc5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-03 16:42:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49ee82935fb74e27ae1a895643dff8db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/32134fc7ca0d4909bf91501e213f6b3e : SUCCESS in 3m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1cf557ec79114859a907e79f79c1ab42 : SUCCESS in 3m 13s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c3d0006fed2942ba82a54d64d009fd87 : SUCCESS in 3m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc390b3af53746818650bb2b4860b2a6 : SUCCESS in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60188f5b8f68449989393eab58abf679 : SUCCESS in 3m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/448e54e70c04483ab0adeb11895d5799 : SUCCESS in 5m 56s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/94dc5c4e232b427a8e53f18cb29d5c2b : SUCCESS in 3m 40s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/7f416674dd5a4196a64b20b6cd3459ba : SUCCESS in 42m 34s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1dc2ef5feea748229781963e5218ec89 : SUCCESS in 50m 40s","accounts_in_message":[],"_revision_number":1},{"id":"5cf1957ce20aca06deeb13343bd845ae26d03b96","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2023-01-04 08:16:46.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"052fb106fd2c5a303f3f7ecac8c95fb6a7126def","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-01-12 10:15:38.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0b91db95a9d985ae26cb85f88ce546ebf8953e76","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 10:16:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"009c0f8453bd6d6b472a2d29707a9f88ef224218","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:25:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bed53c190b314ee1b3ab2df9bbfa49ea\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bf516e6746c54f1d91577a667e5d6d30 : SUCCESS in 3m 28s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/180856ce3f6e43209b3276a9a883e39f : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab574fe0d3614ad7a37e470872f85c85 : SUCCESS in 3m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ebb85e68356347b6ad71738dc54f5900 : SUCCESS in 5m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0e7a12c4c3894e8cb11f0ddd9d77d2cb : SUCCESS in 3m 08s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b3bf9dd1c9744c51ab370d1bfc20048a : SUCCESS in 1h 01m 03s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/afa1adb96617485a92ec3bdab28afa9a : SUCCESS in 1h 04m 16s","accounts_in_message":[],"_revision_number":1},{"id":"9a7e134818c744d15bad8dcf8f0e0c2d5c5540f6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:25:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5fdbf38dca82a7bfea36f0b3f9607543d3062b66","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 11:26:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/122d4ab792aa44eb83f28a409c6bc77e\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9cd547a60074c3ca0c0e37dac118d0a : SUCCESS in 38s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a6515fc15a3427c8d74ea6664cbe8d3 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"be95c7c45ae1322d132beffc1651200a10b97685","revisions":{"be95c7c45ae1322d132beffc1651200a10b97685":{"kind":"REWORK","_number":1,"created":"2023-01-02 07:44:17.000000000","uploader":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"ref":"refs/changes/78/868778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/78/868778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/78/868778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/78/868778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/78/868778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/78/868778/1"}}},"commit":{"parents":[{"commit":"a20ecddadcd6afccb1697ddb9a830631cb6a1879","subject":"Handle a different error code for missing TransferProtocolType","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a20ecddadcd6afccb1697ddb9a830631cb6a1879"}]}],"author":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2022-11-17 15:06:29.000000000","tz":0},"committer":{"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2023-01-02 07:44:17.000000000","tz":0},"subject":"Retry on iDRAC SYS518 errors for all requests","message":"Retry on iDRAC SYS518 errors for all requests\n\nWhen iDRAC is still loading, some of the components is not available\nfor short period of time and returns HTTP 500 error with SYS518 error\ncode and message \"iDRAC is currently unable to display any information\nbecause data sources are unavailable.\". There is already retry on GET\nerrors with Server side error, but this can happen also for other\nrequests, for example, DELETE.\nThis patch adds retry for all SYS518 failures disregarding request\nmethod. This helps to fix a known intermittent issue when deleting\nset of volumes one after another and iDRAC is reloading after\ndeleting each volume.\n\nCo-Authored-By: Swapnil Machikar \u003cswapnil.machikar@yahoo.com\u003e\nChange-Id: Ief49157438813c650fdd34d60dfe7ea8ea2d1702\n(cherry picked from commit 2b760a6c9ef23ca2a9f736ca6a6c8c146490f1c3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/be95c7c45ae1322d132beffc1651200a10b97685"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/be95c7c45ae1322d132beffc1651200a10b97685"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
