)]}'
{"id":"openstack%2Fsushy~869313","triplet_id":"openstack%2Fsushy~stable%2Fxena~Ia421ff025f1f0bc72252b5441589beb2e53cc25a","project":"openstack/sushy","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"27909":{"account":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"last_update":"2023-01-12 14:27:57.000000000","reason":"Reviewer/Cc was removed"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-01-16 13:31:36.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},"33244":{"account":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"last_update":"2023-01-24 10:04:37.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Ia421ff025f1f0bc72252b5441589beb2e53cc25a","subject":"Fix volume deletion on newer iDRACs","status":"MERGED","created":"2023-01-05 08:01:52.000000000","updated":"2023-01-24 10:05:42.000000000","submitted":"2023-01-24 10:04:37.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"869313","cherry_pick_of_change":869313,"cherry_pick_of_patch_set":1,"meta_rev_id":"2376cce5982a5f2bf7f0bd843ea0cf9a3d14c55b","_number":869313,"virtual_id_number":869313,"owner":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-01-24 10:04:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2023-01-17 23:31:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2023-01-16 13:31:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2023-01-24 08:53:53.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-05 09:00:48.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-05 09:05:08.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2023-01-05 12:45:43.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"CC"},{"updated":"2023-01-12 11:48:32.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-01-12 14:27:57.000000000","updated_by":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"reviewer":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"state":"REMOVED"},{"updated":"2023-01-17 23:31:02.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"7ae992804071f4f717cd4ec9c69f1228f7de7afe","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-05 08:01:52.000000000","message":"Patch Set 1: Cherry Picked from branch stable/yoga.","accounts_in_message":[],"_revision_number":1},{"id":"39456330750d87e5b41cfec272dc8933238eb447","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-05 08:02:09.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"0869523d1e2791768f9bbe810a9bf8b596063a26","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-05 09:00:48.000000000","message":"Patch Set 2: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9c99a80a4865491fafaf5802c5d8df1a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/703b51e03be447468ade7db7413ff137 : FAILURE in 2m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e78b3d10fd14a5c9148a63187658a3a : FAILURE in 2m 46s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/86f4c8cccb194c85aa3e078163306dfe : FAILURE in 3m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/872f0ccc25dc4d4ba971a1f05b51b5d6 : FAILURE in 2m 59s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e638389ddb664e8d901788a917e43af1 : FAILURE in 3m 07s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fb91ba8d2d54b3383ebfef7c8e29dbb : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5854fa5fdc444efdbff157b1ed9b4854 : SUCCESS in 2m 59s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c8f2758aa451457f948daa42b92887a9 : SUCCESS in 53m 11s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/427ea233e7af44aaa961c89b0012a713 : SUCCESS in 50m 36s","accounts_in_message":[],"_revision_number":2},{"id":"6ac249dcade0a777a038ed3d55a69d3f13f9b19d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-01-05 09:05:08.000000000","message":"Patch Set 2: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d9687ddbdf7ea7814e2e034f5292222e093b72f8","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2023-01-05 12:45:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"076647017c1946d710b688e11d835b8a7071deb7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-12 11:32:17.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"8b7bea378163de7587d99694a9bda666c1e4cdd1","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-01-12 11:48:32.000000000","message":"Patch Set 3:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/13/869313/3/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/df565c0/df565c0.html : FAILURE in 13m 51s","accounts_in_message":[],"_revision_number":3},{"id":"8d1b0a7e93b98212fc5b76a966469e3c53165fa9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-12 12:30:52.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58ea0441e66a4d398998dbdfe3a88abc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b20b60004a0a4690b2de287563d6d569 : SUCCESS in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/badf74589d08430eaa4a0adf3ef9ea57 : SUCCESS in 2m 56s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/c7dec05871fc486faa156aba7b80576b : SUCCESS in 3m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ebfe1e300859406dac396e23b92e30fc : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4393dc82424a422897a8872cdd9f8707 : SUCCESS in 2m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/729f1378f35640139a6d26b2a3ba4969 : SUCCESS in 5m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a59ada724d154b759675d7534e1afd5a : SUCCESS in 3m 45s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b2c84d91e70b457ab41fac673a7e335e : SUCCESS in 36m 25s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ebef00c7a6774baca228f42afa03c222 : SUCCESS in 53m 30s","accounts_in_message":[],"_revision_number":3},{"id":"18cc91cb646479a80cf225aa0a56e9c0356bc5d4","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-12 12:42:53.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"33f392d274fca543f0fcd9afa06b81d6f515454d","tag":"autogenerated:gerrit:deleteReviewer","author":{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"},"date":"2023-01-12 14:27:57.000000000","message":"Removed cc \u003cGERRIT_ACCOUNT_27909\u003e.","accounts_in_message":[{"_account_id":27909,"name":"Aija Jauntēva","email":"code@clusums.eu","username":"ajya"}],"_revision_number":3},{"id":"8d948f2bde0cc9ff9f4cc93a9b38b9046dd729fe","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-01-16 13:31:36.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"f21e44803818bd9df837e217aba5545275bdefc7","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-17 09:08:41.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":3},{"id":"bee28f5578947b1e6db6a58b26d317bf0b419469","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-01-17 23:31:02.000000000","message":"Patch Set 3: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"4d7027a9289925bc4677c1c12a21aac603c6457e","author":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"date":"2023-01-23 07:20:16.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"cb73a9698580ce9d0712fbe6538148af0f088f4d","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-01-24 08:53:53.000000000","message":"Patch Set 3: Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"3a634debb5e2b36fc634c186523a104b46782e92","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 08:54:33.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"9e4d138082a62bdedd009bc5aefa91b293aeadd5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:04:37.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ed515fa4eeb34cd9a42e6305c343628b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb3e3b129df84a51bec5ada0d2e6495d : SUCCESS in 2m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24744ca52be140488eb250362ad4d8c9 : SUCCESS in 2m 31s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d0448e4419f5492aad5f377614b6fff5 : SUCCESS in 3m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb6f508a28ab4b318d640cf5497e5d8f : SUCCESS in 4m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0f94514031f4b42933a61bab13d858b : SUCCESS in 3m 42s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/07eef2fd630d473e860f82fae48c9edf : SUCCESS in 39m 41s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b36305a8257a4049baa6e760e4d1a7ea : SUCCESS in 1h 04m 56s","accounts_in_message":[],"_revision_number":3},{"id":"be7d4221d5cd5ae33d0ce857722ed1ad438843c8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:04:37.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"2376cce5982a5f2bf7f0bd843ea0cf9a3d14c55b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-24 10:05:42.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7ac5aab3409b47c4aeac3501fe366c2d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5672d9c8ff6742429522d549ddd4ebed : SUCCESS in 42s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23ec706ebad14a0aa7b86dbe995ad046 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"dc384630e6ad095a6c9723ff9fc7af5d187eb4dd","revisions":{"335f9a9bcd0a254bdac7e2f37a70740f4c377ec0":{"kind":"REWORK","_number":1,"created":"2023-01-05 08:01:52.000000000","uploader":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"ref":"refs/changes/13/869313/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/13/869313/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/13/869313/1"}}},"commit":{"parents":[{"commit":"3765efc00c7979a622da169b8b44b34edfdb9c4c","subject":"Fix tox4 and setuptools errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3765efc00c7979a622da169b8b44b34edfdb9c4c"}]}],"author":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2022-11-17 07:46:29.000000000","tz":0},"committer":{"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2023-01-05 08:01:52.000000000","tz":0},"subject":"Fix volume deletion on newer iDRACs","message":"Fix volume deletion on newer iDRACs\n\nFixes \u0027Unsupported parameter name @Redfish.OperationApplyTime\u0027 General\nerror on iDRAC firmware version 6.00.02.00 or newer when deleting\nvolumes.\nHandles the failure of volume.delete by retrying without operation\napply time. In older iDRAC firmware versions operation apply time was\nsilently ignored, but now in newer iDRAC firmware versions it is\nreturning 501 Not Implemented error.\n\nChange-Id: Ia421ff025f1f0bc72252b5441589beb2e53cc25a\n(cherry picked from commit 74f95b1dd511cbbf36b0e85a596212d96047e345)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/335f9a9bcd0a254bdac7e2f37a70740f4c377ec0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/335f9a9bcd0a254bdac7e2f37a70740f4c377ec0"}]},"branch":"refs/heads/stable/xena"},"60065593e78f6b4dde1545cf401cdaf09976086f":{"kind":"NO_CHANGE","_number":2,"created":"2023-01-05 08:02:09.000000000","uploader":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"ref":"refs/changes/13/869313/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/13/869313/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/13/869313/2"}}},"commit":{"parents":[{"commit":"3765efc00c7979a622da169b8b44b34edfdb9c4c","subject":"Fix tox4 and setuptools errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3765efc00c7979a622da169b8b44b34edfdb9c4c"}]}],"author":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2022-11-17 07:46:29.000000000","tz":0},"committer":{"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2023-01-05 08:02:09.000000000","tz":0},"subject":"Fix volume deletion on newer iDRACs","message":"Fix volume deletion on newer iDRACs\n\nFixes \u0027Unsupported parameter name @Redfish.OperationApplyTime\u0027 General\nerror on iDRAC firmware version 6.00.02.00 or newer when deleting\nvolumes.\nHandles the failure of volume.delete by retrying without operation\napply time. In older iDRAC firmware versions operation apply time was\nsilently ignored, but now in newer iDRAC firmware versions it is\nreturning 501 Not Implemented error.\n\nChange-Id: Ia421ff025f1f0bc72252b5441589beb2e53cc25a\n(cherry picked from commit 74f95b1dd511cbbf36b0e85a596212d96047e345)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/60065593e78f6b4dde1545cf401cdaf09976086f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/60065593e78f6b4dde1545cf401cdaf09976086f"}]},"branch":"refs/heads/stable/xena"},"dc384630e6ad095a6c9723ff9fc7af5d187eb4dd":{"kind":"REWORK","_number":3,"created":"2023-01-12 11:32:17.000000000","uploader":{"_account_id":33244,"name":"kamlesh chauvhan","email":"kamlesh.chauvhan@gmail.com","username":"kamlesh.chauvhan"},"ref":"refs/changes/13/869313/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/13/869313/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/869313/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/13/869313/3"}}},"commit":{"parents":[{"commit":"3765efc00c7979a622da169b8b44b34edfdb9c4c","subject":"Fix tox4 and setuptools errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3765efc00c7979a622da169b8b44b34edfdb9c4c"}]}],"author":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2022-11-17 07:46:29.000000000","tz":0},"committer":{"name":"Kamlesh Chauvhan","email":"kamlesh.chauvhan@gmail.com","date":"2023-01-12 11:27:36.000000000","tz":0},"subject":"Fix volume deletion on newer iDRACs","message":"Fix volume deletion on newer iDRACs\n\nFixes \u0027Unsupported parameter name @Redfish.OperationApplyTime\u0027 General\nerror on iDRAC firmware version 6.00.02.00 or newer when deleting\nvolumes.\nHandles the failure of volume.delete by retrying without operation\napply time. In older iDRAC firmware versions operation apply time was\nsilently ignored, but now in newer iDRAC firmware versions it is\nreturning 501 Not Implemented error.\n\nChange-Id: Ia421ff025f1f0bc72252b5441589beb2e53cc25a\n(cherry picked from commit 74f95b1dd511cbbf36b0e85a596212d96047e345)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc384630e6ad095a6c9723ff9fc7af5d187eb4dd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/dc384630e6ad095a6c9723ff9fc7af5d187eb4dd"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[]}
