)]}'
{"id":"openstack%2Fsushy~872049","triplet_id":"openstack%2Fsushy~master~I0c4690925d086af593a691565a66ce90c19be156","project":"openstack/sushy","branch":"master","topic":"zt-hardware","attention_set":{},"removed_from_attention_set":{"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2023-03-08 21:54:14.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-02-28 15:44:40.000000000","reason":"removed on reply"}},"hashtags":["ironic-week-prio"],"change_id":"I0c4690925d086af593a691565a66ce90c19be156","subject":"Handle non-default language for registries","status":"MERGED","created":"2023-01-27 20:43:43.000000000","updated":"2023-03-08 21:55:11.000000000","submitted":"2023-03-08 21:54:14.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":12,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"872049-zt-hardware","meta_rev_id":"c7f88ee8fd2cd885030e01fc09ca85ae378f1976","_number":872049,"virtual_id_number":872049,"owner":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-08 21:54:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-03-08 20:51:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-01 12:01:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2023-03-08 20:51:32.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"}],"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-01-27 21:37:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-01-27 22:28:33.000000000","updated_by":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"reviewer":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"state":"CC"},{"updated":"2023-02-15 18:02:17.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"CC"},{"updated":"2023-02-27 11:37:42.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2023-03-08 20:51:32.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"a931b73138128164696a2c5cd9d3debaa54904a1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-01-27 20:43:43.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4b2751ce73e01bbd4c1a0c7da2c983930a1290e8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-01-27 21:37:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a37547b5614848078c4dcfc80c0022b2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cbdaf3b83ac4a36a134e5b582d32463 : SUCCESS in 4m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56c630c9658a476fb0b3627b77678863 : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e31341d30cf3499b968dec859a3b31ba : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c16be48620154355a201f3b688b67b18 : SUCCESS in 2m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6685c636cc8e49e9b2ddec4966c7408e : SUCCESS in 6m 04s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/04a6e92687544873b823b13620a0786a : SUCCESS in 12m 04s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9e97d5375a4b474b8711bdb7996f1baf : SUCCESS in 39m 55s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1c237e1b9b88469d9560f6dc76211494 : SUCCESS in 52m 39s","accounts_in_message":[],"_revision_number":1},{"id":"35ca83271ffc1f981b86408fd740befce21dd35d","author":{"_account_id":14629,"name":"HPE Proliant","email":"proliantutils@gmail.com","username":"proliantci"},"date":"2023-01-27 22:28:33.000000000","message":"Patch Set 1:\n\nBuild failed.\n\n- hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3 https://hpproliant.github.io/hpeproliant.github.io/logs/49/872049/1/check/hpe-gen10-dsvm-sushy-standalone-wholedisk-uefi-redfish-virtual-media-direct-python3/5ee008c/5ee008c.html : FAILURE in 45m 44s","accounts_in_message":[],"_revision_number":1},{"id":"980b6d4bbb95891372bc7e58a9e6dc312a9f2cbd","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-02-15 18:02:17.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82f57779fefbefc70303c12ab4d0eadbbbace504","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-02-27 03:20:22.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"4513a74659870e5bc7b4951661d070256f44d598","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-02-27 03:20:44.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":2},{"id":"9d164d1b1aa7518b6b0b872bba32c0c23b03b263","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-27 03:49:09.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5bec420f540b4791b416d33546aa5bb0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52d48de76bce489489364d62bd10bc54 : SUCCESS in 3m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2adeaccb4c7c47c18343a9a25c6fe4f8 : SUCCESS in 2m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/af0d75c3ba5e428f850d5a5e7c5ed0f8 : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0fa8e8c19e0e4a00b9bf3a402bbc24fc : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e519dca88d9b421cbf65ddfccf0b2d4a : SUCCESS in 6m 18s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b465ad3e71a749a896c5538b288f38fc : RETRY_LIMIT in 1m 58s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c04596b840c14ad8af4e3940eebcc42e : SUCCESS in 8m 41s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/72d48945c1694c0bb557dcf965e373e6 : FAILURE in 5m 44s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/283ff50fdf844c959c3117528a55e1e4 : FAILURE in 5m 38s","accounts_in_message":[],"_revision_number":2},{"id":"bf321142864c49a4c3ed4dc8b601bfe6268d886d","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-02-27 11:37:42.000000000","message":"Patch Set 2: Code-Review-1\n\n(3 comments)","accounts_in_message":[],"_revision_number":2},{"id":"f7e6e51cea66923d8481e9e70e797c181fb625b7","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-02-27 11:45:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"590e1e8c947675c82e312cfecbd1bd702cc646a6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-02-28 00:06:14.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"388b82102f4e4529ec5294d7f0c8564d65075e53","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-02-28 00:07:36.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a0efa6070e5be4dc2e58b091fb086f17755d29f4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 00:44:32.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f60b257f512f42a28587ebc0f9db71d5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/00f44453209040c98db8bbf7e9516b03 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9aa84b7e932e4ca7b743075d3e9a0676 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e75be9c99469491693ae82107ef7cd20 : SUCCESS in 3m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9c62eeda1dab4904b86f1ca6bd4fcafb : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dfcfb2fde7134c869238eaf7d89f1766 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cc2fcaaaa44f43e7bbe027ec54d7a1b9 : SUCCESS in 3m 45s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c3178d6acba24024be13051e31ee2b90 : SUCCESS in 6m 45s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/dc0ada7e22c94fb5ae03ffc26265eee5 : SUCCESS in 34m 34s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6f1303d27b694287bd83963b26cc9ef7 : SUCCESS in 36m 13s","accounts_in_message":[],"_revision_number":3},{"id":"914c776e43d14a77f48407d2ba48b44ca1a946eb","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-02-28 15:44:40.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"1bb1631de60c87c81bab15cccbb117e78e236b2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-02-28 15:59:10.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"798b11e0ab7ac4a143a8229d5db97de51721e3b8","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-02-28 15:59:23.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"894387a4ddbd0abe51b668601d5d26990cccae92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-02-28 17:11:54.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c1d0200f0b33404c9186c85df5d9b108\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b559b425d42c4a24b2a4429913415d6a : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/081788c4ac604082a4289aaa28a0e0ab : SUCCESS in 2m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26039c9a367048c599709ac8e57cef3e : SUCCESS in 3m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7a537cf4183d458aafcb7e453509b123 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb732062cff24587aadbf6db4f117ff6 : SUCCESS in 5m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7f93a738d8e4993bfdfa4a21bf0feb5 : SUCCESS in 2m 51s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/89b6e4eaca20471d8d74a62a5eeb6b49 : SUCCESS in 11m 56s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9bebe8d33f4040c59366df399fa20fbd : SUCCESS in 46m 14s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/63a75bca144f487fa7f379c3300e16d5 : SUCCESS in 1h 05m 16s","accounts_in_message":[],"_revision_number":4},{"id":"29c174eeebd3a0a23c3154ed8706cff97b74f7bc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-01 11:23:00.000000000","message":"Patch Set 4: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"2dcbce221fb247f3a31d9e4be38205cc00e5e748","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2023-03-01 11:59:15.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"5f4b97bf463fcd2f114420abafaf679e3ddd7acc","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-01 12:01:46.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"c06c767ac47b64ff93bdd813fb1c47779bc1fe75","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-01 13:01:49.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bedbec2fb7c14cb5bc51ce63aa9c2e22\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/25ca9e8772f648ad8dda5fa2558c6ceb : SUCCESS in 3m 07s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6cc4220dbe00414498c80d8c4b120b45 : SUCCESS in 2m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ba8f0cae5d9048cb89ae0e716dbdbaca : SUCCESS in 2m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0bef95eebf5e4e2f846361aec176de59 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a7e7310d1322457a8278e67817f0004a : SUCCESS in 5m 55s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a33b6c5ba002410497339b3326c1f716 : SUCCESS in 3m 05s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e9601a3096f24678822206fc1562fffe : SUCCESS in 9m 50s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/4556eb4fa1e7469fb0b80da0ef829eeb : SUCCESS in 45m 04s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bbe283f8e4304554b49f3c66a19ec0bc : SUCCESS in 56m 20s","accounts_in_message":[],"_revision_number":5},{"id":"c0483908420ee8d63c1db624e72e7dcb40ac694b","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-03-08 20:51:32.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"6c58eddb801df53749eeabb645f3372a9b26ab4b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 20:52:05.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"a40bf5db2a098b22be30fb8df4f0cf2d78ef6946","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 21:54:13.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c24afd5ed52545c68cafd46b8c683ad2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19b3236108ac4fa6befb9bfe265e2b34 : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/da2b2a98bc3c402bb7171c00b6d47070 : SUCCESS in 2m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/042d8ca818014c018d37e6f2a930d0c3 : SUCCESS in 3m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c266142d6bfe47608432f7f9eeae0581 : SUCCESS in 5m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bfd5c48eab7747b7bb56a64736c3054e : SUCCESS in 3m 04s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/64e6fea840184d4f84fb5de3aa80ade3 : SUCCESS in 8m 59s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1122404f53df4bd8af5417c07fb3213e : SUCCESS in 42m 29s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1df126c09067455c8cb78cc87b96567f : SUCCESS in 57m 20s","accounts_in_message":[],"_revision_number":5},{"id":"f1b908ed526411741afdac57e2e9a71a81bdc9cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 21:54:14.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c7f88ee8fd2cd885030e01fc09ca85ae378f1976","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-08 21:55:11.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e3be78ecc664184968002cf81b774d8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5e10c0bf0bf04085b9f76cdfd405fc24 : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/98d739dc6878410da5db0ce4f812de6b : SUCCESS in 38s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"fbf44e92c4a820ce0063cfbb4d8a6ad4c224f7ca","revisions":{"a093b6c0c6d94819036e674c76bd103d12c12f56":{"kind":"REWORK","_number":1,"created":"2023-01-27 20:43:43.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/49/872049/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/872049/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/872049/1"}}},"commit":{"parents":[{"commit":"98c899997f4f7b7aaccf911c41655aad6548fe6c","subject":"workaround: requests verify handling if env is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98c899997f4f7b7aaccf911c41655aad6548fe6c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-01-27 20:29:36.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-01-27 20:43:42.000000000","tz":-180},"subject":"[WIP] Handle non-default language for registries","message":"[WIP] Handle non-default language for registries\n\nThe default value for language in sushy is \"en\",\nin the ZT hardware some registries have \"en-US\"\nas their language, in this case we can\u0027t get the\nright information.\n\nThis commits changes the logic to accept the case\nwhere the registry language starts with the default.\n\nChange-Id: I0c4690925d086af593a691565a66ce90c19be156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a093b6c0c6d94819036e674c76bd103d12c12f56"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a093b6c0c6d94819036e674c76bd103d12c12f56"}]},"branch":"refs/heads/master"},"f5b0aa4656e51cd6ee7924f84fcfa63f2f896023":{"kind":"REWORK","_number":2,"created":"2023-02-27 03:20:22.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/49/872049/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/872049/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/872049/2"}}},"commit":{"parents":[{"commit":"98c899997f4f7b7aaccf911c41655aad6548fe6c","subject":"workaround: requests verify handling if env is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98c899997f4f7b7aaccf911c41655aad6548fe6c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-01-27 20:29:36.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-02-27 03:20:12.000000000","tz":-180},"subject":"Handle non-default language for registries","message":"Handle non-default language for registries\n\nThe default value for language in sushy is \"en\",\nin the ZT hardware some registries have \"en-US\"\nas their language, in this case we can\u0027t get the\nright information.\n\nThis commits changes the logic to accept the case\nwhere the registry language starts with the default.\n\nChange-Id: I0c4690925d086af593a691565a66ce90c19be156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f5b0aa4656e51cd6ee7924f84fcfa63f2f896023"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f5b0aa4656e51cd6ee7924f84fcfa63f2f896023"}]},"branch":"refs/heads/master"},"29d040eb881457638dd7d3beb5ea2285d9045fa0":{"kind":"REWORK","_number":3,"created":"2023-02-28 00:06:14.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/49/872049/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/872049/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/872049/3"}}},"commit":{"parents":[{"commit":"98c899997f4f7b7aaccf911c41655aad6548fe6c","subject":"workaround: requests verify handling if env is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98c899997f4f7b7aaccf911c41655aad6548fe6c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-01-27 20:29:36.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-02-28 00:05:59.000000000","tz":-180},"subject":"Handle non-default language for registries","message":"Handle non-default language for registries\n\nThe default value for language in sushy is \"en\",\nin the ZT hardware some registries have \"en-US\"\nas their language, in this case we can\u0027t get the\nright information.\n\nThis commits changes the logic to accept the case\nwhere the registry language starts with the default.\n\nChange-Id: I0c4690925d086af593a691565a66ce90c19be156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/29d040eb881457638dd7d3beb5ea2285d9045fa0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/29d040eb881457638dd7d3beb5ea2285d9045fa0"}]},"branch":"refs/heads/master"},"bf12d67b333f03e1d8c622af3c61ee9c898d1d2f":{"kind":"REWORK","_number":4,"created":"2023-02-28 15:59:10.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/49/872049/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/872049/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/872049/4"}}},"commit":{"parents":[{"commit":"98c899997f4f7b7aaccf911c41655aad6548fe6c","subject":"workaround: requests verify handling if env is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98c899997f4f7b7aaccf911c41655aad6548fe6c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-01-27 20:29:36.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-02-28 15:58:21.000000000","tz":-180},"subject":"Handle non-default language for registries","message":"Handle non-default language for registries\n\nThe default value for language in sushy is \"en\",\nin the ZT hardware some registries have \"en-US\"\nas their language, in this case we can\u0027t get the\nright information.\n\nThis commits changes the logic to accept the case\nwhere the registry language starts with the default.\n\nChange-Id: I0c4690925d086af593a691565a66ce90c19be156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bf12d67b333f03e1d8c622af3c61ee9c898d1d2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bf12d67b333f03e1d8c622af3c61ee9c898d1d2f"}]},"branch":"refs/heads/master"},"fbf44e92c4a820ce0063cfbb4d8a6ad4c224f7ca":{"kind":"REWORK","_number":5,"created":"2023-03-01 11:59:15.000000000","uploader":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"ref":"refs/changes/49/872049/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/872049/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/872049/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/872049/5"}}},"commit":{"parents":[{"commit":"98c899997f4f7b7aaccf911c41655aad6548fe6c","subject":"workaround: requests verify handling if env is set","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/98c899997f4f7b7aaccf911c41655aad6548fe6c"}]}],"author":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-01-27 20:29:36.000000000","tz":-180},"committer":{"name":"Iury Gregory Melo Ferreira","email":"imelofer@redhat.com","date":"2023-03-01 11:58:27.000000000","tz":-180},"subject":"Handle non-default language for registries","message":"Handle non-default language for registries\n\nThe default value for language in sushy is \"en\",\nin the ZT hardware some registries have \"en-US\"\nas their language, in this case we can\u0027t get the\nright information.\n\nThis commits changes the logic to accept the case\nwhere the registry language starts with the default.\n\nChange-Id: I0c4690925d086af593a691565a66ce90c19be156\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fbf44e92c4a820ce0063cfbb4d8a6ad4c224f7ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fbf44e92c4a820ce0063cfbb4d8a6ad4c224f7ca"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"OK","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"OK","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[]}
