)]}'
{"id":"openstack%2Fsushy~877077","triplet_id":"openstack%2Fsushy~master~I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","project":"openstack/sushy","branch":"master","topic":"exclude-local-flake8","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-03-16 16:00:46.000000000","reason":"Change was submitted"},"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2023-03-16 15:16:39.000000000","reason":"removed on reply"}},"hashtags":["ironic-week-prio"],"change_id":"I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","subject":"Exclude all files starting with . from flake8 tests","status":"MERGED","created":"2023-03-10 08:22:17.000000000","updated":"2023-03-16 16:01:53.000000000","submitted":"2023-03-16 16:00:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"877077-exclude-local-flake8","meta_rev_id":"8a3d64165c4df5ed18d71ae5a47f5f8b0612df24","_number":877077,"virtual_id_number":877077,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-16 16:00:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2023-03-14 17:44:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-16 15:16:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-16 15:16:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-10 09:36:12.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-10 09:41:34.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2023-03-14 17:44:15.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2023-03-16 15:16:39.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"04f396f815e5dc804eb53e0520e7388d7211bbaa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-10 08:22:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6219bd09ffda50b6da531e008c0923fdf28c69ef","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-10 08:45:18.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"ee835d27b1acffc774a7d42b6bd2d778db5e2fac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 09:36:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/102817367bcc41008274c6b6c39456c8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb146f6e550e45b0b17debda27ae3860 : SUCCESS in 3m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c2c5411c189a4e279b5f84bfcaba9349 : SUCCESS in 2m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c00d90da45b44898e7d9e8dd3f232d6 : SUCCESS in 3m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2382d2b93fed4a44a571ef0e1ca4ff08 : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6844018e69bf46bb86e861ae29f8d146 : SUCCESS in 6m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fd3caa6dff9d406e952ad9f434544fe5 : SUCCESS in 3m 19s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/5fa9cd00eb1f4ed8a6c863ba65e181bd : SUCCESS in 9m 57s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9a21cf3982114ddd98d625d03a976416 : SUCCESS in 43m 53s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d1186244124747adb20abe4eaef1a632 : SUCCESS in 55m 52s","accounts_in_message":[],"_revision_number":1},{"id":"41f488438c1d8f725ecb326e5608bee2af32105b","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-10 09:41:34.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bae42f03d0c9361a54788da77312d24fba04bd86","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-10 11:16:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eecf80a3d9e951899a31778dcc9314379b4ba9fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-10 11:19:58.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"e45f08d40f06faccb76dc623a89810aaf1f9888f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-10 12:21:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f01db02f0ce14b03ab9f54c0b19c6aa5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/045f4f2f5e97409a832d2861c8e6d341 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c22b64c1cfd64ee697719db048970973 : SUCCESS in 2m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/89f02bf507eb4f7089561bcc5c9af860 : SUCCESS in 3m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c12b823ccf5042b79bcc3bb750eab5c5 : SUCCESS in 3m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7064fd4cad684b80b1492f2c0972b2cb : SUCCESS in 5m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/af4b738b9c634c138fdabdd790ec3621 : SUCCESS in 3m 23s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/87371638db0345c5a7ca98ffdb912e6b : SUCCESS in 7m 44s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/241b2691792248f291b3e221491e055d : SUCCESS in 44m 49s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6406ecb1a85446ccb801f57227df2ab1 : SUCCESS in 56m 10s","accounts_in_message":[],"_revision_number":2},{"id":"f99cb2189dbd2ac0227dd5772bb228cf67b22786","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2023-03-14 17:44:15.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b7485ccfba72b5165889a7ee122e1cb1843aca27","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-16 15:16:39.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"703228e9dfd65647666639dca0eb1946c9bba367","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 15:17:50.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"aab23c162a000e3f67fb87b1939fc912e0051345","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 16:00:46.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0153919bd4da43488ad3319c3abb8e52\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d46d022409e9417f801b08a526ab7d80 : SUCCESS in 3m 28s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/aed369d5e03f4a6b8a7ce265875ed291 : SUCCESS in 3m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5ac4017d7c3e4463bae5a655cf1b2d0e : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c1ba6ff80174689a26ef94fb908c111 : SUCCESS in 6m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6543a16e367b4ba699b53615ef4b2cf4 : SUCCESS in 3m 39s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/937a7209f59c47338e7c4d79a4d80ebd : SUCCESS in 8m 08s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/35485bf71b324d5fbff97a1b24cd2807 : SUCCESS in 30m 39s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b587700191d64834a134224a13c02eb4 : SUCCESS in 41m 23s","accounts_in_message":[],"_revision_number":2},{"id":"67ea4674049da70a32ee6768b0440bf0fc6a94de","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 16:00:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8a3d64165c4df5ed18d71ae5a47f5f8b0612df24","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-16 16:01:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/320eccf32c0a44c1b7deebd9054372ab\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/97cef3cfc6114cdfbb450e465aabff3c : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/68f7485e229e442fa5fc2e9a47687938 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"a59cf6c6986c251fab0ba4f116c2c1e635028f34","revisions":{"97dec7da5a4308c0d0db7a2d63ecca86e2a8ba8e":{"kind":"REWORK","_number":1,"created":"2023-03-10 08:22:17.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/77/877077/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/77/877077/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/77/877077/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/77/877077/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/77/877077/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/77/877077/1"}}},"commit":{"parents":[{"commit":"40f460183a14040db97b0c16e7d42438a2d4ac98","subject":"Merge \"Handle non-default language for registries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/40f460183a14040db97b0c16e7d42438a2d4ac98"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-10 08:20:01.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-10 08:20:01.000000000","tz":60},"subject":"Exclude .local from flake8 tests","message":"Exclude .local from flake8 tests\n\nSome tests tool and IDE may create a .local directory inside the repo\nwith virtualenvs for dependencies.\nWhile this can be removed in a second time, or configured differently,\nit\u0027s advisable to just exclude .local from the flake8 tests to avoid\nconfusion and possible unexpected errors.\n\nChange-Id: I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/97dec7da5a4308c0d0db7a2d63ecca86e2a8ba8e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/97dec7da5a4308c0d0db7a2d63ecca86e2a8ba8e"}]},"branch":"refs/heads/master"},"a59cf6c6986c251fab0ba4f116c2c1e635028f34":{"kind":"REWORK","_number":2,"created":"2023-03-10 11:19:58.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/77/877077/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/77/877077/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/77/877077/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/77/877077/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/77/877077/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/77/877077/2"}}},"commit":{"parents":[{"commit":"40f460183a14040db97b0c16e7d42438a2d4ac98","subject":"Merge \"Handle non-default language for registries\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/40f460183a14040db97b0c16e7d42438a2d4ac98"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-10 08:20:01.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-10 11:19:44.000000000","tz":60},"subject":"Exclude all files starting with . from flake8 tests","message":"Exclude all files starting with . from flake8 tests\n\nSome tests tool and IDE may create a .local directory inside the repo\nwith virtualenvs for dependencies, other tools may create . directories\nor files for temporary reports.\nWhile they can be removed in a second time, or configured differently,\nit\u0027s advisable to just exclude all file starting with . from the\nflake8 tests to avoid confusion and possible unexpected errors.\n\nChange-Id: I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a59cf6c6986c251fab0ba4f116c2c1e635028f34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a59cf6c6986c251fab0ba4f116c2c1e635028f34"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
