)]}'
{"id":"openstack%2Fsushy~878018","triplet_id":"openstack%2Fsushy~stable%2Fzed~I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","project":"openstack/sushy","branch":"stable/zed","topic":"exclude-local-flake8-stable/zed","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-03-23 16:53:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","subject":"Exclude all files starting with . from flake8 tests","status":"MERGED","created":"2023-03-21 10:17:41.000000000","updated":"2023-03-23 16:55:12.000000000","submitted":"2023-03-23 16:53:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878018-exclude-local-flake8-stable/zed","cherry_pick_of_change":877077,"cherry_pick_of_patch_set":2,"meta_rev_id":"dfd831ec90bad38befccd16e9700901c8e9ff757","_number":878018,"virtual_id_number":878018,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 16:53:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-23 15:37:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-23 15:37:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 11:13:50.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 15:37:18.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"db4cf69a7b660b8a090e0435190998039f472731","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-21 10:17:41.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"7d16cb0a6b23d78c71e6e34ed5bf825103f67ded","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 11:13:50.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f453d5313acf43de9b7e6616931c4c77\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/842dade1e1b14de1b73852e8e518eb44 : SUCCESS in 3m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8f018a5747e44cf8cb8c8e1b2892f44 : SUCCESS in 2m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/102ef5a143164eea8b44e328732f7497 : SUCCESS in 2m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/97a2602e571f4f87876224b28067fc9b : SUCCESS in 3m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/62125e334e7948bdb3d9fa2bc14278ff : SUCCESS in 3m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be3ff3bea0d4425b86fac9ae4b75c1e8 : SUCCESS in 6m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4493dcd5d67d40a09166d4e5efb77404 : SUCCESS in 3m 03s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d1a39ba31e314cdc9a3c7cb17df45750 : SUCCESS in 9m 15s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/779914e7623e409caa02413306fa4892 : SUCCESS in 40m 38s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/229efbd6f2a64643be727eaaf7ce9a5c : SUCCESS in 50m 33s","accounts_in_message":[],"_revision_number":1},{"id":"ab200838fd5eeb78b1970b1f0f31a1d65bd2e951","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-23 15:37:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a52255c4ece69cefed75035c78460a76e0d61b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:37:55.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"2eb133b3251fb4811e9ef33d89cea1a344fbbf02","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:53:45.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0533966c2b544820b28b7645c96160f3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/33e35d89e54849fdbee64da8bfff128b : SUCCESS in 2m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb2b73e79bd64494823788c1e698ec9a : SUCCESS in 3m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5fe34cb887424443a9213e8b6a8b6cf6 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7029903aaa5e446480653d191b195bde : SUCCESS in 5m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0803d81af4d41488741ed655b61fb9c : SUCCESS in 3m 13s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/3a661481a64946c2a33a93cb7e0c2a51 : SUCCESS in 8m 01s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/cd72d54fa7b540129f2d73c2c55af433 : SUCCESS in 45m 49s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/3801b980753f4ad5a44ca395a836a422 : SUCCESS in 55m 35s","accounts_in_message":[],"_revision_number":1},{"id":"b1a4aa04c70fdb41716df71338854437e485a2eb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:53:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"dfd831ec90bad38befccd16e9700901c8e9ff757","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:55:12.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0a3546b0d01c4175a6a2dc8c115ac3fb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8444062c255446a81367844f7520cae : SUCCESS in 39s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ee3af8fb5f24425b923f4fc1b3e94173 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e4108b3d7a8eaa9da97bb9278b797c22698c12dc","revisions":{"e4108b3d7a8eaa9da97bb9278b797c22698c12dc":{"kind":"REWORK","_number":1,"created":"2023-03-21 10:17:41.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/18/878018/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/18/878018/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/878018/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/878018/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/878018/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/18/878018/1"}}},"commit":{"parents":[{"commit":"844d4ac864787b94fd1ac637861d92755f520f44","subject":"Merge \"Retry on iDRAC SYS518 errors for all requests\" into stable/zed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/844d4ac864787b94fd1ac637861d92755f520f44"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-10 08:20:01.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-21 10:17:41.000000000","tz":0},"subject":"Exclude all files starting with . from flake8 tests","message":"Exclude all files starting with . from flake8 tests\n\nSome tests tool and IDE may create a .local directory inside the repo\nwith virtualenvs for dependencies, other tools may create . directories\nor files for temporary reports.\nWhile they can be removed in a second time, or configured differently,\nit\u0027s advisable to just exclude all file starting with . from the\nflake8 tests to avoid confusion and possible unexpected errors.\n\nChange-Id: I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7\n(cherry picked from commit a59cf6c6986c251fab0ba4f116c2c1e635028f34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e4108b3d7a8eaa9da97bb9278b797c22698c12dc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e4108b3d7a8eaa9da97bb9278b797c22698c12dc"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
