)]}'
{"id":"openstack%2Fsushy~878019","triplet_id":"openstack%2Fsushy~stable%2Fyoga~I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","project":"openstack/sushy","branch":"stable/yoga","topic":"exclude-local-flake8-stable/yoga","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-03-23 16:53:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","subject":"Exclude all files starting with . from flake8 tests","status":"MERGED","created":"2023-03-21 10:17:54.000000000","updated":"2023-03-23 16:55:10.000000000","submitted":"2023-03-23 16:53:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878019-exclude-local-flake8-stable/yoga","cherry_pick_of_change":877077,"cherry_pick_of_patch_set":2,"meta_rev_id":"fef9781930653e792ddb75babc38e0c9898c4748","_number":878019,"virtual_id_number":878019,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 16:53:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-23 15:37:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-23 15:37:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 11:38:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 15:37:14.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"20ac7ef3a5e45b2803172b5c1097694e9e18ef37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-21 10:17:54.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"d72df55668cc29315fd32f7726c78468db16107d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 11:38:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5f6b4090f1db4bcda266f1a924d775ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/478d213e1c304dbf82b93f3136552361 : SUCCESS in 3m 14s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4de804ffa4b14a3a91529f9d7bed1fa0 : SUCCESS in 2m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/1b50ca937afa41aa952ada3401cdf6fd : SUCCESS in 2m 49s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a500a41400de491b9ab45d50e3c7454b : SUCCESS in 3m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e231ba0e99540dcae89208929a5a43e : SUCCESS in 5m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dada05f067df4b6d950c0a259a2d8ddf : SUCCESS in 3m 30s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7c0c2c41c9114a738018d1005aa7a4a7 : SUCCESS in 6m 18s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bc6f25f078584e92b635576eefd8436f : SUCCESS in 1h 11m 19s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8d734f93547d466b93497d87a1398d12 : SUCCESS in 1h 04m 36s","accounts_in_message":[],"_revision_number":1},{"id":"20b0cdde3e159615f92b9e3dd45b80cab69550aa","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-23 15:37:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11515e674c47c1de77dcef7cb20544b08f0ed01a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:37:50.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b22efe1c031580cffbd2c08d52a608bb8ae371b8","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:53:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9612e93ea6e4efe8bf83f0f9dbae31b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93082b5ad66a4cdaa5197e6fc7df1b68 : SUCCESS in 3m 32s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/db76859196054eb3a737209c6f507513 : SUCCESS in 3m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20f6169e61eb4d568087644c6d25d858 : SUCCESS in 3m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f73204eb4ed3450a961acf6de852898e : SUCCESS in 5m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d4b6b18b61ec4ba7a3cb3eea6af0ebd7 : SUCCESS in 3m 39s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/6da57065d6db4321b5e207570c0ce47a : SUCCESS in 8m 02s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/733d9237a8d44e7da18eb599bf24f55b : SUCCESS in 1h 01m 37s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/72b428265f93492baff391b3f40da292 : SUCCESS in 1h 11m 02s","accounts_in_message":[],"_revision_number":1},{"id":"cc1f38477a7ad1371f4cdf32e4c1f78da4bab251","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:53:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"fef9781930653e792ddb75babc38e0c9898c4748","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:55:10.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0827f92df90942a6b5427dd2ad3bad55\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ffb116200cf4b1a972620282dcfface : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e4ff7e6961fc4a27a5833e9c197411e2 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"e5a43fa2cc2aeb55cfe203d48a9005be11ced6d3","revisions":{"e5a43fa2cc2aeb55cfe203d48a9005be11ced6d3":{"kind":"REWORK","_number":1,"created":"2023-03-21 10:17:54.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/19/878019/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/19/878019/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/878019/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/878019/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/878019/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/19/878019/1"}}},"commit":{"parents":[{"commit":"6d7951f6cbbcca3ad7dfeaffd1bb92111dee67a6","subject":"Merge \"Handle proper code_status in unit test\" into stable/yoga","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6d7951f6cbbcca3ad7dfeaffd1bb92111dee67a6"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-10 08:20:01.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-21 10:17:54.000000000","tz":0},"subject":"Exclude all files starting with . from flake8 tests","message":"Exclude all files starting with . from flake8 tests\n\nSome tests tool and IDE may create a .local directory inside the repo\nwith virtualenvs for dependencies, other tools may create . directories\nor files for temporary reports.\nWhile they can be removed in a second time, or configured differently,\nit\u0027s advisable to just exclude all file starting with . from the\nflake8 tests to avoid confusion and possible unexpected errors.\n\nChange-Id: I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7\n(cherry picked from commit a59cf6c6986c251fab0ba4f116c2c1e635028f34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e5a43fa2cc2aeb55cfe203d48a9005be11ced6d3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/e5a43fa2cc2aeb55cfe203d48a9005be11ced6d3"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
