)]}'
{"id":"openstack%2Fsushy~878020","triplet_id":"openstack%2Fsushy~stable%2Fxena~I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","project":"openstack/sushy","branch":"stable/xena","topic":"exclude-local-flake8-stable/xena","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-03-23 16:53:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7","subject":"Exclude all files starting with . from flake8 tests","status":"MERGED","created":"2023-03-21 10:18:06.000000000","updated":"2023-03-23 16:55:12.000000000","submitted":"2023-03-23 16:53:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"878020-exclude-local-flake8-stable/xena","cherry_pick_of_change":877077,"cherry_pick_of_patch_set":2,"meta_rev_id":"7d886a1ed855a2e29023b6941242401a1d29a6e5","_number":878020,"virtual_id_number":878020,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-03-23 16:53:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-03-23 15:37:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-03-23 15:37:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-03-21 11:10:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-03-23 15:37:21.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"5f384f9f22f1f916770e5460d694f527f47e56ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-03-21 10:18:06.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"156324f576b379fe055f022e8b901c51bb7b79f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-21 11:10:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cdee387e56674bf5a9d17751d217e2f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e0a1fc29eaf42589eb584f242dee142 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/51cbf3e4227941588823b31875d862ce : SUCCESS in 2m 55s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/99c13ed28f234ee6ab0e5003c0ffc8d4 : SUCCESS in 3m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e720f91d64ac4a8aab11959aea3458f7 : SUCCESS in 3m 32s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b137197f873942cbbb4fdf6e82c65433 : SUCCESS in 3m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fb0071e1c31b4df88c9c577d5148748c : SUCCESS in 6m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e7ad4ae0ec441a195b65397f59f2a54 : SUCCESS in 3m 17s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/463c5936176546a3988dc212ef7b9965 : SUCCESS in 44m 23s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bd7cd37cdc1240819cb4aed0124adc3e : SUCCESS in 49m 42s","accounts_in_message":[],"_revision_number":1},{"id":"2640cee18ba20f178ed5ce07b9535f0ba5b6e2b2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-03-23 15:37:21.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3baea4313cd38afc0d3308a383bc7a3e4b702301","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 15:37:59.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"413cb122f1ef7e0a2a366beef1a6951c2c23bde2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:53:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/125b92445e574b0a8140ef89c759f960\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bbc84b42a124fc38e91399c465376a0 : SUCCESS in 2m 42s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/aca355c01de54e6cbeacaaaebc4775a6 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/30683c00e0a14123884ca8a13c0aeca9 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/52c2fadef30c43a2b81b0a47fb19642f : SUCCESS in 6m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5d7a2a8223d44cfc9dcce648f53866fd : SUCCESS in 2m 49s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d3c7d8b9400a4b888e5175d3cefa19c0 : SUCCESS in 54m 40s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a2dc7d28ade0447c99093bf148a817b9 : SUCCESS in 48m 58s","accounts_in_message":[],"_revision_number":1},{"id":"ccb031482d4f370a0a55009bd55248ab1233a0d1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:53:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"7d886a1ed855a2e29023b6941242401a1d29a6e5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-03-23 16:55:12.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/77c7cada6b704d3cafbc28fc6c90e4f8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c978ca3f2a514247ad5def11f442e95b : SUCCESS in 37s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0b9094984bdd4c55b1c86be2b8e90bab : FAILURE in 33s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f5e1511c2326f67d0537314874e1ab63c7bb655e","revisions":{"f5e1511c2326f67d0537314874e1ab63c7bb655e":{"kind":"REWORK","_number":1,"created":"2023-03-21 10:18:06.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/20/878020/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/20/878020/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/20/878020/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/20/878020/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/20/878020/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/20/878020/1"}}},"commit":{"parents":[{"commit":"d67f16ac8e1bd470c4c932e26cb8b8aafee95776","subject":"Merge \"Fix volume deletion on newer iDRACs\" into stable/xena","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d67f16ac8e1bd470c4c932e26cb8b8aafee95776"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-10 08:20:01.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2023-03-21 10:18:06.000000000","tz":0},"subject":"Exclude all files starting with . from flake8 tests","message":"Exclude all files starting with . from flake8 tests\n\nSome tests tool and IDE may create a .local directory inside the repo\nwith virtualenvs for dependencies, other tools may create . directories\nor files for temporary reports.\nWhile they can be removed in a second time, or configured differently,\nit\u0027s advisable to just exclude all file starting with . from the\nflake8 tests to avoid confusion and possible unexpected errors.\n\nChange-Id: I6fdee20603437773c8a8c08cce3bbdcb4f8ef0e7\n(cherry picked from commit a59cf6c6986c251fab0ba4f116c2c1e635028f34)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f5e1511c2326f67d0537314874e1ab63c7bb655e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f5e1511c2326f67d0537314874e1ab63c7bb655e"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
