)]}'
{"id":"openstack%2Fsushy~882749","triplet_id":"openstack%2Fsushy~stable%2Fwallaby~Ib78198a60a8924de934bda0c9a0b9298541496cf","project":"openstack/sushy","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2024-03-14 13:04:39.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ib78198a60a8924de934bda0c9a0b9298541496cf","subject":"Retry on ilo state error","status":"ABANDONED","created":"2023-05-09 19:32:59.000000000","updated":"2024-03-14 13:04:39.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"caeaa54c59d773b38b526ec2e2483d12c7cbef1b","_number":882749,"virtual_id_number":882749,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-05-09 20:19:25.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-05-09 20:19:25.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-03-14 13:04:39.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"fef3a8b0015098b198909aa004a1d70102f53561","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-05-09 19:32:59.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7edb3c19a8dee707c42c5a75de7885356f669eda","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-05-09 20:19:25.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9c5b1b99ac34bbb9d967e4b94be55fe\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0944dd1c88924bdfa7ef85079401edb3 : FAILURE in 2m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b931c6a839f64b8a9539169ec0e34b6c : FAILURE in 2m 23s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/24155918faa54db883b6f194e21247cb : FAILURE in 2m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/910fdddc7d414d7085a9f938f3d10bf7 : FAILURE in 3m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0fc832427c5f4094888ada8b1e766830 : FAILURE in 2m 32s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f7e52778daf4ef2ac0b0dba1f56cd4c : SUCCESS in 5m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ef2fc380de1a44e4b5e5f5b2c958b5be : SUCCESS in 2m 44s\n- sushy-tempest-partition-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/30d7ca3ab1594bcb9f57994bae5c76ea : SUCCESS in 37m 37s\n- sushy-tempest-partition-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/fe1d82f7de844e09986ed86077047abd : SUCCESS in 45m 24s","accounts_in_message":[],"_revision_number":1},{"id":"caeaa54c59d773b38b526ec2e2483d12c7cbef1b","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-14 13:04:39.000000000","message":"Abandoned\n\nstable/wallaby branch of openstack/sushy is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/wallaby if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2ac3edcc0bcf0279cf6156d284962527fdeb0478","revisions":{"2ac3edcc0bcf0279cf6156d284962527fdeb0478":{"kind":"REWORK","_number":1,"created":"2023-05-09 19:32:59.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/49/882749/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/49/882749/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/882749/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/882749/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/49/882749/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/49/882749/1"}}},"commit":{"parents":[{"commit":"0ed4f93a36a890517cf0efa96fa1e595c14ad34f","subject":"Fix volume deletion on newer iDRACs","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0ed4f93a36a890517cf0efa96fa1e595c14ad34f"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-04-14 18:31:43.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-05-09 19:32:52.000000000","tz":-420},"subject":"Retry on ilo state error","message":"Retry on ilo state error\n\nThis not great patch allows us to automatically retry when we\nencounter a message from the BMC indicating that the requested\noperation was invalid.\n\nWe\u0027ve seen similar issues on other BMCs in the past, however\nin this case the BMC uses a HTTP 400 error as opposed to a 500\nerror. It is not perfect, but easily accounted for with\ndiffering approaches to the client being aware or not of the\nunderlying state.\n\nChange-Id: Ib78198a60a8924de934bda0c9a0b9298541496cf\nCloses-Bug: #2016307\n(cherry picked from commit eee778a70233e61863acc1709f4554a1862de749)\n(cherry picked from commit fa4b0f781a0460d05a4c33275706ab5733b29eb9)\n(cherry picked from commit b758c8fda657fc9c0390fc056d99e4e58e1c0903)\n(cherry picked from commit 3baaa20fa823486182905dcee9d0df0a8b1c0204)\n(cherry picked from commit b86ecdf393b689c705b793db2b482d93c4e5d271)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2ac3edcc0bcf0279cf6156d284962527fdeb0478"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2ac3edcc0bcf0279cf6156d284962527fdeb0478"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
