)]}'
{"id":"openstack%2Fsushy~892261","triplet_id":"openstack%2Fsushy~master~Ifd01b5f633cb5076abb09a13c223b19f07e68fe6","project":"openstack/sushy","branch":"master","topic":"fix_wrong_resource_base_get_registry","attention_set":{},"removed_from_attention_set":{"30566":{"account":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"last_update":"2023-10-15 02:12:35.000000000","reason":"\u003cGERRIT_ACCOUNT_30566\u003e replied on the change","reason_account":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"}}},"hashtags":[],"change_id":"Ifd01b5f633cb5076abb09a13c223b19f07e68fe6","subject":"Fix wrong _get_registry logic in ResourceBase","status":"MERGED","created":"2023-08-22 06:27:58.000000000","updated":"2023-10-15 04:57:46.000000000","submitted":"2023-10-15 04:56:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"892261-fix_wrong_resource_base_get_registry","meta_rev_id":"80bfd1164a37b49c2e258387aee1dbd705c6cd7b","_number":892261,"virtual_id_number":892261,"owner":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-15 04:56:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2023-09-28 17:03:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 16:05:00.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-11 16:05:00.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-08-22 07:29:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-28 17:03:07.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2023-10-11 16:05:00.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"d884ce8539fad7b99cea2c7670ef860a1b873023","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"date":"2023-08-22 06:27:58.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a7e446caebe74b4586025b7b35f530909b67f6db","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-08-22 07:29:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/35bb1ece40794508a3e27ab95f5b2adf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/42d53743cdad4e579e262af68ca348a4 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6778177289c34d0a8fd8c67726071d0b : SUCCESS in 2m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5569165a0a540089907ab20f94806a9 : SUCCESS in 2m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0974b47cdace4d038ef53ddf250b9960 : SUCCESS in 2m 38s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9877ab0ad217438f8f327e5a89978972 : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/579b617452b44c78868389225f0b2b12 : SUCCESS in 2m 46s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc16c8a3b82e401b86c657c860073523 : SUCCESS in 4m 56s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/79e37a365f9f42e69b6f8bef9f05636c : SUCCESS in 7m 52s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/06fa8e9290644d48982e46ba6d9d1ab6 : SUCCESS in 45m 23s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/5377d7c7cb4b499698be64290b63ad83 : SUCCESS in 56m 37s","accounts_in_message":[],"_revision_number":1},{"id":"259f130e20942169d60bf4db00f02cdb1b37f882","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-28 17:03:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"d9fef6a864faef77e221d0a33049c0780f97e068","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2023-10-11 16:05:00.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"cac8b76edc1e1602c0c4935cb6578229d78bc149","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:05:36.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"eaab733d4daa29224bd0744847950854fe134b73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 16:40:58.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/faacfa39670a47b1ad1336bd1391caaa\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3b9a84dfd64c41a79b133dc2bc989238 : SUCCESS in 2m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ba59934e2724a23addf9e7f3f1e5381 : SUCCESS in 2m 56s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fda3c7cfd874460c91ee708f509d3f48 : SUCCESS in 3m 28s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/66aa439906d64c41bf38322d442ad749 : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bb1182aaefa14cfb8c6e84e93d624fe9 : SUCCESS in 5m 54s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a5d80fc8e08c45a9a561e6ecd2ea5357 : SUCCESS in 8m 19s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/b167e63f11014ab6ab23668772fb5e72 : FAILURE in 30m 46s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/bb83ba4fa7dc4b00a26bf167840c12ed : FAILURE in 24m 59s","accounts_in_message":[],"_revision_number":1},{"id":"d49fbfc342a084f2e7f488bad30d57f103723322","author":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"date":"2023-10-15 02:12:35.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2fb77a56c6290e3267c69098f859299fed989d92","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-15 03:55:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/629f4e7088524629bf153ce8d1e7b56d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ae6bad16923b47d6a4a370c4eee28276 : SUCCESS in 5m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/dda71348aa914bf99553595d86187e7d : SUCCESS in 3m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1b4c95ce29c441bca47fa60a5c5fad7f : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/13d00d7fd6b446ca9a3448718081b1d5 : SUCCESS in 3m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1696eb28516a4ff7bd2098791f56d5b3 : SUCCESS in 3m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/202f49c1afef4f8c9463eb78a08cec7b : SUCCESS in 3m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d568ab96ae4f418c9e681b7c24c7971f : SUCCESS in 8m 17s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1ec4386ed5de46209735b758bf0c6973 : SUCCESS in 9m 21s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/03741b48463d452fa1b1c126c7b6968f : SUCCESS in 40m 07s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/02fbfa4b84274d4f8a189dabc134ccc9 : SUCCESS in 58m 32s","accounts_in_message":[],"_revision_number":1},{"id":"17fbaabf49c798d4e1a1b39bf1393bd44dc78844","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-15 03:55:51.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1a338c8a3f9b2eec872428e82e08fd0ef0e948ca","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-15 04:56:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/12a5e1efb7cb4cb483777fe323fa84cf\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/345dc3dbc45c4a6c8b213d1c481ee3be : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b5ed3019d0e24b78b4cbdfe214c47086 : SUCCESS in 3m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5cbfaddfa7654c358463213f4bd5f2f1 : SUCCESS in 3m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/083f68565c17475aade3c585fb821836 : SUCCESS in 2m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b33627577da474d886d10ffeed0e79a : SUCCESS in 6m 15s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/77baa4bdfcb145b1b6dd1f436099f70d : SUCCESS in 6m 58s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5479962a898343569d79dc985cc33073 : SUCCESS in 29m 04s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d7a8a6ef1fc740b09b3e649c42e8020d : SUCCESS in 56m 18s","accounts_in_message":[],"_revision_number":1},{"id":"99a1a14377517206e7484d1bb7244e7e473ae741","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-15 04:56:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"80bfd1164a37b49c2e258387aee1dbd705c6cd7b","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-15 04:57:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/408078dfff7148459ec815d1ba03033b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/73f874ef47b64d0fab8fd2ba0ae69a30 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"89a098830baedacaa0870f1e5e8a7fbbf01ec933","revisions":{"89a098830baedacaa0870f1e5e8a7fbbf01ec933":{"kind":"REWORK","_number":1,"created":"2023-08-22 06:27:58.000000000","uploader":{"_account_id":30566,"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","username":"vanou"},"ref":"refs/changes/61/892261/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/61/892261/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/61/892261/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/61/892261/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/61/892261/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/61/892261/1"}}},"commit":{"parents":[{"commit":"f03d77ebcb7199fa08a82d5b34d7a6d677d86e0f","subject":"Merge \"Retry on ilo state error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/f03d77ebcb7199fa08a82d5b34d7a6d677d86e0f"}]}],"author":{"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","date":"2023-07-19 05:18:46.000000000","tz":-240},"committer":{"name":"Vanou Ishii","email":"ishii.vanou@fujitsu.com","date":"2023-08-22 06:13:22.000000000","tz":-240},"subject":"Fix wrong _get_registry logic in ResourceBase","message":"Fix wrong _get_registry logic in ResourceBase\n\n_get_registry method of ResourceBase retrieves\nRegistry-like resource specified by identity\nparameter of _get_registry.\nHowever current _get_registry method of ResourceBase\nretrieves Registry-like resource whose name is in\n_attribute_registry. This works with Bios resource\nbut doesn\u0027t with another resource. This commit fixes\nthis.\n\nChange-Id: Ifd01b5f633cb5076abb09a13c223b19f07e68fe6\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/89a098830baedacaa0870f1e5e8a7fbbf01ec933"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/89a098830baedacaa0870f1e5e8a7fbbf01ec933"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
