)]}'
{"id":"openstack%2Fsushy~896835","triplet_id":"openstack%2Fsushy~master~Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5","project":"openstack/sushy","branch":"master","topic":"ironic-week-prio","attention_set":{},"removed_from_attention_set":{"11655":{"account":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"last_update":"2023-11-22 16:29:45.000000000","reason":"Change was submitted"},"4571":{"account":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"last_update":"2023-11-22 16:29:45.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2023-11-06 14:23:34.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":["ironic-week-prio"],"change_id":"Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5","subject":"Add a boot progress indicator","status":"MERGED","created":"2023-09-28 21:00:30.000000000","updated":"2023-11-22 16:30:51.000000000","submitted":"2023-11-22 16:29:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896835-ironic-week-prio","meta_rev_id":"7fb0cdb096f124e68b958407500170f977ab1cd5","_number":896835,"virtual_id_number":896835,"owner":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-22 16:29:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2023-11-22 14:38:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":2,"date":"2023-11-22 14:27:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2023-11-22 14:38:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-28 21:45:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-02 04:08:47.000000000","updated_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"reviewer":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"state":"REVIEWER"},{"updated":"2023-10-03 13:06:22.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"1f299be2a9e9864fcc587339b7b108fe535f3118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-28 21:00:30.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9f0360135ddb07c58ee3c49011647ef875401d5c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 21:45:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9df33023d49d4428b0deb0a69b439584\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9c539f4b56fa479a93a914cc4c0e9e59 : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a20f3f10210f4281b94aed75379eeeef : SUCCESS in 2m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f311972e1f3f406a84c54b5d3f7cc6b5 : SUCCESS in 3m 13s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c75565515a1a468c9f3ec2ce139a9591 : SUCCESS in 2m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1efb0a29a1274cf9af9c92075dc332c2 : SUCCESS in 3m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5631a709fc024b8f891a8c6906bac743 : SUCCESS in 2m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6edd17f9c14d45cb9a2cf9d14902da8f : SUCCESS in 4m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f5e48b4ad161462aa892b36148a29e25 : SUCCESS in 5m 37s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/064e597f41594808afda6866815e31a2 : SUCCESS in 6m 23s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/f4e2c3afe2cd4b14af6e76cdc50a19a9 : SUCCESS in 37m 06s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6d1f2a5e7c184cbe9504e3aee17b4d54 : SUCCESS in 42m 36s","accounts_in_message":[],"_revision_number":1},{"id":"39f670d99031accc82e385894c84e9f013f1420f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-09-28 22:05:36.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"7a2d4808ba52650ad0e9305a2ec779da1880a21c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-28 23:09:58.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e6aa9c00d184c34b302247f48389f88\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6c24000b58cd448cb9921480b2b4b347 : SUCCESS in 3m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06bbd277954f441aad0c814ac2f40941 : SUCCESS in 2m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7eddbbd1f41b4c00b1dc57f1751acc60 : SUCCESS in 2m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/68333a28147441159d3ec2c1284031d5 : SUCCESS in 2m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/993c053e5e4d4430a2f30a7707714808 : SUCCESS in 2m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9539a102b1e6479d88ce96dae834420f : SUCCESS in 5m 41s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a14ae9a9ce54cd19e219b256bc3df5b : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1d130e3fccf04bd5bf20e395facca846 : SUCCESS in 4m 12s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f5f37c8c7d48450fb736bcba09540893 : SUCCESS in 8m 20s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ddf467fde4e54ba9b70145c93916baee : SUCCESS in 37m 58s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/47f29e87a81d400a97fb3fd0ae7c6232 : SUCCESS in 58m 30s","accounts_in_message":[],"_revision_number":2},{"id":"9406740e70b9dfd21fefc517f3f7159c9006838f","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-10-02 04:08:47.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8c2adbc6a014bc28f8549f87957ce2a9cc06c05b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-03 13:06:22.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5f25674e68d24e57a25634b49ff177d72045cf3c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-10-06 20:34:17.000000000","message":"Patch Set 3: Published edit on patch set 2.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":3},{"id":"7ba4929ea56438f057781468fdb5df3052a407e9","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-10-06 20:34:25.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"72e6f462694bf8e44749f5cb0271d0eab75bc2ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-10-08 20:44:11.000000000","message":"Patch Set 4: Patch Set 3 was rebased","accounts_in_message":[],"_revision_number":4},{"id":"8644a834797a1a91ac8a7aa8bd6f894892848977","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-08 21:40:55.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c872bd31819a4dc3be8918057e14a0e2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c20c4e7deedf497c8174a0553d252d22 : SUCCESS in 4m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cbf270605884bbd9ae691e04b38400c : SUCCESS in 2m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c30eecfb33074bf6944d31136a1823aa : SUCCESS in 2m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4a14163632a744848d3126481a47206d : SUCCESS in 2m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c99f0907ef94f57b863b13791f7254a : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d5832e44a0134d2b97e7fba875acba68 : SUCCESS in 4m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f01af1382b01435db952161724319b6f : SUCCESS in 8m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/06b5b7ede8544703ad6a56e874b64c30 : SUCCESS in 3m 36s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/445b6aaff94944a68429e0939d4897c4 : SUCCESS in 8m 58s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3f7bff36beb44d278b0948a936353ee8 : SUCCESS in 31m 25s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6acb91ee17e94c58bad79ef8b0b4988b : SUCCESS in 46m 53s","accounts_in_message":[],"_revision_number":4},{"id":"804625146cb33290b9797a8df0c3a16cb6d1d830","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-10-10 07:30:25.000000000","message":"Patch Set 4: Code-Review+2","accounts_in_message":[],"_revision_number":4},{"id":"479599944dc111b564f2d677cad37d97c088836c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-10-10 13:04:33.000000000","message":"Patch Set 5: Published edit on patch set 4.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"770223b7a7d5ad6bfcc726ea6c893c2d4b1d6bf8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 14:13:02.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cc24425736ab4d598348a2e0d3a8702c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2adef4830ed44c0a9703c133691fe711 : SUCCESS in 3m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d61d30fd964d46f09a36c1dd0ba30502 : SUCCESS in 2m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/93260cfe76dc457db94e32c470c83909 : SUCCESS in 3m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6130e97c05c943fa8767cc061f2f713f : SUCCESS in 3m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/29955d25f5fb48ccbe830cbb310abd24 : SUCCESS in 3m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cab1daa33c894fbf82f535a225a8afe1 : SUCCESS in 2m 52s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/aef15b52a575485bba1bc0c056d0f86a : SUCCESS in 6m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73e4f84af9ea47b092794570bbda4c89 : SUCCESS in 3m 55s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2276a4f378c8416580330964d35bcba6 : SUCCESS in 8m 38s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1975229e17d64f948ec48a7013648290 : SUCCESS in 54m 10s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/54313df9a0104c078efca77e3552f18e : SUCCESS in 1h 02m 46s","accounts_in_message":[],"_revision_number":5},{"id":"6daa8a9ad1f09c7c12c1ebc0d60009b992279f15","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-10-30 17:32:12.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0da934b3bb994ca747e38283592287381dd36f47","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-11-03 14:13:43.000000000","message":"Patch Set 6: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"21d6af13df57f122bbd3c6d637588fd3836158d7","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-11-03 14:13:54.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4d71bc2b446116406db31e2efb197d6847dec4f1","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-11-03 14:13:58.000000000","message":"Topic set to ironic-week-prio","accounts_in_message":[],"_revision_number":6},{"id":"466c7fbfe4549bd92cc1a19687d6cdb8f7801bec","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-11-06 14:23:34.000000000","message":"Patch Set 6: Code-Review+2","accounts_in_message":[],"_revision_number":6},{"id":"296bec4d22ab6f780bdf66d490771ae85bc53234","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-11-06 19:41:42.000000000","message":"Patch Set 7: Patch Set 6 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":7},{"id":"905d70d3b64f5397923a3fb89847a0c8ad3210a5","author":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"date":"2023-11-06 19:43:31.000000000","message":"Patch Set 7: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":7},{"id":"b0308620274dd44ddb4ccee21606d9b4d10950d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 20:58:32.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1a65c6dbdee040e797295d643dfcea58\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9fa8ac0e41f4e129a7f7237bb08a399 : POST_FAILURE in 8m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7376e6908e1644ecbfb84b2e375a5e1e : SUCCESS in 2m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/464d53bc1dae4319bae1f6d565498b12 : SUCCESS in 2m 55s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9f26bafc3e1c4baeb74241964097d5ae : SUCCESS in 3m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c29e4ee0837f4e5dbbcc87fd02504daf : SUCCESS in 5m 47s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b73f53318ae147efad60d7c4ad98778a : SUCCESS in 3m 05s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/2e6db9307ef64b0694e2d8957e5131d0 : SUCCESS in 9m 26s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6e5f3acb67994cbc99274b4d6d4c07ef : POST_FAILURE in 39m 43s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/60dbc2517e44473cbfdef5bae57a360f : SUCCESS in 1h 07m 52s","accounts_in_message":[],"_revision_number":7},{"id":"7b338a3b7ddcf58db83ccb4ed4a33cd75e2e8ef0","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-11-09 18:30:13.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":7},{"id":"b108951fdcb8c5be300d687e0d9fec146b882618","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-11-09 18:31:02.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"dda113cf437a698552700d0c57c92fcd8ecabf76","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 19:34:16.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eeeca857bbac4545a93ff2813704e39f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/018eb068c5eb408fa53b4e50a0816356 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb31a113569c405c88562fa90097a21c : SUCCESS in 2m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/63c1ab1743e74611ad23b579edfff377 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ec2a242d50e948958a4587efc86d97b2 : SUCCESS in 3m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb854769dfff44ca8ffb63cebbbd71aa : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d7884e6128114e0fbe71d866e97f2c7c : SUCCESS in 3m 02s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/1bb7777c78654591af384160a454bf68 : SUCCESS in 10m 00s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a3ff361cb63b4c94827c8bb9b8f568e6 : SUCCESS in 47m 49s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/262fc5ef479142ce940f092a2b453c5d : SUCCESS in 57m 17s","accounts_in_message":[],"_revision_number":7},{"id":"9007c886dd9ea6cad27817c5804e2c6cb14dddde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2023-11-22 14:27:04.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n* Workflow+1\n","accounts_in_message":[],"_revision_number":8},{"id":"9cdda33b3e81078e64f22f3cefe490193142e72c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2023-11-22 14:38:04.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":8},{"id":"1edfcd283020ec89997bb01967c91f0bcc430819","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 15:22:45.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8dfd10e10d814964aae3030552aaf03d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fb57932937bc48abacaee15e280e8bf5 : SUCCESS in 3m 44s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c24d90a7a7c04f0683002ab36c89aec9 : SUCCESS in 2m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e5204e012209419ba52345ca063ada7c : SUCCESS in 3m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1d1ff7dcee64f0bb55d8c923169cfe9 : SUCCESS in 2m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/88c7abeae18e4dd995b4f70eb8a314d3 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8b3da589184047c0aa1a379cc0301692 : SUCCESS in 3m 23s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/853850083d074d7eba34779f6e9f48a0 : SUCCESS in 8m 12s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/48a35cd379f74a09baa0bf80d3fe766f : SUCCESS in 27m 18s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/0f29df8d19c14b9b81cad9a0b695e38f : SUCCESS in 50m 05s","accounts_in_message":[],"_revision_number":8},{"id":"cabec0f057af6144ea5bca23e77ffeb7294813f7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 15:23:19.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"bbd45e414083584d97febba9537c344a90f7a017","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 16:29:45.000000000","message":"Patch Set 8: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/07a45fb9dd0644c38463fb3d7d003da2\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a16ef1a96f4249689946fd486a92790c : SUCCESS in 2m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aced7efef5a44956a32e836643ba6e9e : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dad1c07b7037406388bc50c305bc7de0 : SUCCESS in 3m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5fde7cdd43974c4082d879eadc183133 : SUCCESS in 5m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/717ce41a787d4a45830a17983511b70b : SUCCESS in 3m 25s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/ec70058ec3cc44c491e1700d44b2a39e : SUCCESS in 9m 18s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c057f340ec4048b2974a4a01e2a6f33c : SUCCESS in 44m 56s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/faa3521bfe78423388c82028c27ed7be : SUCCESS in 1h 01m 02s","accounts_in_message":[],"_revision_number":8},{"id":"aa252ec1d23f8c3e9db9051f33fa523d9b8196cf","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 16:29:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":8},{"id":"7fb0cdb096f124e68b958407500170f977ab1cd5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 16:30:51.000000000","message":"Patch Set 8:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4a8261f0007248b0853925d27cc9df23\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/85a1e181bae046388077c6a1351cc1e3 : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b2f2f2d99476401183f6066b7387119e : SUCCESS in 46s","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"3c49a85c5fe60ed1494ac5e7e263d50e1bd62609","revisions":{"0cb9b9e2c21f300d85776e7ab0c44ff6ba2fa79b":{"kind":"REWORK","_number":1,"created":"2023-09-28 21:00:30.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/896835/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/1"}}},"commit":{"parents":[{"commit":"591eb98e27649ee7229a8cb1955e60545e31838d","subject":"Adds an option for setting the http boot uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/591eb98e27649ee7229a8cb1955e60545e31838d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\nvalue ``BootProgress``, which allows for API consumers to determine\nthe current status, when booted in UEFI mode.\n\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0cb9b9e2c21f300d85776e7ab0c44ff6ba2fa79b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0cb9b9e2c21f300d85776e7ab0c44ff6ba2fa79b"}]},"branch":"refs/heads/master"},"c522936f2a515f206ce1af479f4f2dbb4d1daa1c":{"kind":"REWORK","_number":2,"created":"2023-09-28 22:05:36.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/896835/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/2"}}},"commit":{"parents":[{"commit":"591eb98e27649ee7229a8cb1955e60545e31838d","subject":"Adds an option for setting the http boot uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/591eb98e27649ee7229a8cb1955e60545e31838d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 22:04:56.000000000","tz":-420},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\n``BootProgress`` field and subvalues, which allows for API\nconsumers to determine the current status.\n\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c522936f2a515f206ce1af479f4f2dbb4d1daa1c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/c522936f2a515f206ce1af479f4f2dbb4d1daa1c"}]},"branch":"refs/heads/master"},"89d076ba0cbd4a5cb6f67827f381893cbb418fb7":{"kind":"REWORK","_number":3,"created":"2023-10-06 20:34:17.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/896835/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/3"}}},"commit":{"parents":[{"commit":"591eb98e27649ee7229a8cb1955e60545e31838d","subject":"Adds an option for setting the http boot uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/591eb98e27649ee7229a8cb1955e60545e31838d"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-10-06 20:34:14.000000000","tz":0},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\n``BootProgress`` field and subvalues, which allows for API\nconsumers to determine the current status.\n\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/89d076ba0cbd4a5cb6f67827f381893cbb418fb7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/89d076ba0cbd4a5cb6f67827f381893cbb418fb7"}]},"branch":"refs/heads/master"},"731ed2c3b44a4a8799177728ce728b958cdb5d2e":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-10-08 20:44:11.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/35/896835/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/4"}}},"commit":{"parents":[{"commit":"9e85db638f7208d883f9573be84c1d6c291c8def","subject":"Adds an option for setting the http boot uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9e85db638f7208d883f9573be84c1d6c291c8def"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-10-08 20:44:11.000000000","tz":0},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\n``BootProgress`` field and subvalues, which allows for API\nconsumers to determine the current status.\n\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/731ed2c3b44a4a8799177728ce728b958cdb5d2e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/731ed2c3b44a4a8799177728ce728b958cdb5d2e"}]},"branch":"refs/heads/master","description":"Rebase"},"626ad28654738900aba806f77a8ba51e01c59277":{"kind":"REWORK","_number":5,"created":"2023-10-10 13:04:33.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/896835/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/5"}}},"commit":{"parents":[{"commit":"9e85db638f7208d883f9573be84c1d6c291c8def","subject":"Adds an option for setting the http boot uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9e85db638f7208d883f9573be84c1d6c291c8def"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-10-10 13:04:32.000000000","tz":0},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\n``BootProgress`` field and subvalues, which allows for API\nconsumers to determine the current status.\n\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/626ad28654738900aba806f77a8ba51e01c59277"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/626ad28654738900aba806f77a8ba51e01c59277"}]},"branch":"refs/heads/master"},"260f230005d9d733283d8a0d5d179394063af84d":{"kind":"NO_CODE_CHANGE","_number":6,"created":"2023-11-03 14:13:43.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/896835/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/6"}}},"commit":{"parents":[{"commit":"9e85db638f7208d883f9573be84c1d6c291c8def","subject":"Adds an option for setting the http boot uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9e85db638f7208d883f9573be84c1d6c291c8def"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-11-03 14:13:43.000000000","tz":0},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\n``BootProgress`` field and subvalues, which allows for API\nconsumers to determine the current status.\n\nRelated-Bug: #2041902\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/260f230005d9d733283d8a0d5d179394063af84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/260f230005d9d733283d8a0d5d179394063af84d"}]},"branch":"refs/heads/master","description":"Edit commit message"},"66c4abfa6e1f857713930a8991136b39025569d2":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2023-11-06 19:41:42.000000000","uploader":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"},"ref":"refs/changes/35/896835/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/7"}}},"commit":{"parents":[{"commit":"b99648cca2683ef557d7e33913a7893ba7faf726","subject":"Adds an option for setting the http boot uri","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b99648cca2683ef557d7e33913a7893ba7faf726"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Steve Baker","email":"sbaker@redhat.com","date":"2023-11-06 19:41:42.000000000","tz":0},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\n``BootProgress`` field and subvalues, which allows for API\nconsumers to determine the current status.\n\nRelated-Bug: #2041902\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/66c4abfa6e1f857713930a8991136b39025569d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/66c4abfa6e1f857713930a8991136b39025569d2"}]},"branch":"refs/heads/master","description":"Rebase"},"3c49a85c5fe60ed1494ac5e7e263d50e1bd62609":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2023-11-22 14:27:04.000000000","uploader":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"ref":"refs/changes/35/896835/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/896835/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/896835/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/896835/8"}}},"commit":{"parents":[{"commit":"b611b93bed0e5076a6ffafffe5226a6cd55019f0","subject":"Merge \"Adds an option for setting the http boot uri\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b611b93bed0e5076a6ffafffe5226a6cd55019f0"}]}],"author":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-09-28 20:57:14.000000000","tz":-420},"committer":{"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","date":"2023-11-22 14:26:59.000000000","tz":-480},"subject":"Add a boot progress indicator","message":"Add a boot progress indicator\n\nAdd the indicator for boot progress, as mapped to the System\n``BootProgress`` field and subvalues, which allows for API\nconsumers to determine the current status.\n\nRelated-Bug: #2041902\nChange-Id: Icaf8d1ba1a46ab5ffbf8a98fc7eff020fdcecfa5\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3c49a85c5fe60ed1494ac5e7e263d50e1bd62609"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3c49a85c5fe60ed1494ac5e7e263d50e1bd62609"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":4571,"name":"Steve Baker","email":"sbaker@redhat.com","username":"steve-stevebaker"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
