)]}'
{"id":"openstack%2Fsushy~906218","triplet_id":"openstack%2Fsushy~master~Ibc76ddbc6f0d8e449e3d9a5a9caa0c9715c759e8","project":"openstack/sushy","branch":"master","topic":"py3.11","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-01-25 16:08:57.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ibc76ddbc6f0d8e449e3d9a5a9caa0c9715c759e8","subject":"Update supported python versions","status":"MERGED","created":"2024-01-21 14:39:02.000000000","updated":"2024-01-25 16:10:03.000000000","submitted":"2024-01-25 16:08:57.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906218-py3.11","meta_rev_id":"787e7c389da9110a4e0481fb9caaf94b7318aa22","_number":906218,"virtual_id_number":906218,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 16:08:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2024-01-25 15:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 12:14:07.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":1,"date":"2024-01-25 15:05:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-21 16:04:54.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 12:14:07.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-01-25 15:05:39.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"}],"messages":[{"id":"39c2ab013462cb262375ef2d618c42b0af432292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 14:39:02.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c7ff5a127bda8819f53c0c7f7e8ba24241ffe213","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-21 14:39:54.000000000","message":"Topic set to py3.11","accounts_in_message":[],"_revision_number":1},{"id":"7491a8579f2ceb47bbe03776ea7d39e13226fa6a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-21 16:04:54.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6a66092f9ad3489a9113d3624dcf6f3d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/fe2c8f0d3c2644968c699b4bf7781ed6 : SUCCESS in 2m 33s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e9b7a12f102b47639ab5d900cf3fd8b4 : SUCCESS in 2m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c4a89577c1f0431bb800915b1c43d471 : SUCCESS in 2m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e32ec0c263d24d91a46dfdd27579e4bc : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ee16853907945ac8aaa9ef7bc437a76 : SUCCESS in 3m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/137d1e96a85948f192b311c6ee1e753f : SUCCESS in 7m 04s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c9a904f308f241f083a170f5e2a9e5cb : SUCCESS in 7m 38s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/bd9aae4b6cc94d8eb8f1b2c4fba0894a : SUCCESS in 45m 33s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4d7291cb28e04ceea8923ae42d4f9d44 : FAILURE in 1h 20m 22s","accounts_in_message":[],"_revision_number":1},{"id":"c0040d87f8838ec994721f460a1285405ae3cac4","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-22 09:15:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"491d8c25e7754160d7d2f3c736a07635ca1df69b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 10:29:15.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47664c1ec3674cf2a8515b0e79ef5513\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0b7048602f4045e2ac0b80284c39ca68 : SUCCESS in 2m 02s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9fe203b1e89f4cbda3ae1b87b6efe068 : SUCCESS in 3m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a05b91f0af0a4e9eb7e724080412fefd : SUCCESS in 2m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffd85295fa5248348c7532d2b057c1b9 : SUCCESS in 3m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f289b96dd7df469693d8fcec26182bcd : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8cb90eeadcdf451e8f7f75d5f0a32ddf : SUCCESS in 5m 39s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/318e59469b574b9ca0d25b9e15e94d6b : SUCCESS in 9m 22s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5101f32383b64c269786aac423c46bd1 : SUCCESS in 1h 04m 07s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/68d173511cd74ebe80492114386a49e8 : FAILURE in 1h 06m 49s","accounts_in_message":[],"_revision_number":1},{"id":"841e1a0a13975f6de90611d9f94658f518da7de1","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-23 03:37:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a286f6496e70526b08ffe1ff0dcd90d8ff6989e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 04:57:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de45ceb59a684ad68545f0959bf8eff7\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/28d2e2f1ac014dd1bbd7a5d5c3c7c8a7 : SUCCESS in 2m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9f8adec3007e4151892af5dbbda926d4 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/349bcc7e263d4e03aa06457c4735db82 : SUCCESS in 2m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/809d8c73b5d846d4a1faf930abd62064 : SUCCESS in 3m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f3f6d75aacc74f6fb816df1a49b1bb3a : SUCCESS in 2m 59s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6e43430ed3cf459e87e8861b7fb85466 : SUCCESS in 6m 01s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/dabb3b660ce64c3cb7cfcd1c4e81fe51 : SUCCESS in 8m 46s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1a46547b0cc1471d862c67bf1e605fbb : SUCCESS in 46m 41s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/746e367206834830ba6836fc2ebfb2f0 : SUCCESS in 58m 33s","accounts_in_message":[],"_revision_number":1},{"id":"c0d986371e7bba946912befa98ee5c76ddc30cc3","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-01-23 12:14:07.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5a558f96c8eca6dc79cb57592587970bf7074fd","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-01-25 15:05:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"87d04600aab98e110086433e31c936fe88070fc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 15:06:12.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f6d64a67ae070b50471934339d261e696e837f37","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 16:08:57.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab09d7a58f1475187f7b54f40ac508b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/125f81b11e7c4bc1acf6cfe9e4d4d5c1 : SUCCESS in 1m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/efa42cf085bd45be80670f50c253a239 : SUCCESS in 2m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/eb86792e19274320aca8471392fbef11 : SUCCESS in 3m 08s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/726e0c9754f14b9a8a876939a97babde : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d3018e827f547e89074832312f8753d : SUCCESS in 7m 20s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8558fc302e65439492988a43720717d5 : SUCCESS in 9m 04s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/77a2a9a0aa5e4ddeb5c79653ea716ad3 : SUCCESS in 37m 25s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/15f5b01d9da54bcb993ad6e57103dda8 : SUCCESS in 1h 01m 07s","accounts_in_message":[],"_revision_number":1},{"id":"882fae23943f48bbcc6f096ccecb23c72a63a1a6","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 16:08:57.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"787e7c389da9110a4e0481fb9caaf94b7318aa22","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 16:10:03.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a9e9c6111294c8ab100fa6d39683aaf\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/201feae3c0e74c25a68330ff8339bb62 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"80fd7a3e627066a9f43f0569f0f4a5dc2258c63f","revisions":{"80fd7a3e627066a9f43f0569f0f4a5dc2258c63f":{"kind":"REWORK","_number":1,"created":"2024-01-21 14:39:02.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/18/906218/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/18/906218/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/906218/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/906218/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/906218/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/18/906218/1"}}},"commit":{"parents":[{"commit":"3745794d43b865cffffba5651a0dc8e578e6379e","subject":"Merge \"Handle a different related properties for missing TransferProtocolType\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3745794d43b865cffffba5651a0dc8e578e6379e"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:38:11.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2024-01-21 14:38:58.000000000","tz":540},"subject":"Update supported python versions","message":"Update supported python versions\n\n... according to the current PTI.\n\nChange-Id: Ibc76ddbc6f0d8e449e3d9a5a9caa0c9715c759e8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/80fd7a3e627066a9f43f0569f0f4a5dc2258c63f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/80fd7a3e627066a9f43f0569f0f4a5dc2258c63f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
