)]}'
{"id":"openstack%2Fsushy~906798","triplet_id":"openstack%2Fsushy~master~Ib439c9873d601bb140bd3f97acb5ee2d551c7063","project":"openstack/sushy","branch":"master","topic":"ironic-codespell","attention_set":{},"removed_from_attention_set":{"36102":{"account":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"last_update":"2024-02-05 18:53:01.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-01-26 23:19:25.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}}},"hashtags":[],"change_id":"Ib439c9873d601bb140bd3f97acb5ee2d551c7063","subject":"[codespell] Adding Tox Target for Codespell","status":"MERGED","created":"2024-01-26 05:06:42.000000000","updated":"2024-02-05 18:54:05.000000000","submitted":"2024-02-05 18:53:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"906798-ironic-codespell","meta_rev_id":"8f5305e0f3add133a6b09dc4d3d3cc42e8dc30af","_number":906798,"virtual_id_number":906798,"owner":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-05 18:53:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-01-26 23:19:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-05 17:30:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-05 17:30:54.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-26 05:07:13.000000000","updated_by":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-01-26 05:56:09.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-05 17:30:54.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"9a9e9a02c46f113010b46a19a778ef697cf5dbd4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-26 05:06:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a195d44c9c99e04bb310fb1877243dd4db1267ee","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-26 05:07:17.000000000","message":"Topic set to ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"88bb86d2b45e9d5f2d9a2693c2502842d71e69ef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-26 05:56:09.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2904e6a33bdc486f914770e3e5e3694f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/6194fc095331424aa2263f9cdaf34c25 : SUCCESS in 3m 04s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/867fcd21dcca4b708e45b81cd70e3c7f : SUCCESS in 5m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d8c85a37858e49d8b5b1087c7a1f46ec : SUCCESS in 3m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0006297d1f3e4e388cee5dc862bd21fb : SUCCESS in 2m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1037f2bf57224d6784b0553dde1883f5 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1d9f82fa85c240bca2cf797b358d3555 : SUCCESS in 5m 26s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/85d1ec4340af4ab3b5b969bf9126647a : SUCCESS in 2m 54s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a271668301814307bd163ea63baca49d : SUCCESS in 9m 46s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/36434ce2e3a34ab590d67e610528da88 : SUCCESS in 39m 03s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/51c874f89b2942a7af0902e695a58140 : SUCCESS in 47m 06s","accounts_in_message":[],"_revision_number":1},{"id":"2ace868e289682ed4f718096318664bf084e6a89","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-01-26 23:19:25.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2877cfe256c4d36dd23811205bef09610499a1c0","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-27 06:29:11.000000000","message":"Topic ironic-week-prio removed","accounts_in_message":[],"_revision_number":1},{"id":"2de5f4ea2750b527d445085928aeb1c909cc6619","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"date":"2024-01-27 06:29:14.000000000","message":"Topic set to ironic-codespell","accounts_in_message":[],"_revision_number":1},{"id":"3d0d854b5c4ad0c1d3b29bde792308dce4c412f4","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-02-05 17:30:54.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"0dbdeca601d3bf4cb6b0da940655f17410dc339d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 17:31:29.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef6b8619ebd482f0f248d2f4106eb81b586a3e96","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 18:53:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16ae5d3a4e754dd5af4ac899dfb235cd\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/326cff14e55240bfba560a671c872562 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/793c03801bb743e0a242d6f9271c3280 : SUCCESS in 3m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e60244ac0ea24919b7f7946c6a782b25 : SUCCESS in 3m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4dfa626b58bb4871bcec875ae7592f4b : SUCCESS in 3m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bdb99e87848d485696765a23df7d3ad2 : SUCCESS in 5m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4961047bd57e43d7ae04c211d79ea293 : SUCCESS in 3m 25s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f55cf40213f2434bb2d8664fee7aec45 : SUCCESS in 7m 34s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/1ea03e9e55384637b684b215d3b9888d : SUCCESS in 33m 44s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/c75fdcf807c248c28052be3bf98c9288 : SUCCESS in 45m 31s","accounts_in_message":[],"_revision_number":1},{"id":"ddffad90119ac6bcdfaa1a55401ee198d2292aa0","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 18:53:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8f5305e0f3add133a6b09dc4d3d3cc42e8dc30af","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 18:54:05.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2e6037775cc044968f1a9be1a0cf2007\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/44d07b02f19249fdade57c01b35ebcb3 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/639cccdab0a54a88878f4311ac23fc66 : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"eeb1158e8bec2d5192f8ccf295141e94aa4f0506","revisions":{"eeb1158e8bec2d5192f8ccf295141e94aa4f0506":{"kind":"REWORK","_number":1,"created":"2024-01-26 05:06:42.000000000","uploader":{"_account_id":36102,"name":"Adam McArthur","display_name":"Adam McArthur","email":"adam@mcaq.me","username":"adamcarthur","status":"PhD Candidate in Computational Imaging. Open Source Software Engineer at G-Research."},"ref":"refs/changes/98/906798/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/98/906798/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/98/906798/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/98/906798/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/98/906798/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/98/906798/1"}}},"commit":{"parents":[{"commit":"a7f94923723716b18beaca6c9b0c103c4461971a","subject":"[codespell] Fixing Spelling Mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a7f94923723716b18beaca6c9b0c103c4461971a"}]}],"author":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-26 05:06:32.000000000","tz":0},"committer":{"name":"Sharpz7","email":"adam.mcarthur62@gmail.com","date":"2024-01-26 05:06:32.000000000","tz":0},"subject":"[codespell] Adding Tox Target for Codespell","message":"[codespell] Adding Tox Target for Codespell\n\nSecond in a series of commits to add Codespell to Ironic Repos. This one adds the command that was used to fix the spelling errors.\n\nFuture Commits will add CI support and potentially a git-blame-ignore-revs file if their are lots of spelling mistakes that could clutter git blame.\n\nChange-Id: Ib439c9873d601bb140bd3f97acb5ee2d551c7063\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eeb1158e8bec2d5192f8ccf295141e94aa4f0506"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/eeb1158e8bec2d5192f8ccf295141e94aa4f0506"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
