)]}'
{"id":"openstack%2Fsushy~907461","triplet_id":"openstack%2Fsushy~master~I3a9e4748c1cd7aad4c88809b9d3652471b7a3e0d","project":"openstack/sushy","branch":"master","topic":"tox-constraints","attention_set":{},"removed_from_attention_set":{"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-02-05 11:49:52.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I3a9e4748c1cd7aad4c88809b9d3652471b7a3e0d","subject":"Force constraints when installing a package during tox test","status":"MERGED","created":"2024-02-01 17:09:05.000000000","updated":"2024-02-05 11:50:57.000000000","submitted":"2024-02-05 11:49:52.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"907461-tox-constraints","meta_rev_id":"3925acca79c1a40803842277005c9ae94b646678","_number":907461,"virtual_id_number":907461,"owner":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-05 11:49:52.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"all":[{"value":2,"date":"2024-02-02 16:08:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-05 11:00:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-02-05 11:00:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-02-01 18:21:41.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-02 16:08:38.000000000","updated_by":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"reviewer":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"state":"REVIEWER"},{"updated":"2024-02-05 11:00:03.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"72cc2efd7910e932327427998dd68964307eae12","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-02-01 17:09:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"16fc697deac98f2854a81ac7390ed82d353b5e5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 18:21:41.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bcc4cd8074184b2b82465e32668a48f3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9033e01356c4b90b71f4b405e078fff : SUCCESS in 5m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c8cddf1c8431469c9301d308fe6ea90b : SUCCESS in 4m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7412b4c70b2b433c88137a4233a6423e : SUCCESS in 3m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ffebb4d4148f45c0bd09a58b0af31930 : SUCCESS in 3m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b364498a760845b9990bbf472602fa79 : SUCCESS in 6m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac0dcda670ba4df8ac4e5dbee02873b5 : SUCCESS in 4m 09s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/ba1c5c3cbf184bcd80cf0804ddeb5176 : SUCCESS in 8m 50s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/756ea04a63d54efaa5ed7cd86e947148 : SUCCESS in 44m 53s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/ef048fd9bed6410090568ad1db427789 : SUCCESS in 1h 06m 03s","accounts_in_message":[],"_revision_number":1},{"id":"c6cb1d5d98bad147588b238124cfa9d073fdee24","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-02-02 08:47:35.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"733bb42d9859acf871fceb1efdf6a2700ce1f6da","author":{"_account_id":11655,"name":"Julia Kreger","email":"juliaashleykreger@gmail.com","username":"jkreger","status":"Flying to the moon with a Jetpack!"},"date":"2024-02-02 16:08:38.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"7d6b678da8f7e3ef838eab9ee2ca653812346237","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-02-05 11:00:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d7b072d1ca9e3210d82e13b99d171d1d9c675ce4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:00:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"631d0134eb48731ed3504666f6f92041759ac3e0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:49:52.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9da063dac7a34d6d8d7b81e523ba9418\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9ad09b8f009449eea279fa7270a9a98d : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e477e673ba1c448081d08cf06d2a573e : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4495e4ef16b8425c8f6d40ab4768e158 : SUCCESS in 3m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/50c7c0be0eb649db987eb7ac0fef640f : SUCCESS in 7m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/287820aa8fa9442a959b7e48b813e6d4 : SUCCESS in 3m 15s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/fe9a71a109b14334b6dad6ef558e1af4 : SUCCESS in 11m 20s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/a3a88ab6b21549cbb5fc068ed5fab5b3 : SUCCESS in 34m 50s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d07e2428e78c40debe4bdc0549f69c2a : SUCCESS in 48m 12s","accounts_in_message":[],"_revision_number":1},{"id":"afb071dab7fc5188394e2659cf5fc765cac14ac2","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:49:52.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3925acca79c1a40803842277005c9ae94b646678","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-05 11:50:57.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a73c663be59f4d73995f09ea9de282d5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/903cde3febda4821848512f1ebdfdea4 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/20e5d18329454380bc660bfbacd8b154 : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5d217b7e8405c46c44523e77e0bc9ae0f9cbfbe2","revisions":{"5d217b7e8405c46c44523e77e0bc9ae0f9cbfbe2":{"kind":"REWORK","_number":1,"created":"2024-02-01 17:09:05.000000000","uploader":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"ref":"refs/changes/61/907461/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/61/907461/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/61/907461/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/61/907461/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/61/907461/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/61/907461/1"}}},"commit":{"parents":[{"commit":"a7f94923723716b18beaca6c9b0c103c4461971a","subject":"[codespell] Fixing Spelling Mistakes","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/a7f94923723716b18beaca6c9b0c103c4461971a"}]}],"author":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-02-01 17:06:26.000000000","tz":60},"committer":{"name":"Riccardo Pittau","email":"elfosardo@gmail.com","date":"2024-02-01 17:06:26.000000000","tz":60},"subject":"Force constraints when installing a package during tox test","message":"Force constraints when installing a package during tox test\n\nThe tox deps option grants installation of single dependencies and\nrequirements, optionally pinned using constraints, before installing\na package, therefore not granting installation of the correct\nconstraint during the package installation.\nTo fix that tox 4.4.0 has introduced the constrain_package_deps\noption [1]\n\n[1] https://tox.wiki/en/4.12.1/faq.html#using-constraint-files\n\nChange-Id: I3a9e4748c1cd7aad4c88809b9d3652471b7a3e0d\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d217b7e8405c46c44523e77e0bc9ae0f9cbfbe2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/5d217b7e8405c46c44523e77e0bc9ae0f9cbfbe2"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
