)]}'
{"id":"openstack%2Fsushy~925862","triplet_id":"openstack%2Fsushy~stable%2F2024.1~I814e01a8487d0432deec32f9653bfd35099ebe69","project":"openstack/sushy","branch":"stable/2024.1","topic":"storage-controller-name-stable/2024.1","attention_set":{},"removed_from_attention_set":{"10239":{"account":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"last_update":"2024-08-07 13:37:42.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I814e01a8487d0432deec32f9653bfd35099ebe69","subject":"Name is not mandatory for Storage/StorageControllers","status":"MERGED","created":"2024-08-07 11:56:11.000000000","updated":"2024-08-07 13:38:46.000000000","submitted":"2024-08-07 13:37:42.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"925862-storage-controller-name-stable/2024.1","cherry_pick_of_change":925853,"cherry_pick_of_patch_set":1,"meta_rev_id":"bd862d02bc3befc5657aa8b682dab2509abc5ab7","_number":925862,"virtual_id_number":925862,"owner":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-08-07 13:37:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2024-08-07 12:05:39.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":1,"date":"2024-08-07 12:05:39.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-08-07 12:05:39.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-08-07 12:49:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b1ec6bd55ea37de1f67a96d5845d0b953f30d1e1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-08-07 11:56:11.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"c27eb6d3c9038d074e97ad187bf5873ca2d8a8b2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-08-07 12:05:39.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3d40388b0b297afce1356528e4171037d34db122","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:49:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a8c7f293ee534fc1bef70ea8c1fd7c34\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/67b4dee03aab4e17bc32398246d35a14 : SUCCESS in 3m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/182a204cfb9d45938330383d04caed7b : SUCCESS in 2m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a038a40c638457bb93d68bcb3d48e6e : SUCCESS in 2m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/30fd6ce74e5b4fa8ab99bd32f5bcd2b3 : SUCCESS in 4m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9f46394ddfc1457d8ef59653bbc3a3b8 : SUCCESS in 5m 42s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/faebdced53444c5f87d44d9348e0c19a : SUCCESS in 2m 35s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d74ab30ed4bb4932a73e907f72c08c19 : SUCCESS in 9m 37s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/79d1f89e87ac454e9ff1af42eae7fc0d : SUCCESS in 46m 33s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/23c37629005743f68fe02a7e6014a91b : SUCCESS in 35m 26s\n- sushy-tox-codespell https://zuul.opendev.org/t/openstack/build/42c463d2fbc24c27af7f2be8bc4aea1e : FAILURE in 3m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a26b0ad4f46a3c109a765c18cf2d431a31868d29","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 12:49:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"dc093064a932f8a19d9d5f415389ea26186cc642","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 13:37:42.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/943cc47858e640108d54fd4eb1e83a1e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eac4cd851b984bd88d3b0bd6d502d19d : SUCCESS in 3m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3b0675731a9c497795bff749ba026f7f : SUCCESS in 5m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/96deca5ccc734a1e8a74dc179941cf0c : SUCCESS in 2m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/be6529e9f7be46ef9fc6e79e16e16400 : SUCCESS in 6m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6081dd1ef08f4b02a4c5ae67ebe79d74 : SUCCESS in 2m 26s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/ae5e4cb322a64b91bef6690811beaf9f : SUCCESS in 9m 04s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/db05968ff1c14d1696a74317043ce87f : SUCCESS in 42m 52s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/1f09928983f6444cae313fecd46a661c : SUCCESS in 39m 56s","accounts_in_message":[],"_revision_number":1},{"id":"bae68863504b879d46d7e562e503ac4beb664b74","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 13:37:42.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"bd862d02bc3befc5657aa8b682dab2509abc5ab7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-08-07 13:38:46.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ee94781264f491687580b5b9f180892\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db24ee77a2494c40a6dcca8c75910479 : SUCCESS in 45s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/93976eea6abf4de3b9639f9f8fbed5cb : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"285498f1aabef53567bda1c7f102c2168b832ea3","revisions":{"285498f1aabef53567bda1c7f102c2168b832ea3":{"kind":"REWORK","_number":1,"created":"2024-08-07 11:56:11.000000000","uploader":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"ref":"refs/changes/62/925862/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/62/925862/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/62/925862/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/62/925862/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/62/925862/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/62/925862/1"}}},"commit":{"parents":[{"commit":"ddd61c764aceca0e490381480954dd36944f660e","subject":"Handle NotAcceptable when Accept-Encoding: identity is not allowed","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ddd61c764aceca0e490381480954dd36944f660e"}]}],"author":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-08-07 09:32:49.000000000","tz":120},"committer":{"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","date":"2024-08-07 11:56:11.000000000","tz":0},"subject":"Name is not mandatory for Storage/StorageControllers","message":"Name is not mandatory for Storage/StorageControllers\n\nUnlike the separate StorageController object, the StorageControllers\nsub-fields do not mandate a name, and it\u0027s actually missing from some\niLO 5 machines.\n\nChange-Id: I814e01a8487d0432deec32f9653bfd35099ebe69\n(cherry picked from commit c589fae847b9da6ccb9801b91673cfae3316c9c1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/285498f1aabef53567bda1c7f102c2168b832ea3"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/285498f1aabef53567bda1c7f102c2168b832ea3"}]},"branch":"refs/heads/stable/2024.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
