)]}'
{"id":"openstack%2Fsushy~931818","triplet_id":"openstack%2Fsushy~master~Ie8e649307d5392ef96e86aa4184a916590e546c3","project":"openstack/sushy","branch":"master","topic":"PUC-491","hashtags":[],"change_id":"Ie8e649307d5392ef96e86aa4184a916590e546c3","subject":"Change-Id: Ibfa97060d477d794a7ef8fe2a9675a658a13e1f6","status":"ABANDONED","created":"2024-10-08 16:09:12.000000000","updated":"2024-10-08 21:33:18.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"8f8474716d8fced03105771a95fd5fffc4ac730a","_number":931818,"virtual_id_number":931818,"owner":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-10-08 17:02:55.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-08 17:02:55.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"0d4abe4924fbe2031ccd85e1305531de57cb16a2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"date":"2024-10-08 16:09:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6a8dfb5359a9ba8d5354934acfcfee7be3f3c6a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-08 17:02:55.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d08656cb65de4834876dd07c00c2db3b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f87b14e3df340e98355cae0908fcf3f : SUCCESS in 2m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/492da6d104b746ee9e4f4e13e465fc4e : SUCCESS in 5m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c976a281db94d4b973f95e5593c143a : SUCCESS in 3m 19s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5db68da59d60433ea5ae6b868b510388 : SUCCESS in 42m 56s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4374d3e845e0405298c724aa131b3bac : SUCCESS in 45m 40s\n- sushy-tox-codespell https://zuul.opendev.org/t/openstack/build/386b5dfffcf34767847b056bbf0db602 : SUCCESS in 3m 41s","accounts_in_message":[],"_revision_number":1},{"id":"8f8474716d8fced03105771a95fd5fffc4ac730a","tag":"autogenerated:gerrit:abandon","author":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"date":"2024-10-08 21:33:18.000000000","message":"Abandoned\n\nMistaken commit","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"8e4edd1e4a81076e62feb1565fb02ced7633a1ef","revisions":{"8e4edd1e4a81076e62feb1565fb02ced7633a1ef":{"kind":"REWORK","_number":1,"created":"2024-10-08 16:09:12.000000000","uploader":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"ref":"refs/changes/18/931818/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/18/931818/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/931818/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/931818/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/18/931818/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/18/931818/1"}}},"commit":{"parents":[{"commit":"4ff0986aba88ee53665ad89d0e462b3cdfd8237e","subject":"Contribute Port interface to Sushy","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4ff0986aba88ee53665ad89d0e462b3cdfd8237e"}]}],"author":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-08 16:04:56.000000000","tz":-300},"committer":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-08 16:06:34.000000000","tz":-300},"subject":"Change-Id: Ibfa97060d477d794a7ef8fe2a9675a658a13e1f6","message":"Change-Id: Ibfa97060d477d794a7ef8fe2a9675a658a13e1f6\n\nChange-Id: Ie8e649307d5392ef96e86aa4184a916590e546c3\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8e4edd1e4a81076e62feb1565fb02ced7633a1ef"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8e4edd1e4a81076e62feb1565fb02ced7633a1ef"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
