)]}'
{"id":"openstack%2Fsushy~932096","triplet_id":"openstack%2Fsushy~master~I9443bc724816b5929e28c6d95f540bdff785850f","project":"openstack/sushy","branch":"master","topic":"add-ports","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2024-11-07 15:01:20.000000000","reason":"Change was submitted"},"37363":{"account":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"last_update":"2024-11-07 15:01:20.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2024-11-06 17:00:32.000000000","reason":"\u003cGERRIT_ACCOUNT_10342\u003e replied on the change","reason_account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2024-10-23 07:30:29.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":["ironic-week-prio"],"change_id":"I9443bc724816b5929e28c6d95f540bdff785850f","subject":"Add Port resource","status":"MERGED","created":"2024-10-10 16:20:21.000000000","updated":"2024-11-07 15:03:04.000000000","submitted":"2024-11-07 15:01:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"932096-add-ports","meta_rev_id":"c95f7a0c23f004a9621fd1893279a348673a0c15","_number":932096,"virtual_id_number":932096,"owner":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-07 15:01:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"recommended":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2024-11-06 17:00:32.000000000","permitted_voting_range":{"min":1,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2024-11-05 08:41:38.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-07 14:02:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-11-07 14:02:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-10-10 17:24:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-10-11 08:48:38.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2024-10-16 22:02:48.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2024-10-22 17:15:20.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"},{"updated":"2024-11-04 11:17:07.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"CC"},{"updated":"2024-11-07 13:50:38.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"CC"},{"updated":"2024-11-07 14:02:17.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"dc91a4ac2a9a960826873fea4b682ef9152a4ee5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"date":"2024-10-10 16:20:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"0601566ed68799c3c51c81bffb89a8e7584eecac","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"date":"2024-10-10 16:42:01.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"682b6444d6d2dfb8399c72104f0fd0a939dd5784","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 17:24:20.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e397c7d822444ce0ada30c8ee38e80c1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cb51a8b0da33475d843f916159189cdc : SUCCESS in 2m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a22def2c0e2a4ba89fd4a86013babef3 : SUCCESS in 2m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3da21146306a4ec08760bdb8babe200d : SUCCESS in 2m 58s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ec946555181a4d529098a3e5747e6fdd : SUCCESS in 2m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8bc5535bcf44fc38a8b9cb12237f763 : SUCCESS in 4m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d840061591734f64bac9078f8aa71af0 : SUCCESS in 3m 28s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/e365f6ad55324195bbd74518f3da1e38 : SUCCESS in 9m 50s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/2a1e38d693244152b1abf9bc2470c587 : SUCCESS in 37m 05s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b858333d9f3f43ab9fea58dd64bb0c4e : SUCCESS in 40m 02s\n- sushy-tox-codespell https://zuul.opendev.org/t/openstack/build/9e218ef0924648e688584a96039dc604 : FAILURE in 3m 26s","accounts_in_message":[],"_revision_number":2},{"id":"0c4f2eaf66c1e6deeba9f4112dc8c910f23ecdd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"date":"2024-10-10 20:28:11.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"ee272804f1bd5361529931894bf09882eb89b17f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-10 21:02:04.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2dfd26814b4d38a899f02af3fe4ba4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/223d6e69be8a46619322586c1b2d65f4 : SUCCESS in 3m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ede079b99eb645febcbde829681874af : SUCCESS in 3m 16s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9b19bd53710e4eca84430790ee76757c : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f9cb675a7d2d4b5c8a55b1bde3fdb072 : SUCCESS in 3m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7a517ead26e40f6b1bd7e6559c4e511 : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9620879c6a9940ba9b4d2afcf2487463 : SUCCESS in 3m 20s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/10a97f060b794d03bf542a4c1f7307e9 : SUCCESS in 6m 10s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/364c65f57d1841aaad77681acb17e952 : SUCCESS in 22m 50s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a26a622644fb4f13bdb9378c29c7fd92 : SUCCESS in 32m 04s\n- sushy-tox-codespell https://zuul.opendev.org/t/openstack/build/b2d43f0ad1954b7c84b4b0afe57ef81c : SUCCESS in 2m 03s","accounts_in_message":[],"_revision_number":3},{"id":"e11c0a810266dd89521a81ea192cfa5253b007e7","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-10-11 08:48:38.000000000","message":"Patch Set 3: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"8960ac92cefbc84401fad5a2eb6cbe1c92ea39ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"date":"2024-10-11 15:27:56.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5ffdb41f46cd1fb7dbc08009f47aea7fca6fed02","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-11 16:22:26.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/16a2c3307e82461ca87107a7e50d48ab\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16c351c51f404132ab69dad9f714a7c4 : SUCCESS in 2m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/07dd8d2a752f4b9c9dd7275cc176a205 : SUCCESS in 2m 26s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5b440510ce8e46ae87ebda85281967e9 : SUCCESS in 2m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f1b0c721b3fc4247a931b1d7ab3a353c : SUCCESS in 2m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/723493b422af4e7aa315e46caec5e601 : SUCCESS in 6m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c1600640fb0a47958e4ecb252f9703d2 : SUCCESS in 2m 28s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/d9667e139fce4a42a9e146942c0ca099 : SUCCESS in 7m 40s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6fd22fc0449d44659d5e22a391bb60a8 : SUCCESS in 37m 46s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/df0e5abc49d2454a8eaf7132e928a6c9 : SUCCESS in 39m 23s\n- sushy-tox-codespell https://zuul.opendev.org/t/openstack/build/788463275ed94735a5f3a1b74c76ef7d : SUCCESS in 2m 20s","accounts_in_message":[],"_revision_number":4},{"id":"6f6bf6fe59fd7fe064ba16a72625478723072334","author":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"date":"2024-10-12 21:49:54.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6418b764e270d3fc1629d1d6aa08ff60c758e318","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-16 22:02:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"ad1c6ef135d3b752f38ed002d1943028c5de53e3","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-16 22:05:33.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"5f49b0a02cd2b0e75be9db29d9761265200b53f6","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-22 17:15:20.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"afb3d0260a08974c1f78a365059b59a735bd110d","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-10-22 17:16:20.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"446d35f72e65aa5c3a80b1eeb25ca56f1490f032","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-10-23 07:30:29.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e609139a4b09557fb561e3633d7f7a60e654cf16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-29 22:38:20.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"fd0eaf221001a69df855a9b7f9b0e1b3668fb3b5","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-29 22:39:43.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"c404e4f432845fb0fa72de836bcbc7d2a2d99d30","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-29 22:52:01.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"e0c3ee3a3b4dfbbe3a1886fd131bc360387a92e0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-29 23:07:14.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5101401800a041caacedda0cd759e6c7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a66c4884f73946aab64a2c52b777d422 : SUCCESS in 2m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d320152c51da484e9fa07f6ca5945b24 : SUCCESS in 2m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4d10e4fe5bd64e8e98f3a40e91f43b75 : SUCCESS in 2m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/3e813c9b619040c4b2a8e87137a5c651 : SUCCESS in 2m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ccd5a32fd0441cabb947009fba9d60a : SUCCESS in 5m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/394d0299ef6946fdb5df404c3ed9fc0e : SUCCESS in 2m 04s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7707117358154df8be2597ea7ccd51d1 : SUCCESS in 9m 16s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c761bedfeeaf42418ae6484232e90170 : SUCCESS in 23m 49s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9e0c6cf211ee4bb2b5303c21d8b85034 : SUCCESS in 28m 10s\n- sushy-tox-codespell https://zuul.opendev.org/t/openstack/build/4a8cb6f2671243c9967b7ffc19ca9e52 : SUCCESS in 1m 56s","accounts_in_message":[],"_revision_number":5},{"id":"e4713b1b520e44c87b73131a530eb0228f09e1e1","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-10-30 15:42:33.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":5},{"id":"f59bc7eb66ff2015c7ee8262fecd007aeed5bc18","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2024-11-05 08:41:38.000000000","message":"Patch Set 5: Code-Review+2","accounts_in_message":[],"_revision_number":5},{"id":"7484d195362ea4f303f7d24ffba0313c856f0ef3","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-06 17:00:32.000000000","message":"Patch Set 5: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"7f43b8208f7bc773b6ac2434e305c51915cc53a2","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-11-07 14:02:17.000000000","message":"Patch Set 5: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":5},{"id":"21ef00ec58f5ad6c94bfc0a63417824b1b8f832f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 14:02:52.000000000","message":"Patch Set 5: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":5},{"id":"e74a09959c3f3fd925150a2447460772422e4e1d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 15:01:20.000000000","message":"Patch Set 5: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3965aaa20f2e417b9ba76a96e1fc5e13\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cab16c492ce14fd794ad97cbd03f188f : SUCCESS in 1m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ba213070ba9c4d669436e77595e42049 : SUCCESS in 2m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/345261b44be64467b3e5d4fec3ee0827 : SUCCESS in 4m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9684eb3daf4044329a771e383f5b5b13 : SUCCESS in 5m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3ee9dd92c358471f9e17ade6691b43fe : SUCCESS in 2m 28s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7f9ff4cf8f2d46f0b20986761f05c99a : SUCCESS in 8m 16s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/894606ed2a664c1980caff2a2a08fc73 : SUCCESS in 40m 08s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/a569c897435e4a8fb9165b689c7c763c : SUCCESS in 47m 01s\n- sushy-tox-codespell https://zuul.opendev.org/t/openstack/build/2c90c6698cfc46cd9cca466b2723bae7 : SUCCESS in 2m 23s","accounts_in_message":[],"_revision_number":5},{"id":"76b88ab77a9124400509bc5436dcc2ab240994b3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 15:01:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":5},{"id":"c95f7a0c23f004a9621fd1893279a348673a0c15","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-07 15:03:04.000000000","message":"Patch Set 5:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b7f42acc5738439ea511647634fe9be9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/515c76f639ec41b3860b9c60cdbe30c4 : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/19772afc7e614cde893409dc4706b062 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":5}],"current_revision_number":5,"current_revision":"ba4a6d1cf3f76df4c5ed554de6eb873201a14db8","revisions":{"13ff82d5774e5ba5b4b7c6813829fb1514934081":{"kind":"REWORK","_number":1,"created":"2024-10-10 16:20:21.000000000","uploader":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"ref":"refs/changes/96/932096/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/96/932096/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/96/932096/1"}}},"commit":{"parents":[{"commit":"04eca76b939cbce34c2af5ff7bd32b58ee2d8361","subject":"contributor: links to ironic dev docs, launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/04eca76b939cbce34c2af5ff7bd32b58ee2d8361"}]}],"author":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 16:06:06.000000000","tz":-300},"committer":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 16:20:19.000000000","tz":-300},"subject":"Add Port interface","message":"Add Port interface\n\n- Add enumeration for port link status\n- Add port collection and port base resources\n- Add tests to exercise port collection and port base resources\n- Rename old reference of \"port\" to \"network port\" with new \"port\"\n  collection base resource\n- Add sample json files for port collection and port base resources\n- Rename \"network_port.py\" file to be replace with new port file for\n  port base resource\n- Correct tests to match reference to network_port and port\n\nChange-Id: I9443bc724816b5929e28c6d95f540bdff785850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/13ff82d5774e5ba5b4b7c6813829fb1514934081"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/13ff82d5774e5ba5b4b7c6813829fb1514934081"}]},"branch":"refs/heads/master"},"b89ac54a3d9d49f3656c3d5329987b7472b3420e":{"kind":"REWORK","_number":2,"created":"2024-10-10 16:42:01.000000000","uploader":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"ref":"refs/changes/96/932096/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/96/932096/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/96/932096/2"}}},"commit":{"parents":[{"commit":"04eca76b939cbce34c2af5ff7bd32b58ee2d8361","subject":"contributor: links to ironic dev docs, launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/04eca76b939cbce34c2af5ff7bd32b58ee2d8361"}]}],"author":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 16:06:06.000000000","tz":-300},"committer":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 16:41:36.000000000","tz":-300},"subject":"Add Port interface","message":"Add Port interface\n\n- Add enumeration for port link status\n- Add port collection and port base resources\n- Add tests to exercise port collection and port base resources\n- Rename old reference of \"port\" to \"network port\" with new \"port\"\n  collection base resource\n- Add sample json files for port collection and port base resources\n- Rename \"network_port.py\" file to be replace with new port file for\n  port base resource\n- Correct tests to match reference to network_port and port\n\nChange-Id: I9443bc724816b5929e28c6d95f540bdff785850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b89ac54a3d9d49f3656c3d5329987b7472b3420e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b89ac54a3d9d49f3656c3d5329987b7472b3420e"}]},"branch":"refs/heads/master"},"79bbad6d81c0e15a918c0fd1734ab2259f239aed":{"kind":"REWORK","_number":3,"created":"2024-10-10 20:28:11.000000000","uploader":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"ref":"refs/changes/96/932096/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/96/932096/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/96/932096/3"}}},"commit":{"parents":[{"commit":"04eca76b939cbce34c2af5ff7bd32b58ee2d8361","subject":"contributor: links to ironic dev docs, launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/04eca76b939cbce34c2af5ff7bd32b58ee2d8361"}]}],"author":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 16:06:06.000000000","tz":-300},"committer":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 20:27:52.000000000","tz":-300},"subject":"Add Port interface","message":"Add Port interface\n\n- Add enumeration for port link status\n- Add port collection and port base resources\n- Add tests to exercise port collection and port base resources\n- Rename old reference of \"port\" to \"network port\" with new \"port\"\n  collection base resource\n- Add sample json files for port collection and port base resources\n- Rename \"network_port.py\" file to be replace with new port file for\n  port base resource\n- Correct tests to match reference to network_port and port\n\nChange-Id: I9443bc724816b5929e28c6d95f540bdff785850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/79bbad6d81c0e15a918c0fd1734ab2259f239aed"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/79bbad6d81c0e15a918c0fd1734ab2259f239aed"}]},"branch":"refs/heads/master"},"ecdc7e70edfd582829095766e2f469fa82035d2d":{"kind":"REWORK","_number":4,"created":"2024-10-11 15:27:56.000000000","uploader":{"_account_id":37363,"name":"Scott Tran","email":"transt@gmail.com","username":"transt"},"ref":"refs/changes/96/932096/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/96/932096/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/96/932096/4"}}},"commit":{"parents":[{"commit":"04eca76b939cbce34c2af5ff7bd32b58ee2d8361","subject":"contributor: links to ironic dev docs, launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/04eca76b939cbce34c2af5ff7bd32b58ee2d8361"}]}],"author":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 16:06:06.000000000","tz":-300},"committer":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-11 15:27:39.000000000","tz":-300},"subject":"Add Port interface","message":"Add Port interface\n\n- Add enumeration for port link status\n- Add port collection and port base resources\n- Add tests to exercise port collection and port base resources\n- Rename old reference of \"port\" to \"network port\" with new \"port\"\n  collection base resource\n- Add sample json files for port collection and port base resources\n- Rename \"network_port.py\" file to be replace with new port file for\n  port base resource\n- Correct tests to match reference to network_port and port\n\nChange-Id: I9443bc724816b5929e28c6d95f540bdff785850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ecdc7e70edfd582829095766e2f469fa82035d2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ecdc7e70edfd582829095766e2f469fa82035d2d"}]},"branch":"refs/heads/master"},"ba4a6d1cf3f76df4c5ed554de6eb873201a14db8":{"kind":"REWORK","_number":5,"created":"2024-10-29 22:38:20.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/96/932096/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/96/932096/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/96/932096/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/96/932096/5"}}},"commit":{"parents":[{"commit":"04eca76b939cbce34c2af5ff7bd32b58ee2d8361","subject":"contributor: links to ironic dev docs, launchpad","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/04eca76b939cbce34c2af5ff7bd32b58ee2d8361"}]}],"author":{"name":"Scott Tran","email":"transt@gmail.com","date":"2024-10-10 16:06:06.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-10-29 22:37:05.000000000","tz":-300},"subject":"Add Port resource","message":"Add Port resource\n\n- Add Port and PortCollection to system as they are any physical port\n- Add enumeration for port link status\n- Add tests to exercise PortCollection and Port resources\n- Add sample json files for PortCollection and Port resources\n\nChange-Id: I9443bc724816b5929e28c6d95f540bdff785850f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ba4a6d1cf3f76df4c5ed554de6eb873201a14db8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/ba4a6d1cf3f76df4c5ed554de6eb873201a14db8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
