)]}'
{"id":"openstack%2Fsushy~934913","triplet_id":"openstack%2Fsushy~master~Iec4b628a5d5425dd18b5a4fd76f61c504e3d53b0","project":"openstack/sushy","branch":"master","topic":"enable-ruff-checks","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2024-11-25 23:51:09.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"Iec4b628a5d5425dd18b5a4fd76f61c504e3d53b0","subject":"enable ruff with flake8-bandit rules","status":"MERGED","created":"2024-11-12 23:47:08.000000000","updated":"2024-11-25 23:52:07.000000000","submitted":"2024-11-25 23:51:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"934913-enable-ruff-checks","meta_rev_id":"07ff7e22ad04f7855ab22e362fd0a7a546dad5cd","_number":934913,"virtual_id_number":934913,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-11-25 23:51:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2024-11-25 22:13:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-11-25 16:56:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":1,"date":"2024-11-25 22:13:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-11-13 00:30:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-11-25 16:56:29.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"},{"updated":"2024-11-25 22:13:14.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"6b6f84e0b7c2513d6909ea3b983c87f1be87a4c8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-11-12 23:47:08.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b69281a7316553ac5976d79c90dae641bced2661","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-13 00:30:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/58ed32878c304d3eadd5a2d4d102556b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/252c4be435ae47489b8358a689c51969 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7f86731831f497b923bc5b9ec1098fb : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ae5e86c682904f2ba91c5f6f7f2823ce : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2bad76cc7278458f83529aac27dac133 : SUCCESS in 3m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f41d6037cbb40f681c322ee7d72a086 : SUCCESS in 5m 44s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/64c720267b094b5383f02f575e6f695e : SUCCESS in 8m 58s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/db25f9243fdd4302aafe71fdd833255f : SUCCESS in 42m 15s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6b4f09cc423341b58a459f9d74be9deb : SUCCESS in 31m 47s","accounts_in_message":[],"_revision_number":1},{"id":"8fa5e47d10e901cb531919ee61a87bd8ae72f3f0","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2024-11-20 21:57:46.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"505022999533a19278e9118c388b40ca8b20f381","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2024-11-25 16:56:29.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b5b7a699c76d9daa185b07b4d921f8b501c8cbf9","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2024-11-25 22:13:14.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"370dd051dacfa5ad761cb1bf2165cc6b44db1adf","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 22:57:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b74a3e26916abad0a0ec5dbfe8e2d9e0efbb4050","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:51:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fe2ac37257fd4a51a0a8b6f8185c47a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fbf38d9484134edebc9421eb1c3400d7 : SUCCESS in 3m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1de2160a127048c5a8c9bcaf7d862682 : SUCCESS in 2m 31s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/421a6c459ca04d0c9fbb723610482e40 : SUCCESS in 2m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2dbf592d03fb466d93b00c0648ffe3b2 : SUCCESS in 5m 12s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/607e9d74f89749f3957c754811115c57 : SUCCESS in 6m 36s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3943da97c587406aab21b7f37c7734a7 : SUCCESS in 48m 47s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4ef2abfe15bf464f98478770804969c1 : SUCCESS in 33m 49s","accounts_in_message":[],"_revision_number":1},{"id":"b6c1f5f596afac7a0b7c9e1d5af1438a55ac10d5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:51:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"07ff7e22ad04f7855ab22e362fd0a7a546dad5cd","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-25 23:52:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b053c1a3d96e4dbb86a11f213ffa5450\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8036e925af52402db74597c5f732e5ec : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"83293508d8e45e136c7dad488990c42054c5c928","revisions":{"83293508d8e45e136c7dad488990c42054c5c928":{"kind":"REWORK","_number":1,"created":"2024-11-12 23:47:08.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/13/934913/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/13/934913/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/934913/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/934913/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/13/934913/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/13/934913/1"}}},"commit":{"parents":[{"commit":"43fdf124f87d1d94882bea6dc47bca5605172a7f","subject":"ensure all requests calls have a timeout","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/43fdf124f87d1d94882bea6dc47bca5605172a7f"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-11-09 03:42:47.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2024-11-12 23:43:34.000000000","tz":-360},"subject":"enable ruff with flake8-bandit rules","message":"enable ruff with flake8-bandit rules\n\nEnable flake8-bandit checks via the ruff checker. The flake8-bandit\nrules are prefixed with S instead of B like the bandit rules are but the\nnumberical values are different. flake8-bandit provides the default\nconfig without needing it to be generated like bandit has in the ironic\nrepo. The only exceptions to the default config in ironic is the\nskipping of 3 rules which is instead skipped just for tests instead of\nglobally. The choice of ruff is intentional as other OpenStack projects\nare trending that way and more checks can be used in the future.\n\nChange-Id: Iec4b628a5d5425dd18b5a4fd76f61c504e3d53b0\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/83293508d8e45e136c7dad488990c42054c5c928"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/83293508d8e45e136c7dad488990c42054c5c928"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
