)]}'
{"id":"openstack%2Fsushy~940557","triplet_id":"openstack%2Fsushy~master~Ic7ebcf39494b2f224bb8350c0df611f742fd6c66","project":"openstack/sushy","branch":"master","topic":"merge-sushy-oem-idrac","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-03-20 13:36:45.000000000","reason":"Change was submitted"},"10342":{"account":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"last_update":"2025-03-20 13:36:45.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-02-24 09:10:06.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":[],"change_id":"Ic7ebcf39494b2f224bb8350c0df611f742fd6c66","subject":"migrate sushy_oem_idrac to sushy","status":"MERGED","created":"2025-01-31 20:18:33.000000000","updated":"2025-03-20 13:38:00.000000000","submitted":"2025-03-20 13:36:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":8,"unresolved_comment_count":3,"has_review_started":true,"submission_id":"940557-merge-sushy-oem-idrac","meta_rev_id":"6c7402a0ae2cb15f6a1cca2db23e2de132844edf","_number":940557,"virtual_id_number":940557,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-03-20 13:36:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":2,"date":"2025-02-24 15:21:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-03-20 12:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2025-03-20 12:29:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-01-31 20:50:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 15:39:40.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-02-19 20:17:16.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"CC"},{"updated":"2025-03-20 12:29:50.000000000","updated_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"reviewer":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"state":"REVIEWER"}],"messages":[{"id":"dcd2b5695ec4978fe7554669c47852fbb18bc2c4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-01-31 20:18:33.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"38cf8831bb8a1f397d028e495353d3c6238f5c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-01-31 20:50:16.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/f2d5cdc69d6a4d1093fceeacbdbfe782\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/f9408fe26ab04844848b2939aedd9f1e : SUCCESS in 2m 15s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/58b898c2c22c4d7abf57430dee948de0 : FAILURE in 3m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0edf099f523c4e68a8515a4012c755f3 : SUCCESS in 3m 29s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cfebc173faa49fea59d4d51402d0f47 : FAILURE in 3m 01s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/150e21c43e4f4eae9958d713baf3f222 : FAILURE in 3m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ddae009b4a544e6b5bcbe5f84ce0749 : SUCCESS in 6m 14s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/34a4a5251f284ae08882a7909906ad4f : SUCCESS in 7m 04s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/9cfbc04d059d4f5a8a2608754a4e2eb7 : SUCCESS in 30m 07s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d02654a7a1ff4486bbe3932e4757004b : SUCCESS in 30m 32s","accounts_in_message":[],"_revision_number":1},{"id":"e2b779ef00bd93afe1bbb20138611e4dd77d139c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-18 19:35:34.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2a0f3ec889a18faced748e93d803d590a376be45","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-18 19:36:17.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.","accounts_in_message":[],"_revision_number":3},{"id":"060a6dea10933321a11c69aacfa52e94ef486294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-18 20:30:42.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/b64704b8c400401591e96a83d98363bb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/050152ba232a440db049a36621f31e58 : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d2571485a0d14f1da8b35b4558e2aa2b : FAILURE in 4m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a930907edba04e568a24ed95a445ec42 : SUCCESS in 4m 07s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4050bea0021545548217ab507741b82b : FAILURE in 3m 05s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b65f6b1075ab450ea362fa2e5864f692 : FAILURE in 3m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90a5ff670fa4474fab350ede2ca9a13e : SUCCESS in 6m 33s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/c11dc0b842de418a937786c32e6bfb95 : SUCCESS in 8m 00s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/33c0b060516e4cfea06f70c4ac421b21 : SUCCESS in 38m 54s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8f79622f01dd4654825d101899d15ba4 : SUCCESS in 34m 39s","accounts_in_message":[],"_revision_number":3},{"id":"1ff25cdeb0f02dd74c66c9e20b4a6f622ac14719","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-18 23:08:50.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fcac603799f69ef475efb16ff1e10e71f4deb4d0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-18 23:20:53.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"15df8c50650dc80bc65b8fbfbd9b55e731041a3a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-18 23:36:48.000000000","message":"Uploaded patch set 6: Patch Set 5 was rebased.","accounts_in_message":[],"_revision_number":6},{"id":"f4c457ae7b78a66039ece503a25aa17dc9a37054","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 00:10:08.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/d9a78b6636914fd5a603a7a269abeed8\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/50d821d18e254190b1e06d66539b962e : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/538828540fa246d88bd7951f0a73148c : FAILURE in 2m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e0855bc243d4d918e04fe6a15dfccf5 : FAILURE in 3m 44s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59cdbf55f7d4428584d6d0dc9bd4ee24 : FAILURE in 2m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/32d00928575e40f8b9c31fbca5743a25 : FAILURE in 3m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce348b66f5cc46fe9847590fb4a43efa : SUCCESS in 3m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d78a85ab95e54672ad8ece54e9ff2e6b : SUCCESS in 2m 20s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/13d1458f0e32465d9e02f2581b50fbd0 : SUCCESS in 8m 31s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e89511e693054656b223a6fd6cf817d9 : SUCCESS in 30m 48s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b25522e512254727801121d7a6efa36a : SUCCESS in 32m 03s\n\nWarning:\n  Comments left for invalid file sushy/tests/unit/resources/oem/test_common.py","accounts_in_message":[],"_revision_number":6},{"id":"da6a28a51c09f26de27c5615708e4fe70b8115a3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-19 05:16:00.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"c4c71827632ce77fcad24259f48a327285ffad2c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 06:08:56.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d41288e5f9bc4db09ced391930ea014c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c891428a76f9442bb88d488b6c68f736 : SUCCESS in 4m 41s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/41e7398d3fef4e4e9d4eca7ed679b5d4 : SUCCESS in 3m 41s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0bbfd89428114e3c878e7deaf022ea38 : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/023dc476e6a94893bfbe2ead5a2e6047 : SUCCESS in 2m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/40c5f0e283594c51b7e6c4256770c00e : SUCCESS in 2m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b34c30cc671d4cfe9bb40cc2265faf80 : SUCCESS in 6m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6000d3f4f6464f5c8d89e8834562073f : SUCCESS in 2m 46s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/107923421db947f78c9917987dde608a : SUCCESS in 6m 45s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/158f336590514ea2bd58fd0bab658564 : SUCCESS in 25m 47s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/44cd0b121f6545298c083ad7ce49b73e : SUCCESS in 28m 42s","accounts_in_message":[],"_revision_number":7},{"id":"db9456cfa3237ee116b66c2acf98c761b82389cb","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-19 15:39:40.000000000","message":"Patch Set 7: Code-Review+2","accounts_in_message":[],"_revision_number":7},{"id":"7015c3bc116a76e3a4ce4f9c2e1337e4aa252191","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-02-19 20:17:16.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"745009992119f1c9baa0ed8a48f6dc34cdae0b09","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-19 20:21:01.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"8fcec96c7232b2a6ec3847c70b5e0116ffc6bbba","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-20 09:46:42.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"15e1aa1bf8a2cf2dc576e3b4a76de2c94efe8c23","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-22 17:45:21.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"0e1a778028438a280765fe64dc4372a3d85f6339","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 17:57:46.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/e7fb88a783564db4a1af0f5e7eb85db5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/016a4c45b08d4badb078d8e63ab033ec : SUCCESS in 2m 17s\n- requirements-check https://zuul.opendev.org/t/openstack/build/edc549e8305341e19cf360ca0195c4a7 : FAILURE in 2m 08s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/df869be846a1464a9b10502a623c1b3b : FAILURE in 2m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad825580247541e799f33df0534ba4f5 : FAILURE in 2m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cb9362d1e588459c9415462a7f8ae503 : FAILURE in 1m 42s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/242ffca2141941dcbf68036c4e827e57 : FAILURE in 2m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3e420f97eeaf408ab374ecf2ae326d4a : FAILURE in 2m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3fc864343bb4451b95faef380309e244 : SUCCESS in 2m 22s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/7bb5d2a131254556ac230e7a0f588564 : FAILURE in 2m 58s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/44113d81a4d143d5b3178b5a2572d15c : POST_FAILURE in 11m 22s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9514f6a60c47434faa0ec7ae981aca1c : POST_FAILURE in 11m 10s","accounts_in_message":[],"_revision_number":8},{"id":"7c7fb785f9b860b6ab4f86ab4ba92b60cd26bf6f","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-22 19:02:37.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":9},{"id":"1b3cb8376091c124a70a5091ba284c50318c67ac","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-22 19:02:59.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"b6ac34b49f29736cd5a44d8f4b806dcd4938ab09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-22 19:56:41.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2cc9fc3b4f44d4693253161fa265f24\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/748bccafe7ae4097ae6d0433cd4673ad : SUCCESS in 4m 48s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/041dde36b3244ea6adc1c1f20d3faa8a : SUCCESS in 3m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d4f8366a4a2e4d5ab32ce213292ca46e : SUCCESS in 7m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cd27e92fc734d2990cc838c03c89f3f : SUCCESS in 2m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f38148e701934d58b0066e1899d2569e : SUCCESS in 3m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b995f0a58c34022856857e0264c3544 : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5bbd76318cdb4f0485777d98f7412341 : SUCCESS in 2m 12s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f6281b8a56f34c58b374bc32a9f8c73b : SUCCESS in 7m 45s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/377527cc01124f2fa3dc8f47894b525c : SUCCESS in 42m 02s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8795377be5cc42e59de10aef75eae048 : SUCCESS in 48m 06s","accounts_in_message":[],"_revision_number":9},{"id":"dbc874240f0af47f9ab11378a24dfc7dbec6d21c","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-24 09:10:06.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"444f707499d5cfb52dbd002651c2f1142f561d0b","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-24 15:21:30.000000000","message":"Patch Set 9: Code-Review+2","accounts_in_message":[],"_revision_number":9},{"id":"ea5df66b179ca4cbcb99c6a5360c2f06639b4691","author":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"},"date":"2025-03-20 12:29:50.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1\n\n(3 comments)","accounts_in_message":[],"_revision_number":9},{"id":"7827006eb438e76d16954e3bb1129175f8d2792d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 12:30:26.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"4539290222615a478a50bb45ed918a37e4b40a14","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 13:36:45.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4586d3222c84468a948f44a973b9d8f5\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/99351f59daaf438c9b24bd11c0dfde43 : SUCCESS in 2m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9fedebb2fee44c38bfefc90e2600060f : SUCCESS in 2m 20s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a4399cd60a0d4dccae5acc32fe4f41f0 : SUCCESS in 3m 00s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/206c336f1f824d17b6f01311eda99ec4 : SUCCESS in 2m 00s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2b369da746104df288d84b2700eb276a : SUCCESS in 7m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/99fc179750894d73b9c769f76e73ecd2 : SUCCESS in 1m 37s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/0c1962343c744baf9984d474ed8aba1b : SUCCESS in 7m 51s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/55a808aaa2eb4374a4d4ce5d14954707 : SUCCESS in 33m 55s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8842cfc683894c3f94a02ce719e3d6fd : SUCCESS in 1h 02m 05s","accounts_in_message":[],"_revision_number":9},{"id":"56bad64763eeb9263f0cb187b7164fb9853fb865","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 13:36:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"6c7402a0ae2cb15f6a1cca2db23e2de132844edf","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-03-20 13:38:00.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/71efff14b3e74fd789f72539ed0f0680\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2041912f7ff14246b1bd2e2a6c2a4acf : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ac9f663ecd2a4a48a7f37527d0a0ab53 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"da092394650e952d1991f24f13403b5e020567d9","revisions":{"8b2010e3b0fd7f017caaa68d985a4110987204d8":{"kind":"REWORK","_number":1,"created":"2025-01-31 20:18:33.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/1"}}},"commit":{"parents":[{"commit":"52d9078836ebd9ea0df2fc2afe228aa4f48f3f8c","subject":"enable pyupgrade via ruff to Python 3.7","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/52d9078836ebd9ea0df2fc2afe228aa4f48f3f8c"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:15:10.000000000","tz":-300},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8b2010e3b0fd7f017caaa68d985a4110987204d8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8b2010e3b0fd7f017caaa68d985a4110987204d8"}]},"branch":"refs/heads/master"},"8d15afc859dab4a2a78058271ba23a3af4d33518":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2025-02-18 19:35:34.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/2"}}},"commit":{"parents":[{"commit":"abf9898c7189633c4d4010c6237f8ab8539d1b84","subject":"fix extension loading test to use correct namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/abf9898c7189633c4d4010c6237f8ab8539d1b84"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-18 19:35:29.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8d15afc859dab4a2a78058271ba23a3af4d33518"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8d15afc859dab4a2a78058271ba23a3af4d33518"}]},"branch":"refs/heads/master"},"6f0384a2af198ea6c10693ee00a936a2a20d1f8b":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2025-02-18 19:36:17.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/3"}}},"commit":{"parents":[{"commit":"87d96ea9b111f618599562dad430cd48ada2d7d5","subject":"fix extension loading test to use correct namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/87d96ea9b111f618599562dad430cd48ada2d7d5"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-18 19:36:12.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6f0384a2af198ea6c10693ee00a936a2a20d1f8b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6f0384a2af198ea6c10693ee00a936a2a20d1f8b"}]},"branch":"refs/heads/master"},"21818cd30b354a326c4177f131016735cd8f05b0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2025-02-18 23:08:50.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/4"}}},"commit":{"parents":[{"commit":"cd75cb53ad923d2f0385de66e620a999dc76f391","subject":"fix extension loading test to use correct namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd75cb53ad923d2f0385de66e620a999dc76f391"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-18 23:08:46.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/21818cd30b354a326c4177f131016735cd8f05b0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/21818cd30b354a326c4177f131016735cd8f05b0"}]},"branch":"refs/heads/master"},"b13ae11f9f10641b4944ca2db3a8cc78f33cbb34":{"kind":"REWORK","_number":5,"created":"2025-02-18 23:20:53.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/5"}}},"commit":{"parents":[{"commit":"590214aa666bba9d4acee4e68af0cc65d17dfe6f","subject":"fix extension loading test to use correct namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/590214aa666bba9d4acee4e68af0cc65d17dfe6f"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-18 23:20:46.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b13ae11f9f10641b4944ca2db3a8cc78f33cbb34"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b13ae11f9f10641b4944ca2db3a8cc78f33cbb34"}]},"branch":"refs/heads/master"},"9da1ea4447fb734720a76aacce36ceae272a23c9":{"kind":"TRIVIAL_REBASE","_number":6,"created":"2025-02-18 23:36:48.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/6"}}},"commit":{"parents":[{"commit":"0ae7bc30ca558df29f0df44cc15fa8f8c3344d0d","subject":"fix extension loading test to use correct namespace","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0ae7bc30ca558df29f0df44cc15fa8f8c3344d0d"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-18 23:36:42.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9da1ea4447fb734720a76aacce36ceae272a23c9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9da1ea4447fb734720a76aacce36ceae272a23c9"}]},"branch":"refs/heads/master"},"3274375fbbe29216993863ca79505458812b704b":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-02-19 05:16:00.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/7"}}},"commit":{"parents":[{"commit":"6e0f5fabdc155385101a58c5aa68ae21c4b61f83","subject":"explicit oem extn names and fix parallel tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6e0f5fabdc155385101a58c5aa68ae21c4b61f83"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-19 05:15:54.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3274375fbbe29216993863ca79505458812b704b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3274375fbbe29216993863ca79505458812b704b"}]},"branch":"refs/heads/master"},"45ced79bcabf06cf6a3aac16011d43826475d49d":{"kind":"REWORK","_number":8,"created":"2025-02-22 17:45:21.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/8"}}},"commit":{"parents":[{"commit":"6e0f5fabdc155385101a58c5aa68ae21c4b61f83","subject":"explicit oem extn names and fix parallel tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6e0f5fabdc155385101a58c5aa68ae21c4b61f83"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-22 17:45:12.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/45ced79bcabf06cf6a3aac16011d43826475d49d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/45ced79bcabf06cf6a3aac16011d43826475d49d"}]},"branch":"refs/heads/master"},"da092394650e952d1991f24f13403b5e020567d9":{"kind":"REWORK","_number":9,"created":"2025-02-22 19:02:37.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/57/940557/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940557/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940557/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940557/9"}}},"commit":{"parents":[{"commit":"6e0f5fabdc155385101a58c5aa68ae21c4b61f83","subject":"explicit oem extn names and fix parallel tests","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6e0f5fabdc155385101a58c5aa68ae21c4b61f83"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-01-31 20:06:27.000000000","tz":-300},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-22 19:02:36.000000000","tz":-360},"subject":"migrate sushy_oem_idrac to sushy","message":"migrate sushy_oem_idrac to sushy\n\nCopied sushy_oem_idrac@585703a22c8 to sushy/oem/dell. The following\nmechanical replacements were made:\ns|sushy_oem_idrac/tests|sushy/tests/oem/dell|\ns|sushy_oem_idrac|sushy|sushy.oem.dell|\n\nAdditionally, code style and format was updated automatically to\nmatch the rest of the sushy repo. And the usage of random was switched\nto secrets.\n\nUsed \u0027dell\u0027 over \u0027idrac\u0027 to match the entry point value, which matches\nthe value in Oem via redfish.\n\nChange-Id: Ic7ebcf39494b2f224bb8350c0df611f742fd6c66\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/da092394650e952d1991f24f13403b5e020567d9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/da092394650e952d1991f24f13403b5e020567d9"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":10239,"name":"Dmitry Tantsur","email":"dtantsur@protonmail.com","username":"dtantsur"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
