)]}'
{"id":"openstack%2Fsushy~940757","triplet_id":"openstack%2Fsushy~master~I96c08c7daa2f4258dda1547a32f01747709d6d0b","project":"openstack/sushy","branch":"master","attention_set":{},"removed_from_attention_set":{"36316":{"account":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"last_update":"2025-02-05 10:31:56.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I96c08c7daa2f4258dda1547a32f01747709d6d0b","subject":"trivial: fix variable in log message","status":"MERGED","created":"2025-02-05 01:36:15.000000000","updated":"2025-02-05 10:33:38.000000000","submitted":"2025-02-05 10:31:56.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"940757","meta_rev_id":"202197a6cbdf0851b5dd6730ee7325bba6f9d89e","_number":940757,"virtual_id_number":940757,"owner":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-05 10:31:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2025-02-05 05:48:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2025-02-05 09:36:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-02-05 01:39:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":1,"date":"2025-02-05 09:36:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-05 01:39:12.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"REVIEWER"},{"updated":"2025-02-05 02:06:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-05 05:48:27.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"},{"updated":"2025-02-05 09:36:02.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"}],"messages":[{"id":"7b6ec53b35abf36ad6f3343f900d70183f1beb57","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"date":"2025-02-05 01:36:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3ea7a931aa967f2b50e340442ffdd464cd61df77","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2025-02-05 01:39:12.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"aa95bf6d1cd2b7f05947f77796db27a6fd47677f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 02:06:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a702d42932804034ac3a9137816360cd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c9e06f7be984a71b998738b894b6075 : SUCCESS in 4m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/05aed2f0bee048d3a443efe43f302a0f : SUCCESS in 3m 18s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/923feaa9dc2a4bd7a8df69a6419aff94 : SUCCESS in 2m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/2cb2117742a34b1bb0b6f89af8f6735a : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/41e478f0009c449d860f3b0b82a2d754 : SUCCESS in 4m 32s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a24f8f46eda740749501db141e2cc38d : SUCCESS in 6m 46s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/c860eab869f442f8b23c857b5d4df897 : SUCCESS in 26m 13s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/9e8b82ed13aa4dd09e2b33b15b00c197 : SUCCESS in 28m 54s","accounts_in_message":[],"_revision_number":1},{"id":"dcde6b0b1b4dc663d8b5f6a1a0ec6513a1cdb18f","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-05 05:48:27.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"482f75431a51444ab7e4b1746885fdc2a2cc3406","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-05 09:36:02.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d02a90175cad579f63d90b1ef4d88e1a9e504335","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 09:36:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"317ab3852e356f53c44ab8aafe85f9b7b8f86320","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:31:55.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c86d03d40c240dfa5f3876f693223d3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/95a6e9efc7a740fe804b7719dc11fb73 : SUCCESS in 5m 23s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/54ba5cb5e8d74a96b5025e28bef96425 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4548b52942d747eeafcde5ded1f05ffa : SUCCESS in 2m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc4e4bdca92f42dc969aea30d7f95047 : SUCCESS in 5m 19s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/bbd60b8620da4e3287fa48cff8dcafd8 : SUCCESS in 7m 37s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/828f8d1e6ae84fa2a54ab05599202076 : SUCCESS in 34m 30s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/8e85a45d62bc4f3abfbda8b6b0f16998 : SUCCESS in 39m 34s","accounts_in_message":[],"_revision_number":1},{"id":"36fab2f09ebc9a56ec80ac039a71b6c3e0590fa3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:31:56.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"202197a6cbdf0851b5dd6730ee7325bba6f9d89e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-05 10:33:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bf056a6d142046fea1e83eaf1b0d0931\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d8dd39dbc8b4e998f0e640839c80415 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"81fbc97cea885905e899c342fd6c30c5bbd51b39","revisions":{"81fbc97cea885905e899c342fd6c30c5bbd51b39":{"kind":"REWORK","_number":1,"created":"2025-02-05 01:36:15.000000000","uploader":{"_account_id":36316,"name":"Alexon Ferreira de Oliveira","display_name":"Alexon Oliveira","email":"alolivei@redhat.com","username":"alexonoliveira"},"ref":"refs/changes/57/940757/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/57/940757/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940757/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940757/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/57/940757/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/57/940757/1"}}},"commit":{"parents":[{"commit":"27434e48e4ce944dedf5da3ee38b09ce54410428","subject":"enable flake8 logging checks in ruff","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/27434e48e4ce944dedf5da3ee38b09ce54410428"}]}],"author":{"name":"Alexon Oliveira","email":"alolivei@redhat.com","date":"2025-02-05 01:21:59.000000000","tz":-180},"committer":{"name":"Alexon Oliveira","email":"alolivei@redhat.com","date":"2025-02-05 01:36:05.000000000","tz":-180},"subject":"trivial: fix variable in log message","message":"trivial: fix variable in log message\n\nProper variable name should be %(err)s instead of %(error)s\n\nChange-Id: I96c08c7daa2f4258dda1547a32f01747709d6d0b\nSigned-off-by: Alexon Oliveira \u003calexon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/81fbc97cea885905e899c342fd6c30c5bbd51b39"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/81fbc97cea885905e899c342fd6c30c5bbd51b39"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
