)]}'
{"id":"openstack%2Fsushy~942166","triplet_id":"openstack%2Fsushy~master~I8f58769b39315caf6e7939665efc9e3cc6faa0f0","project":"openstack/sushy","branch":"master","topic":"merge-sushy-oem-idrac","attention_set":{},"removed_from_attention_set":{"5890":{"account":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"last_update":"2025-02-20 16:02:24.000000000","reason":"Change was submitted"}},"hashtags":["ironic-week-prio"],"change_id":"I8f58769b39315caf6e7939665efc9e3cc6faa0f0","subject":"explicit oem extn names and fix parallel tests","status":"MERGED","created":"2025-02-19 05:16:00.000000000","updated":"2025-02-20 16:04:18.000000000","submitted":"2025-02-20 16:02:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"942166-merge-sushy-oem-idrac","meta_rev_id":"6a60d3e1c8f092678981cb4fd5af9eb969348c5c","_number":942166,"virtual_id_number":942166,"owner":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-02-20 16:02:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"all":[{"value":2,"date":"2025-02-19 20:17:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":2,"date":"2025-02-19 15:39:37.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":1,"date":"2025-02-20 15:13:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-02-19 06:06:21.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-02-19 15:39:37.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-02-19 20:17:54.000000000","updated_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"reviewer":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"state":"REVIEWER"}],"messages":[{"id":"3c2f83328b632e8a132db013764d4865db862b9d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-19 05:16:00.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d680c6c63e26067bfaaa598f4490a900cd0b4118","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-19 06:06:21.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0d2fe325eaf3407abc59c9a244457091\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/286f47c0577f40ff869d0b6b2a541f0d : SUCCESS in 3m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a54779261d58456cb46acc00cf508713 : SUCCESS in 3m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f9531793e38f4313afb624834bbab12e : SUCCESS in 2m 50s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/4fc5990467824dbeae41583a31448ab8 : SUCCESS in 3m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f260f25cc9cf43cc866d933bd386e6d2 : SUCCESS in 6m 10s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/8ac875d1932247a09e74b7c19d71e4a2 : SUCCESS in 7m 07s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/5eb98a9b991f4e74aea7b866e7b2b5a4 : SUCCESS in 44m 14s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/52771c15ed484f5a8f0dd48358466b9d : SUCCESS in 28m 44s","accounts_in_message":[],"_revision_number":1},{"id":"239bdbcd0d427733c491129ca6ee518b156db539","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-19 15:39:37.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2afcdf840cfe084d9fcbf46dd59d7cf2429c6159","author":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"},"date":"2025-02-19 20:17:54.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"fb979d22b3ca3bb2d3d059c935f328e983e1db10","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2025-02-20 02:23:31.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":1},{"id":"c19cae3238577dab65e88b55be1cacdd9df9430e","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-02-20 15:13:15.000000000","message":"Patch Set 1: Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"05e419a8d0843877557f316717531089f0d1e578","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 15:13:53.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b82f05de9ba30f9ec84a9735d94f42ee4175661b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:02:24.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57508c12288147448af9e061514308e6\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4b3a52a6ce1347c9bb176cd9738b69df : SUCCESS in 4m 11s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0cdd325642cd4054b56208adfdf7282f : SUCCESS in 4m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/0d7c29d8da7a4d5e8b2f7d9ee9725f2d : SUCCESS in 3m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4a1458da6b404715be25d857d16031e3 : SUCCESS in 8m 21s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/cee29cfb57f64d93b7123ac29f89b116 : SUCCESS in 15m 00s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/dc1a68b1de1643c4bf4ab397ba41c6d8 : SUCCESS in 42m 37s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/61090e3f03af4629b35d0d9728043cb0 : SUCCESS in 45m 32s","accounts_in_message":[],"_revision_number":1},{"id":"d4f6f834af4f6b5c9f955a487616b1ef37e1ef45","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:02:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"6a60d3e1c8f092678981cb4fd5af9eb969348c5c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-20 16:04:18.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/038f6fcfd4a443e9a61a6004daa007bb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d17a350060fa4649bb31cd51090fe715 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6e0f5fabdc155385101a58c5aa68ae21c4b61f83","revisions":{"6e0f5fabdc155385101a58c5aa68ae21c4b61f83":{"kind":"REWORK","_number":1,"created":"2025-02-19 05:16:00.000000000","uploader":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"ref":"refs/changes/66/942166/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/66/942166/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/66/942166/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/66/942166/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/66/942166/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/66/942166/1"}}},"commit":{"parents":[{"commit":"d0752a5598170c0ad4358f154dad24ef9ef7b938","subject":"Merge \"enable pyupgrade via ruff to Python 3.7\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/d0752a5598170c0ad4358f154dad24ef9ef7b938"}]}],"author":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-19 04:01:29.000000000","tz":-360},"committer":{"name":"Doug Goldstein","email":"cardoe@cardoe.com","date":"2025-02-19 05:12:26.000000000","tz":-360},"subject":"explicit oem extn names and fix parallel tests","message":"explicit oem extn names and fix parallel tests\n\nRather than building up the full name and then trying to parse the\nstring later on, just pass the two parts to the function that needs it.\nThe tests are also running in parallel and the mocks and patching of the\nglobal object falls down when there are any oem extensions installed\npast the ones that match up with what the tests mock out. Currently\nsushy only has one for system under contoso, which is what the first\nmock uses so it just works. But adding even another causes\nundeterministic failure in the tests. This should patch things and keep\nthem in scope to the test.\n\nChange-Id: I8f58769b39315caf6e7939665efc9e3cc6faa0f0\nSigned-off-by: Doug Goldstein \u003ccardoe@cardoe.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6e0f5fabdc155385101a58c5aa68ae21c4b61f83"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/6e0f5fabdc155385101a58c5aa68ae21c4b61f83"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":10342,"name":"Jay Faulkner","display_name":"JayF","email":"jay@jvf.cc","username":"JayF","status":"youtube.com/@oss-gr / podcast.gr-oss.io"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
