)]}'
{"id":"openstack%2Fsushy~963282","triplet_id":"openstack%2Fsushy~master~Ie65d70a7862f4cc7058269921bc512a8a9754bae","project":"openstack/sushy","branch":"master","attention_set":{},"removed_from_attention_set":{"38437":{"account":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"last_update":"2025-10-22 12:42:03.000000000","reason":"Change was submitted"},"23851":{"account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"last_update":"2025-10-22 11:55:59.000000000","reason":"\u003cGERRIT_ACCOUNT_23851\u003e replied on the change","reason_account":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}}},"hashtags":["ironic-week-prio"],"change_id":"Ie65d70a7862f4cc7058269921bc512a8a9754bae","subject":"Check required credentials in a detailed way","status":"MERGED","created":"2025-10-07 13:24:42.000000000","updated":"2025-10-22 12:43:04.000000000","submitted":"2025-10-22 12:42:03.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"963282","meta_rev_id":"fd5b76281210ac0b9bfef93ae61e5c945e279d62","_number":963282,"virtual_id_number":963282,"owner":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-10-22 12:42:03.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"all":[{"value":2,"date":"2025-10-15 12:46:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":2,"date":"2025-10-22 11:55:59.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":1,"date":"2025-10-22 11:55:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2025-10-07 13:30:33.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2025-10-07 14:49:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2025-10-15 12:46:10.000000000","updated_by":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"reviewer":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"state":"REVIEWER"}],"messages":[{"id":"afb152bb140678480783ba2d456b99f64a9d690e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 13:24:42.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"533632ba4b610015d7ace0a5af89fa55e235cab2","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-07 13:30:33.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1633bb90144b50d27dc5b79df0422b1874715ba3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 13:58:52.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"647510b95843e695d6cf1ce7e706d1b968a5c960","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 14:49:59.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c6957ee462a4ab1b6d05299f896e660\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2fbfb3445a264afa80b4ca78855b1e3b : SUCCESS in 2m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2f73906cf8943de9a3bb87dcc000e90 : SUCCESS in 3m 32s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9f4987b2f04840f49ed1c52f1ae0c988 : SUCCESS in 3m 41s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/41c7c1e5d280441287f5911143ac7dd3 : SUCCESS in 3m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/b48b3e1f3bd54a719d4b90277d6c7061 : SUCCESS in 4m 50s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60422fcc9640484588d54c62c44c99b7 : SUCCESS in 5m 01s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/f0ea7883cde642b185aeaec6b3746932 : SUCCESS in 7m 27s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/3fa1fc2ab4b84ecfa59d522d73b07868 : SUCCESS in 45m 30s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/6970ae07b2664091a93277f3d2dea7c7 : SUCCESS in 28m 47s","accounts_in_message":[],"_revision_number":2},{"id":"062326c79094573303a9bf62fda1fa95ab2ed240","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 15:11:40.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"4387990b85fe62ba8519502341b382de5a77f8ab","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 15:14:12.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"4c01b7f6a85e405c71ad03850233cb4c6373402d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 15:21:31.000000000","message":"Patch Set 5: Commit message was updated.","accounts_in_message":[],"_revision_number":5},{"id":"367206bc3d87073bec6adc1b1d56e4fb03618baa","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 15:38:35.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b7b18cd4c71ccbb55b20a93420f5e483132d5394","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 15:39:31.000000000","message":"Uploaded patch set 7: Commit message was updated.","accounts_in_message":[],"_revision_number":7},{"id":"ae68579d9f757d5ac9c2ec6b2a19c13578324d05","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-07 15:47:53.000000000","message":"Patch Set 7:\n\n(1 comment)","accounts_in_message":[],"_revision_number":7},{"id":"69d2dc86c6f28aeb57ec2e264d17575b56fd5fef","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-07 16:29:18.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ab6a8b4a14a41a191058621c9ce7bed\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6f44b75ce4604de1867335a6707ab1cf : SUCCESS in 3m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bcd07975525248d09bf4d26d211b8221 : SUCCESS in 4m 00s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c13288b01eda478a9031f53ee013f41d : SUCCESS in 3m 16s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d23fba4e9ef48fdb9405cca434a6221 : SUCCESS in 3m 19s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9fd1043d481e4534997c1e17790ea0ec : SUCCESS in 5m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/07ef44a678f74676a0776d9bf472464f : SUCCESS in 7m 28s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/633a48a722e54c4f853405422390af8a : SUCCESS in 4m 10s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/4560c1e26cd54c2697a1c1fc6d4e1f6c : SUCCESS in 12m 07s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/aed2c60a319a4822a54efbc7978dcda2 : SUCCESS in 29m 05s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/af3244e28e6a4228806f0f57c97cbed5 : SUCCESS in 44m 37s","accounts_in_message":[],"_revision_number":7},{"id":"4e845f05bfe37ebb64225dda590eea94f57c828d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-08 14:10:12.000000000","message":"Patch Set 8: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"6f70d5c743443e3acb257c1a02ac8838f01c8845","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-08 15:03:02.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6707160c00ad460bb8a3e19e9e87bb72\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6154945cfb474a7399a146f0bd053987 : SUCCESS in 3m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/497ecd648d4b4a898bf6a369db654b61 : SUCCESS in 6m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dcd64cce33ed40a4877d484caa5ea9a3 : SUCCESS in 3m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/addf3496cea34266b5abac8e3ec7e984 : SUCCESS in 3m 10s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/641c45259e624a8788fad83333a606ef : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/663876e2103e48b79b8ddfaecef30962 : SUCCESS in 6m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/204d3ba2b9ee4da09221e2e0eef85d14 : SUCCESS in 2m 45s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/453950c774d743648dcd26a41c8606d0 : SUCCESS in 6m 35s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/e246850c278745a4a2620ac293869e87 : SUCCESS in 46m 15s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/4e1f29baf77649f5a242903708b02500 : SUCCESS in 31m 38s","accounts_in_message":[],"_revision_number":8},{"id":"100c6ffbb63132b67dce779eb9497d5d4d5c96da","tag":"autogenerated:gerrit:setHashtag","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-09 09:45:14.000000000","message":"Hashtag added: ironic-week-prio","accounts_in_message":[],"_revision_number":8},{"id":"f96516595a6d6c465d170801a1b3259387e55243","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-09 09:50:46.000000000","message":"Patch Set 8: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":8},{"id":"2321b643b7a64ca795599880f23609fc0c1548c6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-13 10:51:17.000000000","message":"Uploaded patch set 9.\n\nOutdated Votes:\n* Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"298b6afc293e4c7bb0c5cf3610382bed92ea4c07","author":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"date":"2025-10-13 10:52:24.000000000","message":"Patch Set 9:\n\n(2 comments)","accounts_in_message":[],"_revision_number":9},{"id":"b2832370125692041798d339983b189d23c13649","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-13 11:24:39.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/03fe7844016843fa897426b264dc5ef2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/425410055d884d8a934ae16bec91d20c : SUCCESS in 3m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8f4fc327ef7a46a58f0319986b289b35 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b2ac6e4f26f34065abaf1e8ea6a219db : SUCCESS in 4m 29s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a3b1e99fd9d841ecb002b1968e0c9f71 : SUCCESS in 3m 07s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/59db4e4d92344f92ba840ed350237fb1 : SUCCESS in 6m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e094de2870c547d6949ba3be97a66edf : SUCCESS in 8m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/037bf04eec5746629b5ab28ce360d725 : SUCCESS in 2m 31s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/fb61774e9f3f4ca6bb56743260c3938b : SUCCESS in 10m 54s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/6e8373bdd5244334b728d44d652b73cb : SUCCESS in 21m 36s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/45b2c396f784419ca2801cdb1968dff7 : SUCCESS in 31m 54s","accounts_in_message":[],"_revision_number":9},{"id":"db59a8d5713e28d82df83f3682c3334c35156768","author":{"_account_id":36770,"name":"cid","display_name":"cid","email":"cid@gr-oss.io","username":"cidelight","status":"@gr-oss upstream: Doing good IRONIC things..."},"date":"2025-10-15 12:46:10.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"02f0e66d60a5ea1bac70edf4535105eff639c608","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2025-10-22 11:55:59.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"312724f125b27520aa215c197b698b531f685e9b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 11:56:25.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"eeb69789cbc6cfa1373b4fe487149e970e112992","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 12:42:03.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7131fc2e15934245b3a2620c97841ba4\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/af9e90a67d554be8bf81875adce0cf0e : SUCCESS in 4m 33s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5409a243b3bf48e88c8db483eba1f702 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/febc9a8174324562b42474822f5f5f5d : SUCCESS in 4m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073de377a23b44ceb72fad84d10a01ec : SUCCESS in 5m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25f7945fc8bd4670a705084c30a45376 : SUCCESS in 2m 21s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/20cee480a2484d7ca296451a51f810b6 : SUCCESS in 8m 43s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/d4a390521aa04d99b3452ebc32041a09 : SUCCESS in 41m 16s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/d60b9a892ff24fd098292da7dc7688d6 : SUCCESS in 34m 04s","accounts_in_message":[],"_revision_number":9},{"id":"2e4b6147e1b93a5ed08ad2f0c59aef728d9ac9b7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 12:42:03.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"fd5b76281210ac0b9bfef93ae61e5c945e279d62","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-10-22 12:43:04.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8fc7b70a23a4a7fa7cd58737768a708\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4252024075254e2480c67951ae5d0cda : SUCCESS in 40s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/03a0045c1d4f439a820b4b7a5664d4d7 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"408cfe178be9b21ba8b436afc173c41c2a0f63c1","revisions":{"bdf59944e6c16c0b2d5accd73d6899ab3755cfcb":{"kind":"REWORK","_number":1,"created":"2025-10-07 13:24:42.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/1"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:24:32.000000000","tz":180},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bdf59944e6c16c0b2d5accd73d6899ab3755cfcb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bdf59944e6c16c0b2d5accd73d6899ab3755cfcb"}]},"branch":"refs/heads/master"},"af3bb4ffcb0768ba72f2a1280296a15e8dadb299":{"kind":"REWORK","_number":2,"created":"2025-10-07 13:58:52.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/2"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:58:29.000000000","tz":180},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/af3bb4ffcb0768ba72f2a1280296a15e8dadb299"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/af3bb4ffcb0768ba72f2a1280296a15e8dadb299"}]},"branch":"refs/heads/master"},"8f6001b5c2af9100303ce5fcc818bd97f2d5a45c":{"kind":"REWORK","_number":3,"created":"2025-10-07 15:11:40.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/3"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 15:10:10.000000000","tz":180},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nRelated-Bug: OCPBUGS-62309\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8f6001b5c2af9100303ce5fcc818bd97f2d5a45c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/8f6001b5c2af9100303ce5fcc818bd97f2d5a45c"}]},"branch":"refs/heads/master"},"2f44520b915d55268518d5c1281f8892b0f451f8":{"kind":"REWORK","_number":4,"created":"2025-10-07 15:14:12.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/4"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 15:13:51.000000000","tz":180},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nRelated-Bug: OCPBUGS-62309\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2f44520b915d55268518d5c1281f8892b0f451f8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/2f44520b915d55268518d5c1281f8892b0f451f8"}]},"branch":"refs/heads/master"},"bec1d532d4d07d256453334bfa24ac1a53fd79c8":{"kind":"NO_CODE_CHANGE","_number":5,"created":"2025-10-07 15:21:31.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/5"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"Aykut Bulgu","email":"abulgu@redhat.com","date":"2025-10-07 15:21:31.000000000","tz":0},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bec1d532d4d07d256453334bfa24ac1a53fd79c8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/bec1d532d4d07d256453334bfa24ac1a53fd79c8"}]},"branch":"refs/heads/master","description":"Edit commit message"},"9c3f7ca3f56b693320047fb6e4e731ea5dea34a0":{"kind":"REWORK","_number":6,"created":"2025-10-07 15:38:35.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/6"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 15:37:38.000000000","tz":180},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nImprove Redfish VirtualMedia ``InsertMedia`` error handling on BMCs that\n    require credentials but return unstructured errors (e.g., responses that\n    omit ``error.code`` and report ``ActionParameterMissing`` only via\n    ``@Message.ExtendedInfo`` or free-text messages). Sushy now detects these\n    cases and retries with ``UserName``/``Password`` parameters, allowing ISO\n    mounting to proceed. This also preserves compatibility with legacy\n    ``error.code \u003d\u003d *GeneralError`` responses that mention missing parameters.\n\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9c3f7ca3f56b693320047fb6e4e731ea5dea34a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/9c3f7ca3f56b693320047fb6e4e731ea5dea34a0"}]},"branch":"refs/heads/master"},"fe1d8b6ebd0f823f819206c0077ac03700153e04":{"kind":"NO_CODE_CHANGE","_number":7,"created":"2025-10-07 15:39:31.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/7"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 15:38:57.000000000","tz":180},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nImprove Redfish VirtualMedia ``InsertMedia`` error handling on BMCs that\nrequire credentials but return unstructured errors (e.g., responses that\nomit ``error.code`` and report ``ActionParameterMissing`` only via\n``@Message.ExtendedInfo`` or free-text messages). Sushy now detects these\ncases and retries with ``UserName``/``Password`` parameters, allowing ISO\nmounting to proceed. This also preserves compatibility with legacy\n``error.code \u003d\u003d *GeneralError`` responses that mention missing parameters.\n\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fe1d8b6ebd0f823f819206c0077ac03700153e04"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/fe1d8b6ebd0f823f819206c0077ac03700153e04"}]},"branch":"refs/heads/master"},"3a6bd0a2a0ee5ed0ec930acb90c007d6010dc7c0":{"kind":"NO_CODE_CHANGE","_number":8,"created":"2025-10-08 14:10:12.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/8"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"Aykut Bulgu","email":"abulgu@redhat.com","date":"2025-10-08 14:10:12.000000000","tz":0},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nImprove Redfish VirtualMedia ``InsertMedia`` error handling on BMCs that\nrequire credentials but return unstructured errors (e.g., responses that\nomit ``error.code`` and report ``ActionParameterMissing`` only via\n``@Message.ExtendedInfo`` or free-text messages). Sushy now detects these\ncases and retries with ``UserName``/``Password`` parameters, allowing ISO\nmounting to proceed. This also preserves compatibility with legacy\n``error.code \u003d\u003d *GeneralError`` responses that mention missing parameters.\n\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\nCloses-Bug: #2127088\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3a6bd0a2a0ee5ed0ec930acb90c007d6010dc7c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/3a6bd0a2a0ee5ed0ec930acb90c007d6010dc7c0"}]},"branch":"refs/heads/master","description":"Edit commit message"},"408cfe178be9b21ba8b436afc173c41c2a0f63c1":{"kind":"REWORK","_number":9,"created":"2025-10-13 10:51:17.000000000","uploader":{"_account_id":38437,"name":"mabulgu","display_name":"Aykut Bulgu","email":"abulgu@redhat.com","username":"mabulgu"},"ref":"refs/changes/82/963282/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/82/963282/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/82/963282/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/82/963282/9"}}},"commit":{"parents":[{"commit":"cd6f5851b6147dea4eac3929997531e8a4ff5f56","subject":"Merge \"Add comprehensive PCIe device support to Sushy\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/cd6f5851b6147dea4eac3929997531e8a4ff5f56"}]}],"author":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-07 13:20:37.000000000","tz":180},"committer":{"name":"mabulgu","email":"abulgu@redhat.com","date":"2025-10-13 10:38:58.000000000","tz":180},"subject":"Check required credentials in a detailed way","message":"Check required credentials in a detailed way\n\nImprove Redfish VirtualMedia ``InsertMedia`` error handling on\nBMCs that require credentials but return unstructured errors\n(e.g., responses that omit ``error.code`` and report\n``ActionParameterMissing`` only via ``@Message.ExtendedInfo``\nor free-text messages). Sushy now detects these cases and retries\nwith ``UserName``/``Password`` parameters, allowing ISO mounting to\nproceed. This also preserves compatibility with legacy\n``error.code \u003d\u003d *GeneralError`` responses that mention missing\nparameters.\n\nChange-Id: Ie65d70a7862f4cc7058269921bc512a8a9754bae\nSigned-off-by: mabulgu \u003cabulgu@redhat.com\u003e\nAssisted-By: chatgpt-5 (for research purposes)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/408cfe178be9b21ba8b436afc173c41c2a0f63c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/408cfe178be9b21ba8b436afc173c41c2a0f63c1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
