)]}'
{"id":"openstack%2Fsushy~982219","triplet_id":"openstack%2Fsushy~master~I9e7ee050b6092dbc5529b519c24c18e98008f6e4","project":"openstack/sushy","branch":"master","topic":"lentzi90/connect-double-wrap","attention_set":{},"removed_from_attention_set":{"37872":{"account":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"last_update":"2026-03-26 15:17:12.000000000","reason":"Change was submitted"},"15519":{"account":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"last_update":"2026-03-26 15:17:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I9e7ee050b6092dbc5529b519c24c18e98008f6e4","subject":"Prevent double-wrapping timeout tuple in Connector _op","status":"MERGED","created":"2026-03-26 10:09:24.000000000","updated":"2026-03-26 15:18:03.000000000","submitted":"2026-03-26 15:17:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"982219-lentzi90/connect-double-wrap","meta_rev_id":"9f08a1419111e24a84151bce361bc129b98dbe7d","_number":982219,"virtual_id_number":982219,"owner":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2026-03-26 15:17:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":2,"date":"2026-03-26 14:18:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":2,"date":"2026-03-26 13:24:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"all":[{"value":1,"date":"2026-03-26 14:18:36.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"value":0,"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}],"CC":[{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-03-26 11:09:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2026-03-26 11:24:38.000000000","updated_by":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"reviewer":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"state":"CC"},{"updated":"2026-03-26 13:24:06.000000000","updated_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"reviewer":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"state":"REVIEWER"},{"updated":"2026-03-26 14:18:36.000000000","updated_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"reviewer":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"state":"REVIEWER"}],"messages":[{"id":"5796694a79a32231f68011ef9954179a13e589db","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"date":"2026-03-26 10:09:24.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d6172ab77e3fb9bf6225ee0e705527f4de0ca1d2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 11:09:15.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4c70fcc5569b47dd98a78aee0474a2a1\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ec36bff79f9f469aaaf3cfa823298e8a : SUCCESS in 18m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c75ebd6f40544f3698d667aa4d818c74 : FAILURE in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/98004a8d709d4c018a2f90adc6c49859 : SUCCESS in 2m 46s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/f7dde78f70c4408b9716447cd45ac246 : SUCCESS in 4m 07s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/6504b9435b68427c8471f4af2285f3b4 : SUCCESS in 6m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4fbf8fa6cde14ed8a33f8eb29f5ceed1 : SUCCESS in 4m 54s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/83dce5649cca40429180546f67f92fb7 : SUCCESS in 9m 14s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/8b3d8a6ac9944eb4a7530028808350dc : SUCCESS in 47m 59s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/b1fd4e8b76224dec9bc0cf50c334615b : SUCCESS in 53m 51s","accounts_in_message":[],"_revision_number":1},{"id":"e9213bd06714fb3b26e5671292d4ab7eeb7d55dd","author":{"_account_id":15519,"name":"Iury Gregory Melo Ferreira","display_name":"Iury Gregory","email":"iurygregory@gmail.com","username":"iurygregory"},"date":"2026-03-26 11:24:38.000000000","message":"Patch Set 1:\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"879b3e0b61989391ac4b7dfb5d5a672aa2193d0d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"date":"2026-03-26 11:29:49.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"3cd44ea0bcf4905d46a65b8398fc3e5906160e32","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"date":"2026-03-26 11:35:31.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"7accde3c890518af14bd74bb08a871c616e2091d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 12:31:58.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7e41d04dfb7443e785bdff4737f5f045\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3c6594102c504683a9ed96404a37bc0e : SUCCESS in 3m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73ca6f8753b14c5ab11743c9625435c1 : SUCCESS in 2m 25s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/efe8716c55e64e5192a170bc70a986f9 : SUCCESS in 2m 34s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/69777086a2134dabbf9dc1cf9dab5fca : SUCCESS in 2m 49s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/3a3fb9560b79429dbe7b142591fc82a4 : SUCCESS in 4m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/384107abfab44fd8b646a2d33d40f82f : SUCCESS in 5m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4dca1f69078a468d8c700c63beff2e3d : SUCCESS in 3m 03s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/404eaa967de746d5a6605002687ad1fd : SUCCESS in 10m 17s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/eb1db20cf1624d5d885ee183de11b9f4 : SUCCESS in 25m 53s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/54479c77c1e24dad98e5f291ffc62f4d : SUCCESS in 51m 13s","accounts_in_message":[],"_revision_number":3},{"id":"a87d2b450f4b80096f5767ebdcbc73309e97cb96","author":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"},"date":"2026-03-26 13:24:06.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"c1afa3a6a77b3aed744786f1f1119d873d91ca60","author":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"date":"2026-03-26 13:45:42.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"467109b94a835edd117b245c1b29c6c84fc2f314","author":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"},"date":"2026-03-26 14:18:36.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"a86131ea4c5486092b01660fac267dc7d2d96cfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 14:19:03.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"450b07ccafdb5905d4d48c330ec04a65f93b34d7","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:17:12.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e7756631696247f3bbd223a3b8c7e044\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/205e360d5af74730a84ae63e707ddb59 : SUCCESS in 2m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0aa1cffcc5f495ba3ee1c5429a189a3 : SUCCESS in 3m 16s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/e9103a8570254dc193f9e8d7270c8e1a : SUCCESS in 3m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3f8b73f4f40b40cf9b8e2965d34ea700 : SUCCESS in 8m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1a01cb4c3b3e4b52be7750b5e9afe4b6 : SUCCESS in 2m 36s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/a19498879bd140babfc65d07f9532738 : SUCCESS in 9m 04s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/cdb40cb9ed3e4536b0b79502e7309a8c : SUCCESS in 51m 55s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/067ae963cb794bccbff1ae171209c69d : SUCCESS in 43m 30s","accounts_in_message":[],"_revision_number":3},{"id":"298f36c7de73b8bb2feeb80af4641c5df63822ce","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:17:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"9f08a1419111e24a84151bce361bc129b98dbe7d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-03-26 15:18:03.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/486f4bab4eab42eea538fe3968c9a9eb\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cd6a8a3f72d74c738d93be52f047841a : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/73093e0ef32c4fb0ba3a1d7d88ac8c10 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"b87365bdc29c91a8c88c18c8577820f85d9c712f","revisions":{"0c0130f8a383cfecf478ccfd8fdc7ebbb885eacf":{"kind":"REWORK","_number":1,"created":"2026-03-26 10:09:24.000000000","uploader":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"ref":"refs/changes/19/982219/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/19/982219/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/19/982219/1"}}},"commit":{"parents":[{"commit":"950648a0af4f79239f3f379c57edda03ef7e0e2e","subject":"Revert \"Skipping UsbCd workaround on Supermicro ARS-111GL-NHR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/950648a0af4f79239f3f379c57edda03ef7e0e2e"}]}],"author":{"name":"Lennart Jern","email":"lennart.jern@est.tech","date":"2026-03-26 10:04:57.000000000","tz":0},"committer":{"name":"Lennart Jern","email":"lennart.jern@est.tech","date":"2026-03-26 10:09:13.000000000","tz":120},"subject":"Prevent double-wrapping timeout tuple in Connector _op","message":"Prevent double-wrapping timeout tuple in Connector _op\n\nAdd test to ensure timeout is not nested on recursive retry when\nconnect_timeout is set. Fixes #2146416.\n\nChange-Id: I9e7ee050b6092dbc5529b519c24c18e98008f6e4\nSigned-off-by: Lennart Jern \u003clennart.jern@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0c0130f8a383cfecf478ccfd8fdc7ebbb885eacf"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/0c0130f8a383cfecf478ccfd8fdc7ebbb885eacf"}]},"branch":"refs/heads/master"},"61587a1365d7492a1e8b46e486afaf0580c96ea9":{"kind":"REWORK","_number":2,"created":"2026-03-26 11:29:49.000000000","uploader":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"ref":"refs/changes/19/982219/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/19/982219/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/19/982219/2"}}},"commit":{"parents":[{"commit":"950648a0af4f79239f3f379c57edda03ef7e0e2e","subject":"Revert \"Skipping UsbCd workaround on Supermicro ARS-111GL-NHR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/950648a0af4f79239f3f379c57edda03ef7e0e2e"}]}],"author":{"name":"Lennart Jern","email":"lennart.jern@est.tech","date":"2026-03-26 10:04:57.000000000","tz":0},"committer":{"name":"Lennart Jern","email":"lennart.jern@est.tech","date":"2026-03-26 11:29:32.000000000","tz":120},"subject":"Prevent double-wrapping timeout tuple in Connector _op","message":"Prevent double-wrapping timeout tuple in Connector _op\n\nAdd test to ensure timeout is not nested on recursive retry when\nconnect_timeout is set.\n\nCloses-Bug: 2146416\nChange-Id: I9e7ee050b6092dbc5529b519c24c18e98008f6e4\nSigned-off-by: Lennart Jern \u003clennart.jern@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/61587a1365d7492a1e8b46e486afaf0580c96ea9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/61587a1365d7492a1e8b46e486afaf0580c96ea9"}]},"branch":"refs/heads/master"},"b87365bdc29c91a8c88c18c8577820f85d9c712f":{"kind":"REWORK","_number":3,"created":"2026-03-26 11:35:31.000000000","uploader":{"_account_id":37872,"name":"Lennart Jern","email":"lennart.jern@est.tech","username":"lennartj"},"ref":"refs/changes/19/982219/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/19/982219/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/19/982219/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/19/982219/3"}}},"commit":{"parents":[{"commit":"950648a0af4f79239f3f379c57edda03ef7e0e2e","subject":"Revert \"Skipping UsbCd workaround on Supermicro ARS-111GL-NHR\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/950648a0af4f79239f3f379c57edda03ef7e0e2e"}]}],"author":{"name":"Lennart Jern","email":"lennart.jern@est.tech","date":"2026-03-26 10:04:57.000000000","tz":0},"committer":{"name":"Lennart Jern","email":"lennart.jern@est.tech","date":"2026-03-26 11:35:11.000000000","tz":120},"subject":"Prevent double-wrapping timeout tuple in Connector _op","message":"Prevent double-wrapping timeout tuple in Connector _op\n\nAdd test to ensure timeout is not nested on recursive retry when\nconnect_timeout is set.\n\nCloses-Bug: 2146416\nChange-Id: I9e7ee050b6092dbc5529b519c24c18e98008f6e4\nSigned-off-by: Lennart Jern \u003clennart.jern@est.tech\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b87365bdc29c91a8c88c18c8577820f85d9c712f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/b87365bdc29c91a8c88c18c8577820f85d9c712f"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":23851,"name":"Riccardo Pittau","email":"elfosardo@gmail.com","username":"elfosardo"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":5890,"name":"Doug Goldstein","email":"cardoe@cardoe.com","username":"cardoe"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
