)]}'
{"id":"openstack%2Fsushy~994735","triplet_id":"openstack%2Fsushy~master~I527a767035fc119ac780e03c120597c10d124c8a","project":"openstack/sushy","branch":"master","topic":"release-health-check","hashtags":[],"change_id":"I527a767035fc119ac780e03c120597c10d124c8a","subject":"DNM: gate health test","status":"ABANDONED","created":"2026-06-24 13:23:20.000000000","updated":"2026-06-29 15:19:29.000000000","total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"6578f0c4a03e76f645a1bce25da429d37bb276b3","_number":994735,"virtual_id_number":994735,"owner":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2026-06-24 14:22:20.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Do Not Backport"," 0":"Backport Review Needed","+1":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2026-06-24 14:22:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"eb6364c8a57ee7e1c60f51c7bb1f2441599d0dc8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-24 13:23:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3e0a100b2d8103c7199cd2c941dafd802a9d15ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-06-24 14:22:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1f95f277506a42a1a5c46903ada6a8f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/629f9ba3c89f459a861dd6dbeb600746 : SUCCESS in 3m 50s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e449ff288af14fda8407da2822a2c49c : SUCCESS in 4m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7995c237991c413dad1d2f4db5befbe1 : SUCCESS in 3m 20s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/3a07bf69f3aa409b9d17e91345d83dc6 : SUCCESS in 3m 15s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/d6ddae283777480daf073da86bafe0a9 : SUCCESS in 5m 33s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a8f1d9e6fb56475aa28bc9074c0f7797 : SUCCESS in 4m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a4bec6d21e354ef3a3d1ca42970e03fc : SUCCESS in 4m 11s\n- ironic-cross-sushy https://zuul.opendev.org/t/openstack/build/b4cedb50682c4d96a1f79b1fa36dec9f : SUCCESS in 8m 57s\n- sushy-tempest-bios-redfish-pxe https://zuul.opendev.org/t/openstack/build/ad49c5152ed342d1ad4cdda768134b97 : SUCCESS in 42m 37s\n- sushy-tempest-uefi-redfish-vmedia https://zuul.opendev.org/t/openstack/build/e16ebcd1912743bd9718b33b73b13f19 : SUCCESS in 50m 06s","accounts_in_message":[],"_revision_number":1},{"id":"6578f0c4a03e76f645a1bce25da429d37bb276b3","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2026-06-29 15:19:29.000000000","message":"Abandoned\n\ngate looks healthy. abandoning.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"4e933d500abfc34963fb763c287068e376b66d7b","revisions":{"4e933d500abfc34963fb763c287068e376b66d7b":{"kind":"REWORK","_number":1,"created":"2026-06-24 13:23:20.000000000","uploader":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"ref":"refs/changes/35/994735/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/sushy","ref":"refs/changes/35/994735/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/994735/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/994735/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/sushy refs/changes/35/994735/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/sushy refs/changes/35/994735/1"}}},"commit":{"parents":[{"commit":"162132070b0d60a5172e10698dfb7e31ade24d3c","subject":"filter out drives reporting null size on inspect","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/162132070b0d60a5172e10698dfb7e31ade24d3c"}]}],"author":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-06-24 13:23:09.000000000","tz":120},"committer":{"name":"Elod Illes","email":"elod.illes@est.tech","date":"2026-06-24 13:23:18.000000000","tz":120},"subject":"DNM: gate health test","message":"DNM: gate health test\n\nChange-Id: I527a767035fc119ac780e03c120597c10d124c8a\nSigned-off-by: Elod Illes \u003celod.illes@est.tech\u003e\nGenerated-By: openstack/releases:tools/repository_test_generator.sh\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4e933d500abfc34963fb763c287068e376b66d7b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/sushy/commit/4e933d500abfc34963fb763c287068e376b66d7b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
